)]}'
{"id":"openstack%2Fswift~945889","triplet_id":"openstack%2Fswift~master~I92a427df7f21beb15d92c63894f771234fc5e6f9","project":"openstack/swift","branch":"master","topic":"post-metadata-timestamp-combined-update","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-04-02 20:15:51.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"37720":{"account":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"last_update":"2025-05-07 11:59:58.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I92a427df7f21beb15d92c63894f771234fc5e6f9","subject":"Add atomic DB update for put_timestamp and metadata to fix POST race","status":"NEW","created":"2025-03-28 20:00:13.000000000","updated":"2025-05-16 23:24:30.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":14,"unresolved_comment_count":12,"has_review_started":true,"meta_rev_id":"80eb5acb060e67afaf0886c4e62554bdefb92e14","_number":945889,"virtual_id_number":945889,"owner":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-11 17:20:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-28 20:52:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-28 21:08:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-31 00:10:13.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-04-29 00:22:23.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"46d5742ab0a55b96011d0bbb4601d894d40aeec5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-03-28 20:00:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"333bca669d8e39927b805466fd978bed0604d8f5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-03-28 20:01:13.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"426cbfa23b49fd3519c075fb9000bdb182200988","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-03-28 20:01:21.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":1},{"id":"c3741cbdf5d387380ff9de0375eb58a58bed0635","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 20:52:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32030473244642f9a0254173f8a8135e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4605d25f31da4903be6f45413fadc780 : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":1},{"id":"84c5080897a44f8233aad62d4b97d7b213702b12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-28 21:08:57.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7173730e2c934f66b9877e382a08326f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7938b41a36154741a24f9e13aae1f1f1 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/bcf5f39ec0ac48eba167f8b2a74f09c5 : SUCCESS in 1h 01m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5b5f1f03d0ba4f8dac820a647a8c482e : SUCCESS in 45m 23s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5a519dfee2b34070b50ba177e294dd98 : SUCCESS in 1h 01m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8948c61ecc424a399fdf45701de4fd61 : SUCCESS in 1h 05m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/65bd6b47048c418c9fcedb8e1c02439c : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab97df3515fd42e0a845e2b0b47da077 : SUCCESS in 20m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/710c887ef0df42e7af9fdf3f0003811e : SUCCESS in 20m 33s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c108d2b81b3e437c92ec1e684a893fb2 : FAILURE in 5m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cbb5b9a9213448d8b0106f03724a9f8f : SUCCESS in 19m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b1492ced9944a4d9af781d0024cb2ca : SUCCESS in 20m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8133797b64db4362b58cee9a2b6b1460 : SUCCESS in 19m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6f057b05eabb4087989fd7986d2b33a3 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c6d0e71ac9364f02b3f42897c85b8867 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c1b6ace72eae479da8008d7cab57f47f : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/170624ed580f4b35b366a1dfc81dcb4e : SUCCESS in 5m 44s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b1860c2e5e9d432f9273098de7a4e82e : SUCCESS in 16m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/5c48bf27a0fc4fa08ee117ac3288da04 : SUCCESS in 3m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2d6b44979744ca69efb5bed1c6a5cc4 : SUCCESS in 39m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/21a012d3e4124548a867e37b0f72a4a5 : SUCCESS in 17m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0780186fdde14fe3a471926d99e49dfe : SUCCESS in 22m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2fcaceaa5bac42818e0fb94ca492cf4d : SUCCESS in 21m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43a166e1913646f985a79b4bc4cb597c : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a1132fa9d0dd481a9dc62bba5ede2e37 : SUCCESS in 21m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b8a352c4b7e455dbeba117c2286cca5 : SUCCESS in 49m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f85201f672284404b583bad9a1d354b8 : SUCCESS in 54m 47s","accounts_in_message":[],"_revision_number":1},{"id":"d6fe52921a4f601219c805f07fb9e1bc2c2933c1","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-03-31 00:10:13.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"a7f2f300c7da6a7a883c9e2c5766e6d0003220b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-04-02 20:15:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"dad23cd58907f056a748e146d473a1bbbfd6dbd3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 20:59:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aedb7e14671642a1ac12105aa156d825\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1f1ee1d1c241496b9ed6aef4e56fe1b8 : SUCCESS in 42m 02s","accounts_in_message":[],"_revision_number":2},{"id":"d7648d4359c6a9528c421ed7d43cf2a6613c201f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 21:26:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d869bc475d6f46c9bd1422bddf13b44e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7349b5dfc8844fcca9c30bd270d7d08f : SUCCESS in 11m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/32ce5f56f5a5428fb00d9b39dbd23865 : SUCCESS in 49m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/69e3b80491c9461ba66dc513cbd5b5f8 : SUCCESS in 49m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0840b52a1e2d4b6798491c91adf6e787 : SUCCESS in 52m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31e56f66c89f4183934a1fd97ed6b47c : SUCCESS in 1h 03m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1d368646481f453d9a65a92e2d5c4410 : SUCCESS in 19m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fa9b33fc57249019943cb589635469c : SUCCESS in 19m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a77e128a4e774fc6a4fab8c9f5191aee : SUCCESS in 17m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b4142166252b4fd5a5434681f9522100 : SUCCESS in 5m 25s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/22b170162a25453c8e51b3efed31a079 : SUCCESS in 14m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72768357a28942d29dec4ad6209dec13 : SUCCESS in 24m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9249ea14e5484b8db7f8502b1983f200 : SUCCESS in 34m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d9c67e352abc47d082278c7e8725bcd6 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b1e2dd7413dd4db4bc1ecdb9bfb9e099 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b75b6456dc224d8cab6f1b18465cb792 : SUCCESS in 15m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b5fe8b2c961b4404a22406e3c7682de9 : SUCCESS in 5m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e99017802c24ab3b21fa1abc49c1c5c : SUCCESS in 15m 25s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f1b87946d31844a3a882b22e0af7176a : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/14b8c3a3d2c84337b7a7c80a931eaaf9 : SUCCESS in 49m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e78e0b35de2b4ebf8cc1101feaf06cea : SUCCESS in 17m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c0ff2b94385144afb4db52064f3e1ec5 : SUCCESS in 30m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/27bc3db0e9fa4a89ab776118ee6d0820 : SUCCESS in 19m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ed45d5b589c41c6ba187b1d5ed5755a : SUCCESS in 5m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/973ab6a40dc24b5c9935a05abaa48e72 : SUCCESS in 17m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/97d0c721e02d43109707afaebca1978d : SUCCESS in 42m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2fa92e81acd24ff090c69440dc0b48eb : SUCCESS in 30m 45s","accounts_in_message":[],"_revision_number":2},{"id":"ce9c46813e038e2f33fa6eff885d4110d6e7707e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-04-02 23:10:59.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"adac01c30a5e4fb3940dfe8f7b347b80b562c33f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 23:59:14.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f91665a174d47588fab61878e96dc7d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3d6a30949c014eb9bf48046be44192c0 : SUCCESS in 46m 20s","accounts_in_message":[],"_revision_number":3},{"id":"0e2809b6da52ba31ba1b44d00635c54838b2a690","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 00:15:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eba4780968a644ae879f859df826e6b7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd32fb41d8a2474a98a8ac67eac14b0a : SUCCESS in 4m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/d143e83007b648b1aaf6b5d2576839f8 : SUCCESS in 37m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/07d88f0fedfa4dc4bf9da7bfe864f655 : SUCCESS in 26m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7b4fe16400de4d7384534b03e5bd9078 : SUCCESS in 38m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9a8ab9eab85f48c594caba7b263e98ea : SUCCESS in 58m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cade969a7ca7425191914c4b4b373860 : SUCCESS in 10m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e778cecfad5d4150a1bddb8af38fbbc4 : SUCCESS in 16m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/beb4eaade21446b0b23379b5bbd56403 : SUCCESS in 18m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3aa17a50ace403bb9eda616699d2eff : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f8726fccba8840eda07ad0925f96d840 : SUCCESS in 13m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dce4fca6d138447892fa3b34641bc04e : SUCCESS in 13m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1e4598aac4314da38296ffcdc39ed2bb : SUCCESS in 18m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c34e0a65eb75435687710faa098ce13d : SUCCESS in 12m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da88d974137b4c8e9196c12717923cbf : SUCCESS in 12m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4d14de826c1e4c6c983dd71eab47c5c7 : SUCCESS in 14m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/625bb8739bdf4d40bc96ce507d940be7 : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d64cff4d6a0f466481a4c3f761ab7b80 : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/751ab51fc3cd4c909a2dc72bae04bb8e : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/75df8e18bf4e458f9f8784e5e9c61e3a : SUCCESS in 59m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12f2d9b7199f4b4198d095bb7ca89c50 : SUCCESS in 30m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f5648edc27df4768af585bf5f47e4d4f : SUCCESS in 29m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bbb4da06797a47ff973df3b16323a428 : SUCCESS in 17m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8529cfdf161e41f7b632fe6041541769 : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c7403db3ab554cada2f38fa9e3a9dc7e : SUCCESS in 15m 39s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bf84ea564cf44c8e87ee3275bd2333e6 : SUCCESS in 45m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3672710ea6834442ab45e12431991951 : SUCCESS in 27m 05s","accounts_in_message":[],"_revision_number":3},{"id":"a6114cb73a6c95ce633dfd33f1b22e00b383c22d","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-29 00:22:23.000000000","message":"Patch Set 3:\n\n(5 comments)","accounts_in_message":[],"_revision_number":3},{"id":"514963d625dd01aca229ece448e2aefb305b6ad9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-05-07 11:01:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"b0669e055cfc110f11171e0d69f993bbfbc5068d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:44:16.000000000","message":"Patch Set 4:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1abfe4dd05e34026bd010f88a03ad0e0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e6c46b7938404e08a69dfe972a1fca01 : FAILURE in 40m 41s","accounts_in_message":[],"_revision_number":4},{"id":"3a6240a5ceac9bdecab3d441d55f08e551d9103f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:59:58.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fbaeeba6365344da8a8ac467699fd59a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab45ba7f37304851b463d1f5336d2996 : SUCCESS in 9m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/a530decb0b3049629566df2a0fc52223 : SUCCESS in 49m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b71df598a2064acd92ce79d4d1fe67b1 : SUCCESS in 48m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a5c12279e401426a93d1505365615f49 : SUCCESS in 44m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1dc20ef27f364d859b542cbb9b112e72 : SUCCESS in 52m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90603c98994c47ecb56ed2fa31334736 : SUCCESS in 16m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0302e09737e54139a09c8695a414db7d : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39b68673c8d941e5914bb071104f3a3b : SUCCESS in 22m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6adfbfcb85414be8adab78e371a6cc91 : SUCCESS in 4m 07s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/fce6b540e75f4ba9a8ecd1d39b8742f0 : FAILURE in 19m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/52859e68bae94dd589198ec64aec1d94 : FAILURE in 7m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d9c4d428b9ab4611898984729c8d3edd : FAILURE in 24m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4e06da427b494925bd83c10dbdd92e2f : FAILURE in 28m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/16139663cdd44e758f579e7dcb1d02fe : SUCCESS in 11m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3dc739d97ba4affab092ccbafff9f79 : SUCCESS in 12m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab3a479b537b4b10a98e61876aa35441 : SUCCESS in 12m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9758d1eb8e540a082950412d7aea04e : SUCCESS in 5m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f81dbf2ebe3342ffaeba86c8a7e56808 : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c398ef9eaee14dd19f70bf86c8c694a3 : SUCCESS in 4m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d4487ce2fbc94e889992d69f0fed5140 : SUCCESS in 53m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/662038a863ef4d589d2c61c99f6472b5 : SUCCESS in 17m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4eb14a4c4ce4491b9dc5ca48bcb9e396 : SUCCESS in 39m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da76fe91342442e6b2177d0ba43eaf7c : FAILURE in 17m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d2cb536a9724b9fa88dfc9443b61be5 : SUCCESS in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c704140816d94e68b4d9257bf492db28 : SUCCESS in 10m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/28be01d23b0d450bb337f14c8925f757 : SUCCESS in 37m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0224b57d320b4ebb9bdbdf2127a9f06a : SUCCESS in 53m 58s","accounts_in_message":[],"_revision_number":4},{"id":"8299371c6cba9a064b8ff342293ebaf2f4610653","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"date":"2025-05-11 15:58:33.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"10b4f6664ab19d9633f8897f721147abd010e516","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-11 16:50:28.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/104e8bcdd6d44e96a3e362e75e6d7990\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/80b4a6f69702458799eabb6cd96352f1 : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":5},{"id":"a823f0bfee337b3d511728eac17dbbd5bff99556","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-11 17:20:26.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca708578790a42c8bce1ab677594612a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab3c343c1ba1478b884574be7beb3136 : SUCCESS in 9m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/83e379eccaf74a5ab9700426eda9b3c7 : SUCCESS in 56m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/673633cbf6f942c08c159da26864d25d : SUCCESS in 28m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4f1ed515eec34787a3c94ba061ec5d20 : SUCCESS in 52m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3c7a6b2f3a274d61b4ef25e2e21652e6 : SUCCESS in 1h 14m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/313db31f101343eb9ae4bbdef236fc08 : SUCCESS in 17m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8805e9be34094fdb98c49ab4a32931a6 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4951561d05714c14b4f01b8898c7c939 : SUCCESS in 14m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/647f58645f254da181e3f7fb7a87e716 : FAILURE in 3m 34s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/48079e13830d401692360ee7b8d358f4 : SUCCESS in 15m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/452a0b16ada64cde81e81c6a52ee0e71 : SUCCESS in 17m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a604c7c46aef40d4b6a0893ca3dc5c29 : SUCCESS in 26m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d9d7788c2cbe48df955d41878886ded8 : SUCCESS in 19m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/189b438d31884bb48ef99f41df597bc9 : SUCCESS in 11m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/def3cce2e5194ba9a5df035033b10859 : SUCCESS in 12m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bc75f4f5463f446c9353ddd4f0a54fe1 : SUCCESS in 12m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e86ac1dd20844beb8c5b58f85bb0c676 : SUCCESS in 5m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/768dbf575c484f8d83f1bbdecb7873e5 : SUCCESS in 13m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b443c809421646fe9f93b74e87ecea8c : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a486c7e17712464d841825dd5b9a1b84 : SUCCESS in 49m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/61a0de9eb37d4edc8fef10123d3bec48 : SUCCESS in 41m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e26d7e2a39b7443a8f6a28d7d1d31c62 : SUCCESS in 41m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b1ed9469d50e4a8499ec9f844763d014 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcb11915a3114dbe9a4f10fa9504ba13 : SUCCESS in 5m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e3e109b39c3043d9bb0110e29d479fa3 : SUCCESS in 19m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3d38f39264664474a4905255fc69a473 : SUCCESS in 52m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e63dfcfc2ee5437ab6e1c8d9f039a8f0 : SUCCESS in 54m 00s","accounts_in_message":[],"_revision_number":5},{"id":"80eb5acb060e67afaf0886c4e62554bdefb92e14","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-16 23:24:30.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"f41265a952fcac15a3e4985023bd80b8f7104809","revisions":{"a4a71ec0be7c12657163754bff5195b47b9df76c":{"kind":"REWORK","_number":1,"created":"2025-03-28 20:00:13.000000000","uploader":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"ref":"refs/changes/89/945889/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/945889/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/945889/1"}}},"commit":{"parents":[{"commit":"05143a99f8f3c860a887e4eb49688aef1e7a4a78","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05143a99f8f3c860a887e4eb49688aef1e7a4a78"}]}],"author":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-03-28 19:58:38.000000000","tz":-240},"committer":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-03-28 19:58:38.000000000","tz":-240},"subject":"Add atomic DB update for put_timestamp and metadata to fix POST race","message":"Add atomic DB update for put_timestamp and metadata to fix POST race\n\nAdded a new method to update both put_timestamp and container metadata in a single DB transaction, eliminating a race condition in POST requests.\n\nCo-authored-by: James Nguyen \u003cngmqng@gmail.com\u003e\nChange-Id: I92a427df7f21beb15d92c63894f771234fc5e6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4a71ec0be7c12657163754bff5195b47b9df76c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a4a71ec0be7c12657163754bff5195b47b9df76c"}]},"branch":"refs/heads/master"},"abe917c50a3d24138631585c500e47eff6051591":{"kind":"REWORK","_number":2,"created":"2025-04-02 20:15:51.000000000","uploader":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"ref":"refs/changes/89/945889/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/945889/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/945889/2"}}},"commit":{"parents":[{"commit":"05143a99f8f3c860a887e4eb49688aef1e7a4a78","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05143a99f8f3c860a887e4eb49688aef1e7a4a78"}]}],"author":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-03-28 19:58:38.000000000","tz":-240},"committer":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-04-02 20:15:13.000000000","tz":0},"subject":"Add atomic DB update for put_timestamp and metadata to fix POST race","message":"Add atomic DB update for put_timestamp and metadata to fix POST race\n\nAdded a new method to update both put_timestamp and container metadata in a single DB transaction, eliminating a race condition in POST requests.\n\nCo-authored-by: James Nguyen \u003cngmqng@gmail.com\u003e\nChange-Id: I92a427df7f21beb15d92c63894f771234fc5e6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abe917c50a3d24138631585c500e47eff6051591"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abe917c50a3d24138631585c500e47eff6051591"}]},"branch":"refs/heads/master"},"bd8c9c0eab71ffd134e080cf7307e4d0cf29c415":{"kind":"REWORK","_number":3,"created":"2025-04-02 23:10:59.000000000","uploader":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"ref":"refs/changes/89/945889/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/945889/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/945889/3"}}},"commit":{"parents":[{"commit":"05143a99f8f3c860a887e4eb49688aef1e7a4a78","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05143a99f8f3c860a887e4eb49688aef1e7a4a78"}]}],"author":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-03-28 19:58:38.000000000","tz":-240},"committer":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-04-02 23:10:28.000000000","tz":0},"subject":"Add atomic DB update for put_timestamp and metadata to fix POST race","message":"Add atomic DB update for put_timestamp and metadata to fix POST race\n\nAdded a new method to update both put_timestamp and container metadata in a single DB transaction, eliminating a race condition in POST requests.\n\nCo-authored-by: James Nguyen \u003cngmqng@gmail.com\u003e\nChange-Id: I92a427df7f21beb15d92c63894f771234fc5e6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd8c9c0eab71ffd134e080cf7307e4d0cf29c415"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bd8c9c0eab71ffd134e080cf7307e4d0cf29c415"}]},"branch":"refs/heads/master"},"27c413ff19da8dcf2b878f2b06e1578d080f68bb":{"kind":"REWORK","_number":4,"created":"2025-05-07 11:01:46.000000000","uploader":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"ref":"refs/changes/89/945889/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/945889/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/945889/4"}}},"commit":{"parents":[{"commit":"05143a99f8f3c860a887e4eb49688aef1e7a4a78","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05143a99f8f3c860a887e4eb49688aef1e7a4a78"}]}],"author":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-03-28 19:58:38.000000000","tz":-240},"committer":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-05-07 11:01:09.000000000","tz":0},"subject":"Add atomic DB update for put_timestamp and metadata to fix POST race","message":"Add atomic DB update for put_timestamp and metadata to fix POST race\n\nAdded a new method to update both put_timestamp and container metadata in a single DB transaction, eliminating a race condition in POST requests.\n\nCo-authored-by: James Nguyen \u003cngmqng@gmail.com\u003e\nChange-Id: I92a427df7f21beb15d92c63894f771234fc5e6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27c413ff19da8dcf2b878f2b06e1578d080f68bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/27c413ff19da8dcf2b878f2b06e1578d080f68bb"}]},"branch":"refs/heads/master"},"f41265a952fcac15a3e4985023bd80b8f7104809":{"kind":"REWORK","_number":5,"created":"2025-05-11 15:58:33.000000000","uploader":{"_account_id":37720,"name":"Boosung Kim","display_name":"Boosung Kim","email":"boosung@boosungkim.com","username":"boosungkim"},"ref":"refs/changes/89/945889/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/945889/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/945889/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/945889/5"}}},"commit":{"parents":[{"commit":"05143a99f8f3c860a887e4eb49688aef1e7a4a78","subject":"Update master for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05143a99f8f3c860a887e4eb49688aef1e7a4a78"}]}],"author":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-03-28 19:58:38.000000000","tz":-240},"committer":{"name":"boosungkim","email":"boosung@boosungkim.com","date":"2025-05-11 15:58:07.000000000","tz":0},"subject":"Add atomic DB update for put_timestamp and metadata to fix POST race","message":"Add atomic DB update for put_timestamp and metadata to fix POST race\n\nAdded a new method to update both put_timestamp and container metadata in a single DB transaction, eliminating a race condition in POST requests.\n\nCo-authored-by: James Nguyen \u003cngmqng@gmail.com\u003e\nChange-Id: I92a427df7f21beb15d92c63894f771234fc5e6f9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f41265a952fcac15a3e4985023bd80b8f7104809"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f41265a952fcac15a3e4985023bd80b8f7104809"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
