)]}'
{"id":"openstack%2Fswift~945984","triplet_id":"openstack%2Fswift~master~I32d5ef49c86a7a7f0f821ba2f4fd4a4aa34fc00a","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-10 15:14:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I32d5ef49c86a7a7f0f821ba2f4fd4a4aa34fc00a","subject":"sq: s3request: don\u0027t try to check non-existent trailers","status":"ABANDONED","created":"2025-03-31 17:46:45.000000000","updated":"2025-04-10 15:14:49.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"3b4c937d27b5c33feeedfde3bee35da65cab91cd","_number":945984,"virtual_id_number":945984,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-01 13:10:54.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-31 18:30:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-03-31 19:03:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7db605b4977d254bd7571378de287655e4ffaf44","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-31 17:46:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aafaa6afa73472c567af694ff1fffef3e0993dd3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-03-31 17:56:35.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"91bce19a7f0cd5c215424856dfbc69d496c01051","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 18:30:29.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/552117904459473db50d7187694b9814\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40d66bfca92745d4bb1cf5af586f0435 : FAILURE in 35m 08s","accounts_in_message":[],"_revision_number":1},{"id":"646288fb0858e2eead29d9a22582dbec38aa2b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-31 19:03:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/158599131b7c4ebda130747239d4ace5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087f830e794249a3a1d1bbfbe747ed0c : SUCCESS in 10m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/f33133f9dbe24b97a3679bea3d1aa8c1 : SUCCESS in 56m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8eb6d5bca634b59ae9141a9ddf47d61 : SUCCESS in 55m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b02d96c9c3464b5799199ce475107d66 : SUCCESS in 54m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1811c3e2e5864c0a8b6c75375f2416c9 : SUCCESS in 31m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3149163748a4cc2bea3188b4e2171a7 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/02f7ac8b81f34960a84ae340a6143e61 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79d2fda4895e428b94fa3741989be84d : SUCCESS in 21m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ecfca3f2462c421ab470b2e9a0640a0f : SUCCESS in 7m 10s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d57b170d52e644c7be4774b0a371e340 : FAILURE in 19m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c567d70c4b6b4f5d9a2af61512ad7260 : FAILURE in 17m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/11555746db614bd2be317e07565d2e3f : FAILURE in 30m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a304510d01e4cfb82913a2d9e064b83 : SUCCESS in 12m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6240b893c5614493bb43d1f6dd8658f2 : SUCCESS in 15m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/22540320422945979b384354b8fb0684 : SUCCESS in 15m 07s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/21c0287d301c45a482dfb9740b6e01a5 : SUCCESS in 5m 41s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e1deee29c8d34c33989dfaefe9f0d95b : SUCCESS in 13m 35s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7d2e50debd8e42edb5d890db7c29955d : SUCCESS in 8m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ac62f33b983e402fbfae3d3f27fc70a4 : SUCCESS in 53m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dcd041ce109043e5a139e3698e7eab37 : SUCCESS in 35m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cf5257339ad0440ab7fce5e055846f78 : SUCCESS in 37m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63d43216008747aea403a9ca72830056 : FAILURE in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9752565587741ce889d5b84f06aac2c : SUCCESS in 3m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7d6103250066422d92a51b1087f48234 : SUCCESS in 19m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19d8eb161a1d403680cbede74426d734 : SUCCESS in 41m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f56e794dbd649bda62ee1a6ec3c402a : SUCCESS in 50m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1d82f96a7c09efd843680c5470cc01b103c70f82","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-01 11:47:50.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"fc2abf4cb2279dd8338474ee73eaf0100b9940ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 12:23:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b07116d3e7054216b429cee2d9650a9f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7132bfeeabc04821bb98622091537e61 : SUCCESS in 33m 39s","accounts_in_message":[],"_revision_number":2},{"id":"cc62286ec37955569c7771f49ac802f0fdf93cfb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 13:10:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bd09740c11d4eea992b2f323d6a5d4b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3801aa13e670486dab723e0217ce8a02 : SUCCESS in 7m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/18d4181aef9b4377a6fa1207c0ec8777 : SUCCESS in 1h 00m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4499017ba3014cb487838c7534623ee7 : SUCCESS in 56m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/464f35d0cd1842238d1927f78a10035d : SUCCESS in 44m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6ba797148c9047ad966a059d40c9cb99 : SUCCESS in 53m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4bcd91dc3bc54835a0ea67d0e4ccd415 : SUCCESS in 17m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef72bb9acf384a40b685f9f29786479e : SUCCESS in 22m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0abcc948504a462fb894507ac91fd5c0 : SUCCESS in 20m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee369fd511e44abf8b45ad8deb3b3dcb : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf0119954c8c40f292e9416483fdfb8a : SUCCESS in 17m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c05a6eddc3e2450089826d4e8e016a08 : SUCCESS in 17m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e055516fc61f44d2ad0f09445bc1510e : SUCCESS in 34m 25s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a256127addd14ef1992d08094117bf05 : SUCCESS in 14m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e13abf099c9b4ebbb82ca9cd6fc807df : SUCCESS in 13m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee770c8892ff45c7b71989a838c4ed25 : SUCCESS in 15m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/02d1980aa8a84bf390dae6f601a861ce : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9aff750aeac34085b19b1813270b702d : SUCCESS in 17m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/62f47b0068584be092256ae763006b6b : SUCCESS in 6m 52s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6536c789695c4f9b9b47e0801c9f54b5 : SUCCESS in 54m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d652dfadb10449df8d53d182beba2709 : SUCCESS in 34m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6f32fef003dc4f40aae6b97d66893bde : SUCCESS in 48m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2b393c9785224e57b0c6b11501cc5e96 : SUCCESS in 13m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71cb3f1d84ba430ea859f8a021be1083 : SUCCESS in 3m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d07f84d5bcb84de7adf0150d40116033 : SUCCESS in 17m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f24837555f8346f1a19f6f392925bf9f : SUCCESS in 46m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be218101009a49bbb5516ccbb9374b7a : SUCCESS in 50m 53s","accounts_in_message":[],"_revision_number":2},{"id":"3b4c937d27b5c33feeedfde3bee35da65cab91cd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-10 15:14:49.000000000","message":"Abandoned\n\ncovered by https://review.opendev.org/c/openstack/swift/+/945983/3","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6849821c0b1cb48a2cc6c591663fe84186d7a930","revisions":{"3a179a4c10dc49cf6b0f60413cdfc30b04d83a0b":{"kind":"REWORK","_number":1,"created":"2025-03-31 17:46:45.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/945984/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/945984/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/945984/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/945984/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/945984/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/945984/1"}}},"commit":{"parents":[{"commit":"7ecf297eabe8436722a553241cdcf620e2002ee7","subject":"sq: s3request: only validate x-amz-trailer once","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ecf297eabe8436722a553241cdcf620e2002ee7"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-31 17:35:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-31 17:45:48.000000000","tz":60},"subject":"sq: s3request: don\u0027t try to check non-existent trailers","message":"sq: s3request: don\u0027t try to check non-existent trailers\n\nIf the client sends a non-streaming request with an x-amz-trailer\nheader then return an error early. Previously we\u0027d try to get the\ntrailers from a HashingInput, or even an un-wrapped input, neither of\nwhich have a trailers attribute.\n\nChange-Id: I32d5ef49c86a7a7f0f821ba2f4fd4a4aa34fc00a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a179a4c10dc49cf6b0f60413cdfc30b04d83a0b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3a179a4c10dc49cf6b0f60413cdfc30b04d83a0b"}]},"branch":"refs/heads/master"},"6849821c0b1cb48a2cc6c591663fe84186d7a930":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-01 11:47:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/84/945984/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/945984/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/945984/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/945984/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/945984/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/945984/2"}}},"commit":{"parents":[{"commit":"49dbac8aa055ecf661de77bf3cda689f1d0dab27","subject":"sq: s3request: only validate x-amz-trailer once","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49dbac8aa055ecf661de77bf3cda689f1d0dab27"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-03-31 17:35:44.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-01 11:39:15.000000000","tz":60},"subject":"sq: s3request: don\u0027t try to check non-existent trailers","message":"sq: s3request: don\u0027t try to check non-existent trailers\n\nIf the client sends a non-streaming request with an x-amz-trailer\nheader then return an error early. Previously we\u0027d try to get the\ntrailers from a HashingInput, or even an un-wrapped input, neither of\nwhich have a trailers attribute.\n\nChange-Id: I32d5ef49c86a7a7f0f821ba2f4fd4a4aa34fc00a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6849821c0b1cb48a2cc6c591663fe84186d7a930"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6849821c0b1cb48a2cc6c591663fe84186d7a930"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
