)]}'
{"id":"openstack%2Fswift~946239","triplet_id":"openstack%2Fswift~master~Iff796188b88d834b957dd4c8df066b8d7ad75567","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-12 17:17:21.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iff796188b88d834b957dd4c8df066b8d7ad75567","subject":"s3api: log selected crc32c implementation","status":"ABANDONED","created":"2025-04-03 12:10:35.000000000","updated":"2025-04-12 17:17:21.000000000","total_comment_count":10,"unresolved_comment_count":7,"has_review_started":true,"meta_rev_id":"e42633963abd1d12ba54791ca2ca2722cbe823a4","_number":946239,"virtual_id_number":946239,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-11 15:30:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-04-11 14:17:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-03 13:02:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-03 13:17:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 16:04:01.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"7976a25538940a4de63868c8e0450c1caac98f14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-03 12:10:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"78298d12aebb8ff5a6726eb60579ad70b999d142","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 13:02:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d02ae94f5e64a039872a535dc439987\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42a9fe15b95d40e9ac902cf325f9ddfa : SUCCESS in 50m 17s","accounts_in_message":[],"_revision_number":1},{"id":"61ee801f8fcbe3feb64500be31939e1501ba6804","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 13:17:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c541cda5b7f4d398fa39d6e0d52c1ef\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f908c4aa0274378b8e64c65f1cc230d : SUCCESS in 12m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/bd0c34c9e6e4473eb3fcf04964487b1e : SUCCESS in 27m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/11909ff10ab6413d90d4836277de9a44 : SUCCESS in 52m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51cddc9706024452a70269bde616885b : SUCCESS in 49m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ff4c6271322421398ca514adb02705c : SUCCESS in 1h 00m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f37c2e81216a4093aadf5ffcdb0b9220 : SUCCESS in 20m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/201c98e296bf42d785338ebe1e0a8856 : SUCCESS in 16m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f21af291a2e64f21b29bebbcd99c2057 : SUCCESS in 20m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5f43071ec9614e6494bf28e4ed62358f : SUCCESS in 4m 49s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9ef62675c5f54e55aac327935f649394 : SUCCESS in 20m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/648a4c07399449d39786490fafcb1798 : SUCCESS in 19m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae1c9d2a2a0f473caaa2314fab99e16d : SUCCESS in 33m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/031afc3f057640c1adea79651861b226 : SUCCESS in 7m 54s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b8e97ec181374666b8a3d95d1a25099c : SUCCESS in 9m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d7d65d090e5b4237bdba5008579472a1 : SUCCESS in 12m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/73c59d8897724f3c915df604b1f03d3b : SUCCESS in 7m 35s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/882a9d4b9aef450cb3e4f4e6cc561428 : SUCCESS in 7m 46s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/7736098840f545aaafc814e53faf8a43 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/067a97a167f942939231224cdf71e7f1 : SUCCESS in 57m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6429617396a747e3b2ed69ffd2fcb437 : SUCCESS in 42m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fc03df89a9694258a962d30635fce217 : SUCCESS in 36m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c295f217e144169bae8857cbe9eb066 : SUCCESS in 20m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ae25f7075154eb3b5735c599bc3c5f9 : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cb30cea11c44f0d87054012c450bf2a : SUCCESS in 9m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/486922aaef3d4f9fa4870f7583ccef91 : SUCCESS in 53m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a58909b39a94807ac4de722778b9523 : SUCCESS in 45m 21s","accounts_in_message":[],"_revision_number":1},{"id":"71407931a59a1930a83cc422e890f857c9028446","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-07 17:13:02.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"248068c78672b866de5ca4d38c34a21dd397b213","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-07 17:16:46.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"50750fc026494bbe3146dfd1a0a488ad405d4d27","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 17:53:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dc3c0a83d74640b0a96c150d8c4910e8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40db233efa6343cc87b67b2341c1acfd : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":2},{"id":"3be1767455b42d429ef4cbb61a202f94e62a1481","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 18:21:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b393c73592f64213baa4d5d23472c1bf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d77ac285eba4b69bedfd8c487c4ef75 : SUCCESS in 12m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/e91f94f7d24144b6a70f19458e197eb4 : SUCCESS in 55m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f0bcefdb42314be5be1243a88786d7f0 : SUCCESS in 55m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d7cc0f26dae4983b83456734aa9aaf6 : SUCCESS in 30m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/49465c6f10a444229b3fb82a6eb9f98f : SUCCESS in 47m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afd32acf207c42afb8c9ba884f32242b : SUCCESS in 10m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12338d4a91b543e782feb99eb13266a3 : SUCCESS in 22m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eba06959717a480fa8bee35932433477 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d1fcd722d4b34f1c9d41ca32eadc268d : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/cf2047623a3c47df80d36d256b500fa1 : SUCCESS in 16m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d8211de50c4464aaacaf02a37cb0522 : SUCCESS in 22m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0fc6d0ab6b3f4cd5975fbaa331b1bcdb : SUCCESS in 11m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c89771194d0043f79a38efe3ab0a615a : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b2e10421e2a41aaaa40e87e1a039fb0 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/76543ad332274caa9dbf6537d1df3534 : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a77fdcc965514bb1b0e2a377832b20f0 : SUCCESS in 3m 19s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6fe2c62e2cf24458b9096746189d07ab : SUCCESS in 10m 01s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3515d409f0284dfd94b0dbdb68372d70 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3eb0ecb3c7734768831eec818107753c : SUCCESS in 50m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8df8824768664929b25325be10c06c5a : SUCCESS in 33m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3119f74405904785be539031fdeab1b1 : SUCCESS in 31m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/71b1fa85c5cc4c2398112636e58c605a : SUCCESS in 18m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3702314bacc46dfb8f0963b73d2f609 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/522dda20d840474d8433375436468ad6 : SUCCESS in 8m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8138ed0d159e418dac28dfe98bca6140 : SUCCESS in 35m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/05b1c521f6ce46048c2891e2e333741e : SUCCESS in 41m 00s","accounts_in_message":[],"_revision_number":2},{"id":"a6b24f0fdad7504eb51b5d171f502da711daec27","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 16:04:01.000000000","message":"Patch Set 2: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":2},{"id":"80623b585d5c0a387e7522e68590b0d5344cadb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-11 14:17:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"021ed03eb4608fee3615a47418ac26a57a6b35a7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 15:08:52.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/47260185b46b407bbcfb1f0756c220e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bbc7fcccd7e94c6aaf6df330977ecd02 : SUCCESS in 37m 36s","accounts_in_message":[],"_revision_number":3},{"id":"711ff0a432ab4f0c85ae7235e9fdf17ae380ee88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 15:30:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bedeca61394e4443bf5fa1a73ba810db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e0ae96a00574bc2b460ec615cabd1d5 : SUCCESS in 11m 06s\n- grenade https://zuul.opendev.org/t/openstack/build/8b943676cda24e72a8c33a4f398e36ea : SUCCESS in 54m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d05a07e6da6d4ee89e6bf5c46fed9bf4 : SUCCESS in 47m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1caa7a7d4e6c483690b6ddbc0290a07e : SUCCESS in 43m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa2729ddf69542cfbd9f7e81bed8b02c : SUCCESS in 52m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b081a7d195a428eb2c39ba06bb859de : SUCCESS in 25m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c5c704c02f443818464ea79d7a913b9 : SUCCESS in 13m 13s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c48d3ea895404280adf5655697919d0f : SUCCESS in 24m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/26a645d60b1f488ba84d5a4c3f66f4c3 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/53257d8c7c6345a6af5bc5f40057f22e : SUCCESS in 18m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7a0c30ac3b944233b4cf92c3ec0450ef : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/53bc0ca37d0e4eeabc0385c54f06b5df : SUCCESS in 26m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/368fecdb023a4658827da5d210fdc478 : SUCCESS in 15m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f32bd5e32ce443ad95b145be70de7b77 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b151c3140bf745059c5a5119413e9f34 : SUCCESS in 17m 22s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/846df579111448f68d9e0c35e3f46391 : SUCCESS in 5m 57s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7588ddc59c9145dea63d1c3fee898f36 : SUCCESS in 14m 48s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/84b1cc1efcf146a1bb7db500210d880a : SUCCESS in 7m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4008edfbeeeb4e14af0d9bf85f208436 : SUCCESS in 1h 01m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/30ec576e26a149daaccdceb3a551b08e : SUCCESS in 44m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbdeb25480d04f0f8f1605d3cf4e22ae : SUCCESS in 50m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b8196a59c8d5430b8cc275dbc9b07e06 : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6be45148cfec4b47b89d20d35163fc59 : SUCCESS in 2m 32s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5fe3f0f8b97e49aabc9d18bcb68a4441 : SUCCESS in 9m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/99cec08506a34522b98876ff5984d774 : SUCCESS in 36m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/df40e147653e4af49a14d497aadaaab1 : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":3},{"id":"e42633963abd1d12ba54791ca2ca2722cbe823a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-12 17:17:21.000000000","message":"Abandoned\n\nSquashed into some combination of https://review.opendev.org/c/openstack/swift/+/909800 and https://review.opendev.org/c/openstack/swift/+/944073 (with natural follow-through on https://review.opendev.org/c/openstack/swift/+/946141)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bfd0722e92a61c0f122ddc0c172a3be3876e2b27","revisions":{"fff996a40ad2772c28b6f4b6ea2042464991a6b6":{"kind":"REWORK","_number":1,"created":"2025-04-03 12:10:35.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/39/946239/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/946239/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/946239/1"}}},"commit":{"parents":[{"commit":"67942fd66ea6b2e9c8e3fc41a963d47e8ce29ea4","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/67942fd66ea6b2e9c8e3fc41a963d47e8ce29ea4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-03 11:04:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-03 11:55:09.000000000","tz":60},"subject":"s3api: log selected crc32c implementation","message":"s3api: log selected crc32c implementation\n\nLog the crc32c implementation that is available to s3api. If it is the\npython reference implementation then log at warning level, otherwise\ninfo.\n\nDelegate the logging to a checksum module function.\n\nChange-Id: Iff796188b88d834b957dd4c8df066b8d7ad75567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fff996a40ad2772c28b6f4b6ea2042464991a6b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fff996a40ad2772c28b6f4b6ea2042464991a6b6"}]},"branch":"refs/heads/master"},"4bb2f20b64f46aa408d1511805bc02443103c409":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-07 17:16:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/39/946239/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/946239/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/946239/2"}}},"commit":{"parents":[{"commit":"7fe0d81ed7c0a701cf4d8f5281f9ce90b8cf31a3","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7fe0d81ed7c0a701cf4d8f5281f9ce90b8cf31a3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-03 11:04:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-07 17:02:09.000000000","tz":60},"subject":"s3api: log selected crc32c implementation","message":"s3api: log selected crc32c implementation\n\nLog the crc32c implementation that is available to s3api. If it is the\npython reference implementation then log at warning level, otherwise\ninfo.\n\nDelegate the logging to a checksum module function.\n\nChange-Id: Iff796188b88d834b957dd4c8df066b8d7ad75567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bb2f20b64f46aa408d1511805bc02443103c409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4bb2f20b64f46aa408d1511805bc02443103c409"}]},"branch":"refs/heads/master"},"bfd0722e92a61c0f122ddc0c172a3be3876e2b27":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-04-11 14:17:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/39/946239/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/39/946239/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/39/946239/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/39/946239/3"}}},"commit":{"parents":[{"commit":"d4bd1c90f6eef5fbaa3eef18b8abd19eca398ce9","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d4bd1c90f6eef5fbaa3eef18b8abd19eca398ce9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-03 11:04:21.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-11 14:11:41.000000000","tz":60},"subject":"s3api: log selected crc32c implementation","message":"s3api: log selected crc32c implementation\n\nLog the crc32c implementation that is available to s3api. If it is the\npython reference implementation then log at warning level, otherwise\ninfo.\n\nDelegate the logging to a checksum module function.\n\nChange-Id: Iff796188b88d834b957dd4c8df066b8d7ad75567\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfd0722e92a61c0f122ddc0c172a3be3876e2b27"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bfd0722e92a61c0f122ddc0c172a3be3876e2b27"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
