)]}'
{"id":"openstack%2Fswift~946283","triplet_id":"openstack%2Fswift~master~I15791a9238485190acf7f87a3b562bf67e7c5c09","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-09 14:59:09.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-09 17:47:53.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-09 17:47:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I15791a9238485190acf7f87a3b562bf67e7c5c09","subject":"Add skip reason next to test name in zuul output","status":"MERGED","created":"2025-04-03 16:45:38.000000000","updated":"2025-04-09 17:49:28.000000000","submitted":"2025-04-09 17:47:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"946283","meta_rev_id":"17a69975773e494d1603d7cf497f1400bc664d6f","_number":946283,"virtual_id_number":946283,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-09 17:47:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-04-09 14:59:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-04-09 14:59:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-07 18:07:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-07 18:29:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-08 14:31:11.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-04-08 16:22:29.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-04-08 16:56:39.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6cef73072f94709a04d2e5cc0d723e2b2b02ba9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-03 16:45:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4689814d238957574d923666a2f917d84416cffa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-07 17:25:16.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"5abfcb0907e62dc63bb88d94a836e99f4f3387ba","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 18:07:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb2fdca301eb4cc6b2a4d5d54a1e0181\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab066618b6f24ba595e3f7b65b45804f : SUCCESS in 39m 58s","accounts_in_message":[],"_revision_number":2},{"id":"4f38b2f2a59f37b6d19b844f0e4f96b583153449","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 18:29:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/767dccb031c94de3a11fb83f3b02c61b\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/954800a8445e47ab8f226401cfc62fa4 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b21da0d5e154e0db0996ca5c3bc2a9d : SUCCESS in 8m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/53ee98d1f4bb40bca9f73d1c7369c9f3 : SUCCESS in 4m 56s\n- grenade https://zuul.opendev.org/t/openstack/build/596c7750072a4c86a7a1a29b57130255 : SUCCESS in 41m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/702726347fdd49fca86706434ab9c06a : SUCCESS in 32m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/65114654e1894db3afdc10d0eddbc045 : SUCCESS in 32m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01eb5af0d08c43c9ba30c9a365942abd : SUCCESS in 48m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edfd72fe6be64922b1a955d22e271739 : SUCCESS in 12m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fdcde0cfe8e489b9aed71b910e0fe46 : SUCCESS in 10m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eea1fd4f200646a6bc19ba23b6e9020c : SUCCESS in 17m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38ac27aaf9b4424e87b781190617d264 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/32d7b1f78ad444ce91ff3ccc521ff84c : SUCCESS in 15m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ad57866e31ac4e83bc9a860c9a0bb579 : SUCCESS in 18m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/332a6ddacc9241cab1005ac6c3b67d4f : SUCCESS in 23m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aaad45077d8944409c78f27e3ad24002 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/369801af3c6744f381bdf5103db140ca : SUCCESS in 14m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0810a8b0469140f7a2316abbc66ec019 : SUCCESS in 15m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5e16df9ad85e4ede8d9733522ab46ab2 : SUCCESS in 6m 06s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0081d71a95384217b139978178923cfc : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e63dbe91f520449bba4d697a9c4499e4 : SUCCESS in 7m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef0c1517f84046f2a83ef8228c5d965e : SUCCESS in 58m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e37c1ed7d6564b918952c68311caa4fa : SUCCESS in 47m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/188b764d37004c408a74c535781ae6ea : SUCCESS in 39m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/685afed522644e68b60338e9d3e028f5 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0722826242944755947b51a8fe1e17e0 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f1ff2e3414545be9627cae5dbb3d479 : SUCCESS in 22m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93d616d6082a4d6db7716858b93f518e : SUCCESS in 32m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a7ff6f1b09d94ffca3410afb0a5b0a69 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":2},{"id":"9f3a81df9e491d66141f355276306e157dab186d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-07 23:06:03.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"285eed12b3c57cb8ec7ee5196fcbdc188aac0815","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 23:53:19.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dacc611768049a89746cb35dbd07245\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2aa5df2e3e574abfa7ffc73b6002994f : SUCCESS in 45m 10s","accounts_in_message":[],"_revision_number":3},{"id":"06ee2356d1c0a96df48b8be28843bbea7b59e431","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 00:02:50.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/681156aad953428a86e7c83a8189d041\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/897fbdd79052449a955e998628727a62 : SUCCESS in 3m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/004e46225bac4d2c8f0fd4bf841f8789 : SUCCESS in 8m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee13c1b3a325451898769fedf9d922b4 : SUCCESS in 4m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/832bcd8956764ef6af44a3d05f807cd1 : SUCCESS in 41m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a644d8c0e874f33af4cae9b7911bdff : SUCCESS in 38m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/6f1971b1c5ad4b0a8de1482e8c8616ab : SUCCESS in 38m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/22ee19082e3342ec993cd181d9f7d338 : SUCCESS in 44m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a3357a6fb134e209c82368d0a09a35e : SUCCESS in 18m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ea3264df04f406c93b43fbd74931866 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d80e41b985924d5287c82fc7aea49f44 : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aebba86da9ac4773ae072de975ecac73 : SUCCESS in 6m 00s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/5252a8a243a54f27a2b3b135a54b8ffe : SUCCESS in 15m 46s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0bad19bbf9494247b81211426a5a8ba1 : SUCCESS in 17m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/56ebb468fad140ce84e56a5b3f663a09 : SUCCESS in 10m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/da2fb9036af74ae5a0ec47112175b77d : SUCCESS in 9m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c75e91b5a55e4cd7a3d070dfa4455ba3 : SUCCESS in 13m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e7104df1e30c43318ed5976ab0042bbd : SUCCESS in 10m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d796ebfdc18f4be798d8fc0d223d946c : SUCCESS in 4m 52s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dfb0032253044939b4345312fd5ba506 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/289eb945bf284d8b9de929b264afb409 : SUCCESS in 5m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2aab2011cc764b598bf7362b5a270571 : SUCCESS in 32m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c460b346366a405a94f06f83e9128d29 : SUCCESS in 30m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea29ef6f34944c45a309633449264a95 : SUCCESS in 50m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9fdf5a4cb1c547b5932fc3119c8c7bbe : SUCCESS in 20m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f705100117e4da7bdda26adf4406cce : SUCCESS in 2m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d58f1d68e14e4917a9cf0f504f53ef99 : SUCCESS in 9m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d33d1f7aac4e45a58450181ce6dbf072 : SUCCESS in 47m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/25fefe3ffbba45d88e116fd93c592756 : SUCCESS in 52m 01s","accounts_in_message":[],"_revision_number":3},{"id":"14d9d4d22a7fb9bf01ed84a017061e1ef31793be","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-08 14:31:11.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"cc502cb30e2bb7127a1bda782a1645d38cf86315","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 15:03:18.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"f9034fc2f16a7a0c3b8ed58eac79675941673c4d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 15:04:50.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"25ae1a3a3e51301a18459b561544195d53070996","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 15:45:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0d1a01380014088a1b40e095ffe0092\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7ef9c22722d54a548b063e4b96ca4a69 : SUCCESS in 40m 17s","accounts_in_message":[],"_revision_number":4},{"id":"c87e67dc48241f05b91f7848c865ad90aaeb66ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 16:12:05.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c325a2c56d7f4780b3a8704e7f18e595\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c35cd35f39284620aa08ebfd98372dad : SUCCESS in 2m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ce333c3bc8441bbb6310b36bda9311f : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4f1a7369f278418884d3408b4ff0ff2e : SUCCESS in 2m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/ed3d8a71358846cb92f5ffd661f9822d : SUCCESS in 25m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/abe648694fd14db5b1471ed35c20a2d7 : SUCCESS in 28m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3733ec9d0fb44a428a656488b905e5c7 : SUCCESS in 52m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/51dec47f47654ceea6718e13484ebf3b : SUCCESS in 31m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7e4b6b1cec34b2986b0cdcb6f652ff6 : SUCCESS in 18m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c0811a4bfc34aa4b438e1452327d823 : SUCCESS in 12m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/191eedcb050b4a58b7b7c60959dc755b : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a719f53a10b4443dba78001365786032 : POST_FAILURE in 4m 53s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2aa8e43087b34e85814ec30ac72565be : SUCCESS in 16m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/184fcee261544b79a071d919a75bd4de : SUCCESS in 17m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9bcc58e446374acfa1def8f64a25367e : SUCCESS in 16m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2e5c7c8deb4a48409a3baa175889e5e8 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81cefae54ca643bb9d0b47ba2a210906 : SUCCESS in 9m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/47644873133f48a4bc51bf20c211237f : SUCCESS in 15m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a4b9039b4a6143ba9381de71dfee90b1 : SUCCESS in 8m 27s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/160b7caa4df4409fb6b07e5f1244a443 : SUCCESS in 14m 09s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/eec1fc4992ee44c88419b07d79aac57a : SUCCESS in 3m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1920f753c77c4524a47d4c615f264046 : SUCCESS in 1h 02m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2708a67cb4dc495bb89c27ed1d340ba4 : SUCCESS in 46m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c08894fc811046b6acfa1140d1d2adc8 : SUCCESS in 17m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2211b53fa2bc4fa38d2943a5946f8514 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c41a47e26f0c42c394e373a3329b65ca : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/199577e559b7421e9a137c5b8764e4e4 : SUCCESS in 15m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/759ea85be43c401c8f7f729580e8b1ca : SUCCESS in 31m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fb55f020a7e4464ae6d6689f462f75e : SUCCESS in 27m 49s","accounts_in_message":[],"_revision_number":4},{"id":"fd60e12395d987ce429498c6062beb456645f595","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-08 16:22:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4f98537ff8cadbed087f66062fa17838bb2b0454","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-08 16:52:40.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b1d6371aecfa31069be877058d6b8ca3730577c4","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-08 16:56:39.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6281acd0f51b1813a6d2060d7e3f288d1a20dcba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 17:17:58.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"95fc333c3f1cc4c4959ff45a2d11ce8aab5fd19a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 17:20:06.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f6f47ab6d7aa7c465a78584941ea4ff92a301e59","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-08 17:35:19.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"ee5b4184e25d5b2b9f5d39207f731e5911cb1bd9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 17:54:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e32b5a1f67d4430aa079dde14353ea02\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fcaefb7325445df8905e003a442ece2 : SUCCESS in 34m 15s","accounts_in_message":[],"_revision_number":5},{"id":"4ef0fc2e50f8f4b46778b0683f36d53c3e9cb6fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 18:29:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b88500e98984b45bb9967d7c9994e74\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f983e75006194c5d8a50d76f78428972 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/029770614e754ba480469ee105aed6ef : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c211addb762e4fc299930d6e7d4fe58b : SUCCESS in 5m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/800c3bd7e317447eb14b8faa30c24d41 : SUCCESS in 57m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9fb3c46945ed432482ea729e9b43e861 : SUCCESS in 49m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4210dccb9fb441c8981a42140331f51a : SUCCESS in 49m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e5d124a2d1b414a92bc25fb92e49f60 : SUCCESS in 1h 02m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5545719a2dd9454ead8ea1c75241d07f : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3734b63d75b94bb09f76e59b21397fb8 : SUCCESS in 20m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9809a2c8e0f4e7ea4740832d1c29c24 : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/02832203363f45d6b3868b87abe9ff27 : SUCCESS in 4m 18s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/73312e077f7d4945ba2a718d5d6907ad : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a9e9f95e5574dbaab94205ad007e5f9 : SUCCESS in 16m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/94cb4f234fc94a459c53107eac9ddc89 : SUCCESS in 21m 47s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/88a3075433b94fb48b7169d50918acf4 : SUCCESS in 11m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/25df416d2f984f72ba47ad65ab90bd17 : SUCCESS in 13m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e77ac8d644c248ddbb9b22edaa831378 : SUCCESS in 12m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c1daecb0ac624fc1bd9a5ead1cb47925 : SUCCESS in 5m 20s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c37a3690fe8947ca90cead8d2c7e35aa : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ceb4513e3da640969c440506a76e9714 : SUCCESS in 6m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/be2fae73f94841938d2830b1e7c8f7c9 : SUCCESS in 48m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/12ea05f258a341aba587cfe2a51d4dab : SUCCESS in 37m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c509e0628ca744089c54fd7bdc578ee2 : SUCCESS in 36m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da8e0fb87d7c4b69930a37fd7c3b8e4c : SUCCESS in 21m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7787d2ae591d4a77a2fd81a58136b726 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/57932672857e44d693294134170106a1 : SUCCESS in 22m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/58d431d04ebf4ef3bfeadc6b9c4f721a : SUCCESS in 27m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a72be396cec64fc18c40cc8951830cf7 : SUCCESS in 26m 52s","accounts_in_message":[],"_revision_number":5},{"id":"489bd574970275c06c33d40fc981617c98dc7ef5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 19:48:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"2298fcb4fe57e11538810bfaef88f60afdb91c02","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 19:50:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b574f485be948c6232058184f5b7cac8e7bf1f0c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-08 20:15:28.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e89ca0a476e015281e49f1ad200a41684d10ab87","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 20:23:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77c73576db2a4653bb2d4af1a0f393a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6229a63fecf54f94944e401258b382df : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":6},{"id":"0f6730f3de7176f8b5a85c31d9867db5c615d58c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 20:56:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/13e8b819f9aa432fb113dd343d3f8fe9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/c4db9000158840a0983862e952f0eccc : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18f0e5ee999f469ca5b2a1eb94e74fc7 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ec734fc63fd4a2c9da4547ac9498d37 : SUCCESS in 4m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/a6c9bcc5a2704c70bf7be17754ac9925 : SUCCESS in 54m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a76623ffc85747309cbbb351133e616b : SUCCESS in 50m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca65f58f8c734d12af18054bb69f98d9 : SUCCESS in 43m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4c8de174096741bba84fb31ef6e1d4e6 : SUCCESS in 56m 15s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d794e022096349e9a321f98acd0778ef : SUCCESS in 19m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af496cd562434bb09c56b0f5eff66dc5 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5993cf887b2c4d19a4db2104d6ad50b0 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a143402d6f9d4b5195237f03688eccc1 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e456933a5b2e430cb68ee6031ecf65ee : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/480858ee9cdd48508dc41e5d8a2da08d : SUCCESS in 22m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3eb1a4056d5b485b824942ad2c09eb19 : SUCCESS in 14m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d4e318c861f748099d8a99987c00c8a2 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aa6ad9e59b6b443387ab5547f1a246f3 : SUCCESS in 12m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bceb00944f38457ca5af5bde4e3fce69 : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/13fd69cfab184004813866a3ad5d407b : SUCCESS in 5m 49s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d56875ad052a418181b7175b45e22524 : SUCCESS in 15m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/f569986ff33f4baa922aec27d43e9833 : SUCCESS in 6m 43s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/aeed6de4357743df8fb3cd5f799f3032 : SUCCESS in 50m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b9b938b074fa4f469e4db1df438e74ab : SUCCESS in 46m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2b77ba66857c4a8e89d03983c7904057 : SUCCESS in 37m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b18078a9553c402a942d6f4db59f0b69 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ab4e31cb328406fba1f3bca39f40f3c : SUCCESS in 2m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fdde10fc9cbf44a7a4ee80a1f3ef35e2 : SUCCESS in 17m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/660bbd60447c4cc8b0f2d23e3d546e23 : SUCCESS in 36m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78aeab21d0eb418d81e72c6443e58662 : SUCCESS in 34m 51s","accounts_in_message":[],"_revision_number":6},{"id":"4ff8fa0c539eb18f086a339354c78ce8346db86a","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 00:54:58.000000000","message":"Patch Set 6: Code-Review-1 Workflow-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"cbb7bdeff0424ee0d41fea4a5cac99728512b927","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 01:36:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"84de4886b558deac8fba5fe9766010e06d45e9b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 08:55:13.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":7},{"id":"22bd6ab097687be6fcd625a349370681d4019cfc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 08:57:46.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"476516b64c87efdae5f74b13efb6245d76285e19","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 09:35:10.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84e499a293f24979904bfafdd850447c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/497185d2a48946a0a00b685af0aa420e : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":7},{"id":"c537793ebee158099d0364ef67c2a5c0bcf9b301","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 09:55:01.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f24256a07ed442e38a747bc27c327b20\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e7057539e36145f695dc1684d7cefc4a : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab06ece6e6045e49feaca11fd9d9dd4 : SUCCESS in 8m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/030c3234f4fd4812bf5102ea50a3a557 : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/8a2b72a5cd174cabad9b45d6af34082f : SUCCESS in 50m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6a0970f17b474d6ab0eca45f66184989 : SUCCESS in 47m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c2dd7c16af9f4512a434c57cec098296 : SUCCESS in 31m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0a461ec02e045bdbef3870e0828e3db : SUCCESS in 50m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40a70e9711cb4927bba1210c84b498d5 : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2d48226a7bb4ecfa8266542f20c8612 : SUCCESS in 12m 21s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d83875f49c04481786d0e99c74dd13ed : SUCCESS in 11m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/326d00edb69240488855953e5d7a799e : SUCCESS in 3m 32s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6aeffaf7f2c546a3b21eab3d075e389f : SUCCESS in 7m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cc74f211d5784ee79e6e49d450cc5cd0 : SUCCESS in 20m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3053be8f866a4720acbf8e96de2c2e2e : SUCCESS in 27m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4778d8f3aaa247959d17dd1ddd549dd2 : SUCCESS in 13m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dca41fe54fbb4c1f966454295f59d63b : SUCCESS in 15m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/439b6495c093405b9e85a3b74bc34fc9 : SUCCESS in 14m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7d4a8fda440343a98b33c06a6f685430 : SUCCESS in 5m 01s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c735b7531dff44d9af971fcb71655bce : SUCCESS in 15m 27s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/9a98ef38fcdd443c8482b57cb7227c05 : SUCCESS in 5m 50s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7231fa3d15c54e959fc5db043e7b0b28 : SUCCESS in 54m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1b2b81c6077b4ec7a17860509ccc02e7 : SUCCESS in 33m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04e322b6fb6748a3bc2e9ff0d87c2499 : SUCCESS in 32m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d235f3f3dc104c3da6383c2d7e825a59 : SUCCESS in 7m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb885165a1c141a8bebf5a6e01977b8a : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ac35355a6a5e4c9c8cff0fe8c3c128e9 : SUCCESS in 21m 04s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bcf786442a14ad3a8fceaa6a9af1294 : SUCCESS in 37m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/187617e0b67d4593a7cff4984c3c68d7 : SUCCESS in 45m 43s","accounts_in_message":[],"_revision_number":7},{"id":"c8769ab87fc22917ba7122cfdb240f22b629ff2c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 10:11:20.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"330fdd701413106471779856e1f2b267647955dd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 10:13:47.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e32b86f1aa2d404fd78b3d94c36b3be49b2eadb2","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 13:38:47.000000000","message":"Patch Set 7: Code-Review+2\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"1ebe444578554c02aa4e21421258ff51dc1ee693","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 14:04:11.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"62eb03672585b884a22767a75605a1f34c7b7b94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 14:48:21.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5adf787099ad454791e1437ce2a64705\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3975d717f3c4a23af335125fd035a8c : SUCCESS in 42m 08s","accounts_in_message":[],"_revision_number":8},{"id":"1451277b39ab16fae6693db121f85ef23cb1b827","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 14:59:09.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"fc6b28fbc9798c784746e7f4ce8600a8c8208779","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:13:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aee7fb41e0dd463c831971aad4b3b339\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/7f61aaf1d8e9465ea2c74558ad4813ee : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b5bf8d6db9649809cde5a2dd4a7c8f7 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/030748a0d10e4e969607b60e75538c9f : SUCCESS in 4m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/b18866393db34de59c9855b4b395e106 : SUCCESS in 44m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7d8df5038d19429e9975acf88271c0b3 : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/110e3ad247eb46d68baf9bb131dbaf79 : SUCCESS in 27m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/127d37cc83d54fc8b9333bd542b45c89 : SUCCESS in 58m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ac5660fe2fa0458fbd1bd160c74a8f60 : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/48d33be470b94c8186356a9d6804edac : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aeec16f4c70049969db8b932a42c1b58 : SUCCESS in 11m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a6faab21402f40fca749657f9d90aa5f : SUCCESS in 7m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/92321ba544ce4597b369c083aa91e68e : SUCCESS in 15m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/add156ae4f9e4b4e93a63da7fa2fcba4 : SUCCESS in 7m 51s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a75c8429ecba4913a186a4ed1af7a521 : SUCCESS in 28m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bd451843507a4c21bbc2d1e5d766c889 : SUCCESS in 14m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/806f6b370b1d4f91b43b74ab168ad22a : SUCCESS in 19m 09s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ae56710f28f44dc3a043d26d2875f80d : SUCCESS in 14m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/055df8e348114ee192a938ba64966059 : SUCCESS in 3m 55s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5735ba532cbd498cae8e7a00b86cb540 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/e81ebfa8c7d84a70a30dc42b306bc123 : SUCCESS in 9m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de272861133146dfb6d064bfe79fc5e1 : SUCCESS in 1h 00m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/46e418b3132c4721adbbde76ff2681bd : SUCCESS in 44m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/864031255b484159a951a3d6a6b18894 : SUCCESS in 22m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/65c43239a0104a21aab93e42872418f9 : SUCCESS in 21m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c56df45acd14493a5eef918e041133f : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/11304ca6a87d42e5997b41c08134d12d : SUCCESS in 20m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/78acdd6ac58b41359e7dad5618b368dd : SUCCESS in 50m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdf559546eb841fd866c97c76b3ec2e3 : SUCCESS in 54m 03s","accounts_in_message":[],"_revision_number":8},{"id":"4b84196df045eb7504a87996c30444fd01ec2f21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:14:17.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"ba24e77c8dbab9c24dcbdab7fae59f88e51f13ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 17:47:53.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea698c7a169e436ea496f1061afead03\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/f1bd5d0ba79b4e389696a9e6b202e2ab : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b13a183b61234b58a1bdc0f46165c84c : SUCCESS in 9m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/532b1e9372b04edf93c50bc0f197e5e9 : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c0f625c65c9a42e6a5ca8746218ab422 : SUCCESS in 26m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e8af82fd9702473f8a94432581c8b169 : SUCCESS in 36m 54s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32746e6d74714614bba53c00044b9bd0 : SUCCESS in 1h 01m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21a733165f884f618f34c5ac07d73149 : SUCCESS in 57m 56s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/951e6e5526ce469994dd2144080ef82f : SUCCESS in 7m 30s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/80ab924fb66b47519ada465639192708 : SUCCESS in 19m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c3ca57194b5940f49470d6dca6f70967 : SUCCESS in 15m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/44d16d5080d84f99b0801314dbfaee57 : SUCCESS in 21m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0be1f5910574597989aa1c03860ea98 : SUCCESS in 13m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4fca7357840246d4bb640957d163cad2 : SUCCESS in 14m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cfc8c86b2f57473eba9fc2a4a8b2827d : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e19868f6c1a940d395d8f9c393c1fa84 : SUCCESS in 8m 09s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/22dae6fef47a40dda9fe7b87342d6cfb : SUCCESS in 9m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/39ce2d4951064c06a695d52164779a9e : SUCCESS in 58m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f967b82bb8aa43b1ad43755e4ff4c61a : SUCCESS in 22m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bc42d47e86ec49daa5c965aa340c689f : SUCCESS in 19m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b2147ff013e349fbad3be62b5339a718 : SUCCESS in 20m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d131926fcb7406084a777388255141e : SUCCESS in 5m 15s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42b8cb4a353146f9b240e26ff455683e : SUCCESS in 18m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/22ad019865e249f29df7c4a626eedac9 : SUCCESS in 34m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/41fc08da15f74376986c46b0d9d3d5d7 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":8},{"id":"86602e873de9fd47b35fb79ec9eae96a23e6bf2d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 17:47:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"17a69975773e494d1603d7cf497f1400bc664d6f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 17:49:28.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c60a7fdfedf471daf75026859c51a5e\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/bbf81277862c438390244878245577a0 : SUCCESS in 36s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/167cdadf28274fd3ac7c3374b49c0262 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9b92964d1da435aa41a29ace1843505 : SUCCESS in 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7f863dcc3d074cde82ef1c536016aebb : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"dd0ba770f70507cd71d8342e4b325ee50b628b9a","revisions":{"5c28efee44e41f7fecba046fc4f644ebb5920a1d":{"kind":"REWORK","_number":1,"created":"2025-04-03 16:45:38.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/946283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/1"}}},"commit":{"parents":[{"commit":"b47b1ad91b4e32d54a3f3b7c444aeb3f813b9276","subject":"sq? install awscrt","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b47b1ad91b4e32d54a3f3b7c444aeb3f813b9276"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"subject":"sq? debugging skip tests","message":"sq? debugging skip tests\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c28efee44e41f7fecba046fc4f644ebb5920a1d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c28efee44e41f7fecba046fc4f644ebb5920a1d"}]},"branch":"refs/heads/master"},"db9be344e1cb35510c13ad873996a9e6bba6c995":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-07 17:25:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/946283/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/2"}}},"commit":{"parents":[{"commit":"b1fe93466bc7336e21b53be136d0f6808da0d26c","subject":"sq? Add crt as a boto3 extra","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b1fe93466bc7336e21b53be136d0f6808da0d26c"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-07 17:25:08.000000000","tz":-420},"subject":"sq? debugging skip tests","message":"sq? debugging skip tests\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db9be344e1cb35510c13ad873996a9e6bba6c995"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/db9be344e1cb35510c13ad873996a9e6bba6c995"}]},"branch":"refs/heads/master"},"926074c3887e15b7b9da006acecadda6103b3d81":{"kind":"REWORK","_number":3,"created":"2025-04-07 23:06:03.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/946283/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/3"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-07 23:05:44.000000000","tz":-300},"subject":"Add skip reason to tox pytest output","message":"Add skip reason to tox pytest output\n\nCurrently our test names are long and we pipe pytest output to a file,\nso the skip reason gets truncated; setting this env var in tox.ini will\nallow us to see the reason why pytest is skipping next to the test name.\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/926074c3887e15b7b9da006acecadda6103b3d81"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/926074c3887e15b7b9da006acecadda6103b3d81"}]},"branch":"refs/heads/master"},"b247316b78b49ece16baa8ff445d6b5d5543d5c5":{"kind":"REWORK","_number":4,"created":"2025-04-08 15:03:18.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/946283/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/4"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 15:03:07.000000000","tz":-300},"subject":"Add skip reason to tox pytest output","message":"Add skip reason to tox pytest output\n\nCurrently our test names are long and we pipe pytest output to a file,\nso the skip reason gets truncated; setting this env var in tox.ini will\nallow us to see the reason why pytest is skipping next to the test name.\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b247316b78b49ece16baa8ff445d6b5d5543d5c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b247316b78b49ece16baa8ff445d6b5d5543d5c5"}]},"branch":"refs/heads/master"},"ca51bd7f22fc33749613d9ca330ac0fc3afcf929":{"kind":"REWORK","_number":5,"created":"2025-04-08 17:17:58.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/946283/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/5"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 17:17:10.000000000","tz":-300},"subject":"Add skip reason to tox pytest output","message":"Add skip reason to tox pytest output\n\nCurrently our test names are long and we pipe pytest output to a file,\nso the skip reason gets truncated; setting this env var .zuul and\npassing to tox.ini will allow us to see the reason why pytest is\nskipping next to the test name.\n\nDrive-by: remove passenv from s3api tests in tox.ini so they are\ninherited from testenv\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca51bd7f22fc33749613d9ca330ac0fc3afcf929"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca51bd7f22fc33749613d9ca330ac0fc3afcf929"}]},"branch":"refs/heads/master"},"b07492a0eaf6008af2bfdb8b70df05d1ce36a83b":{"kind":"REWORK","_number":6,"created":"2025-04-08 19:48:20.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/946283/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/6"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 19:47:22.000000000","tz":-300},"subject":"Add skip reason to tox pytest output","message":"Add skip reason to tox pytest output\n\nCurrently our test names are long and we pipe pytest output to a file,\nso the skip reason gets truncated; setting this env var .zuul and\npassing to tox.ini will allow us to see the reason why pytest is\nskipping next to the test name.\n\nDrive-by: remove passenv from s3api tests in tox.ini so they are\ninherited from testenv\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b07492a0eaf6008af2bfdb8b70df05d1ce36a83b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b07492a0eaf6008af2bfdb8b70df05d1ce36a83b"}]},"branch":"refs/heads/master"},"e438ac207316b49cc9f1829a244845881e0ec275":{"kind":"REWORK","_number":7,"created":"2025-04-09 08:55:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/83/946283/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/7"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 08:53:34.000000000","tz":60},"subject":"Add skip reason to tox pytest output","message":"Add skip reason to tox pytest output\n\nCurrently our test names are long and we pipe pytest output to a file,\nso the skip reason gets truncated; setting this env var .zuul and\npassing to tox.ini will allow us to see the reason why pytest is\nskipping next to the test name.\n\nDrive-by: remove passenv from s3api tests in tox.ini so they are\ninherited from testenv\n\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e438ac207316b49cc9f1829a244845881e0ec275"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e438ac207316b49cc9f1829a244845881e0ec275"}]},"branch":"refs/heads/master"},"dd0ba770f70507cd71d8342e4b325ee50b628b9a":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-04-09 14:04:11.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/83/946283/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/946283/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/946283/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/946283/8"}}},"commit":{"parents":[{"commit":"0e2791a88a9199006880bbbe31ff4e6d27c58e88","subject":"Remove deprecated statsd label_mode","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e2791a88a9199006880bbbe31ff4e6d27c58e88"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-03 16:45:13.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-09 14:04:11.000000000","tz":0},"subject":"Add skip reason next to test name in zuul output","message":"Add skip reason next to test name in zuul output\n\nCurrently our test names are long and we pipe pytest output to a file,\nso the skip reason gets truncated; setting this env var .zuul and\npassing to tox.ini will allow us to see the reason why pytest is\nskipping next to the test name.\n\nDrive-by: remove passenv from s3api tests in tox.ini so they are\ninherited from testenv\n\nRelated-Change-Id: I6b7cc64262d2f8ef7e3bbce30bab5550899b5211\nChange-Id: I15791a9238485190acf7f87a3b562bf67e7c5c09\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd0ba770f70507cd71d8342e4b325ee50b628b9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd0ba770f70507cd71d8342e4b325ee50b628b9a"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
