)]}'
{"id":"openstack%2Fswift~946335","triplet_id":"openstack%2Fswift~master~I63ed18f4a004e4d84cfbc72fcc22dbe7b8547f0e","project":"openstack/swift","branch":"master","topic":"labeled-metrics","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-09 10:31:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I63ed18f4a004e4d84cfbc72fcc22dbe7b8547f0e","subject":"tests: add LabeledStatsdClient sample_rate kwarg coverage","status":"ABANDONED","created":"2025-04-04 10:56:08.000000000","updated":"2025-04-09 10:31:15.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8cf1c2bf9bc9a977e63e87cc3346bf2354938a5a","_number":946335,"virtual_id_number":946335,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-04 11:59:05.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-04-04 15:24:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-04 11:37:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-04 11:59:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-04 15:24:49.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"049d1bb0f04f1a17011011cd1cb7066e5203f096","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 10:56:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"742218fcd770398c309c32bfa5028977729396ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-04 11:01:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c1af6a694a694ed4613c965b72aa5a859f843cd8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 11:37:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88b4a3e0113c495684464203a8d7a602\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad6d9fbf6bb14bb28fa44dac6de9e797 : SUCCESS in 34m 35s","accounts_in_message":[],"_revision_number":2},{"id":"9f3d04909bd24117de701a1658ecbb3c0c8aad6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 11:59:05.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ee75c0006564ee4bc2a7d5af7e9e8b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3365b6769c26439eb34b1bb9976bf8e5 : SUCCESS in 9m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/342e47fc05eb4f2681c9a1775534abac : SUCCESS in 51m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8714ea3808f4436eb8f585df1066f955 : SUCCESS in 21m 03s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c453edfcbd064ff2aadbd6a529881124 : SUCCESS in 13m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/14769af2dfd2420e8003140263a43ae2 : SUCCESS in 20m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e0b494475eb7451ab4436e7b73c78d4a : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/17d9a1c256a5442fab22e2a926f29a6d : SUCCESS in 14m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/62a429c12c4149a39d0a224c671affc0 : SUCCESS in 21m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4c06d98d990d4950919dc94effbd6701 : SUCCESS in 37m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9702fd5856814de0b562ea56b2bca1ae : SUCCESS in 9m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7ea01286c9c94f32b582e9e0c1de643f : SUCCESS in 13m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ee1989d06ca848febc471db0b944de80 : SUCCESS in 13m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b6b7b2efe6504fc0b24cf874e5116197 : SUCCESS in 32m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1d21712a172741cb944c6d1f416d72c9 : SUCCESS in 41m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a80c01e7fb04bf4816b9c06d0a8b9ec : SUCCESS in 20m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205e2ef1809b43c2a8748b2148a7ec64 : SUCCESS in 3m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/30f952089bdb4395a606efdfe665f137 : SUCCESS in 16m 32s","accounts_in_message":[],"_revision_number":2},{"id":"ca8f2161a3f52ead0d7d735e34a84e5590e6548e","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-04 15:24:49.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8cf1c2bf9bc9a977e63e87cc3346bf2354938a5a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 10:31:15.000000000","message":"Abandoned\n\nmerged https://review.opendev.org/c/openstack/swift/+/946310 instead","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"500bfb1c4bda6870a5722dbd354b2489897ab4a2","revisions":{"e3ab473eb97e47b913eed2cea25086bbad23e7dd":{"kind":"REWORK","_number":1,"created":"2025-04-04 10:56:08.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/946335/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/946335/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/946335/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/946335/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/946335/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/946335/1"}}},"commit":{"parents":[{"commit":"7e5235894b4f590024d572f64ab0b3ba2b02c68b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e5235894b4f590024d572f64ab0b3ba2b02c68b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-03 22:40:25.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 10:53:57.000000000","tz":60},"subject":"tests: add LabeledStatsdClient sample_rate kwarg coverage","message":"tests: add LabeledStatsdClient sample_rate kwarg coverage\n\nAdd test coverage for passing sample_rate to LabeledStatsdClient\nmethods.\n\nDrive-by: clean up and tighten some existing StatsdClient unit tests.\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I63ed18f4a004e4d84cfbc72fcc22dbe7b8547f0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3ab473eb97e47b913eed2cea25086bbad23e7dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3ab473eb97e47b913eed2cea25086bbad23e7dd"}]},"branch":"refs/heads/master"},"500bfb1c4bda6870a5722dbd354b2489897ab4a2":{"kind":"REWORK","_number":2,"created":"2025-04-04 11:01:01.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/35/946335/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/35/946335/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/946335/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/946335/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/35/946335/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/35/946335/2"}}},"commit":{"parents":[{"commit":"7e5235894b4f590024d572f64ab0b3ba2b02c68b","subject":"stats: API for native labeled metrics","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e5235894b4f590024d572f64ab0b3ba2b02c68b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-03 22:40:25.000000000","tz":-300},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-04 11:00:53.000000000","tz":60},"subject":"tests: add LabeledStatsdClient sample_rate kwarg coverage","message":"tests: add LabeledStatsdClient sample_rate kwarg coverage\n\nAdd test coverage for passing sample_rate to LabeledStatsdClient\nmethods.\n\nDrive-by: clean up and tighten some existing StatsdClient unit tests.\n\nRelated-Change: I115ffb1dc601652a979895d7944e011b951a91c1\nCo-Authored-By: Clay Gerrard \u003cclay.gerrard@gmail.com\u003e\nChange-Id: I63ed18f4a004e4d84cfbc72fcc22dbe7b8547f0e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/500bfb1c4bda6870a5722dbd354b2489897ab4a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/500bfb1c4bda6870a5722dbd354b2489897ab4a2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
