)]}'
{"id":"openstack%2Fswift~946559","triplet_id":"openstack%2Fswift~master~Ie47fa50a226da6ff7b067d6a367758ac9f12ef4c","project":"openstack/swift","branch":"master","topic":"p-statsd-tests-trivia","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-15 04:28:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie47fa50a226da6ff7b067d6a367758ac9f12ef4c","subject":"trivial: statsd tests: remove references to delegate methods","status":"MERGED","created":"2025-04-07 14:34:10.000000000","updated":"2025-05-15 04:29:29.000000000","submitted":"2025-05-15 04:28:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"946559-p-statsd-tests-trivia","meta_rev_id":"2ecc967c1cd3b7a89cbaea246e3ea8c6a554490b","_number":946559,"virtual_id_number":946559,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-15 04:28:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-05-14 16:48:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-14 14:29:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":1,"date":"2025-05-15 02:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-05-14 16:48:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},{"value":0,"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-07 15:20:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-07 15:23:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 14:29:47.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"REVIEWER"},{"updated":"2025-05-14 16:48:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-05-15 02:13:32.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"CC"},{"updated":"2025-05-15 02:13:54.000000000","updated_by":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"reviewer":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"state":"REVIEWER"}],"messages":[{"id":"ea21d0ba6aa30151c1e1025bf670e754e21fc7b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-07 14:34:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c6525e703cd406d1001abf784376aa25fd4e6e94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 15:20:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/664cfc020c6a4a1ea53e9d9f32b834bb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c9b9c56d7c643d2abec795599354a36 : SUCCESS in 43m 45s","accounts_in_message":[],"_revision_number":1},{"id":"7bf25d55a0dc2f0e7dfc70e12f095cf97b5735d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 15:23:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5db6afe1f85e4d63a427bd36a31f2ed7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77ba38c7183f46f7b52cf6327e0157be : POST_FAILURE in 10m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/be9fb93be4e54faa8934945ea79115dc : SUCCESS in 26m 59s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/93afa950e5424eceb35ed3ea2091b8f3 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2122c949b4e4d9f9cc4d8f32a427db8 : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a59da7042fc4114abc1cf15affa0196 : SUCCESS in 10m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6004f96c7b9f4b9595694c24ac32f424 : SUCCESS in 4m 57s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ab91963540f2403aa7712b029e91c62a : SUCCESS in 20m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/98ff547a21414e0081a6ca89620e36fe : SUCCESS in 22m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8eab629dd7964594a9509f17d4dc37e6 : SUCCESS in 10m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c5f00befed574a5c961d0a8521e1129c : SUCCESS in 11m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3bb95ec9ad04ede91de908bde3310f1 : SUCCESS in 9m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fd4d2de7614f498394ad6f253512dbd8 : SUCCESS in 16m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a10a10ca3fb045c6a09705d7eee869bf : SUCCESS in 47m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/20b4e864b4fb4bfb8282902809f8ec3d : SUCCESS in 43m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9ccdc0898ca7460aab5851112739ac44 : SUCCESS in 19m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b59632294f84945bf39a7c6c1b24400 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0e80498d2ef84d81bdf060f9d12430fd : SUCCESS in 15m 50s","accounts_in_message":[],"_revision_number":1},{"id":"c7aec12c90fc8afd58c634da645291a3c28c359e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 10:03:53.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f710494b068353ed232ddfb498b6c2f0fa6f7856","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:58:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3d3103001b34f958ed6f110f1779aeb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bc5f3567ece4185b933104de73bbcfd : SUCCESS in 8m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e956c4d47d584726809dbf301cedbf68 : SUCCESS in 45m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2927114c4dc64bf3bbab1e5c76034d8e : SUCCESS in 21m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c463c0b64774f6285db7f388f85e085 : SUCCESS in 10m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8eb0b1af915f480482d5910dac84a1ce : SUCCESS in 18m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ba3d05269ae640ef827feccef566a189 : SUCCESS in 5m 28s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/547107aa46a54f8c83a9b4a41bdbab4a : SUCCESS in 9m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/fd9e3d55ac674cfb9350ba1b347a5a90 : SUCCESS in 9m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9bf846421a4d434d9500fc82d6cc02ed : SUCCESS in 28m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d41394231e7b43d7a6b5f01c7e978c97 : SUCCESS in 25m 23s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a9eb85570eb1413ab90cccc4f0952508 : SUCCESS in 12m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a851c9ca576d4f78a324fcb11f396316 : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c2748342bf78469994f43b673c0b8723 : SUCCESS in 15m 04s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/078b516d8ade4e1f8995cf1d9602ed0c : SUCCESS in 49m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff83fbe66a57457a9eb2f41517cb5949 : SUCCESS in 42m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2ca342bed6564ec1bb881ce21bfa35f6 : SUCCESS in 19m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2de304b5c2ce47cdb8b2ab04f0104d00 : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/83554c9bf6304211aff32358a338da4f : SUCCESS in 20m 37s","accounts_in_message":[],"_revision_number":2},{"id":"eade38203cbf9a07eec2e686924848aff7e795f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 11:29:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05330fed302a46c99a720af08702b1fe\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a248c7cbc7c1483ba9ddba2b8cd0aba8 : SUCCESS in 1h 16m 58s","accounts_in_message":[],"_revision_number":2},{"id":"81ec65c849e7c8a6dcce293fa13320a95704c898","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-14 14:29:47.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"1cf2b1ee3216f4fdac27ec39c40341acd6d6a879","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-05-14 16:48:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"856a1de146ae67abbe3242b1c98907e0ef00597f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 16:48:45.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"19327471fa99778969fc34ddec15f92c71d554ac","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:52:15.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/61546b20c973455caf21bbd3c12cae49\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45f571e748b14c24a2268c331ca66417 : SUCCESS in 10m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/67eed5d065d74ccbaad242e2a947da8e : TIMED_OUT in 3h 02m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f2b6aef8c5b94b84ba9b2dd8e286fb3b : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7682bf859b4a40e2b184a30cf27b853c : SUCCESS in 21m 05s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2eaca3cd96474a4f8ea2e23096decb94 : SUCCESS in 15m 37s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6d794583404a4e7698c615e7528895df : SUCCESS in 26m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0d8e8f25b0aa47db8526ca2db4d8d046 : SUCCESS in 14m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/227aa0f72c1745be906fa4c23a451bd2 : SUCCESS in 14m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8e153d367f6b417cadff43a157f1145f : SUCCESS in 17m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5132a887cb3240a98872ecc6dd9b82c8 : SUCCESS in 6m 09s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/85b76241ad6e4830945fa7869f737555 : SUCCESS in 7m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/309dd6546c1149a2a8e4c93a6cb80c8c : SUCCESS in 23m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c450c16a43b247a0bda70f6aefa59313 : SUCCESS in 40m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d8d100d2c514f4eaf3954ff5fcb3aac : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ff6fec6313347bfb744e2c5408f32b8 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/101379580fbd4e858f76ec09eed2c898 : SUCCESS in 22m 57s","accounts_in_message":[],"_revision_number":2},{"id":"fda4f70b54ffecce81e618ebc859087cd884c458","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-15 02:13:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ba727da140636da55965cdd8c018b8e4d5a62a5d","author":{"_account_id":35790,"name":"Shreeya Deshpande","email":"shreeyad@nvidia.com","username":"shreeyad"},"date":"2025-05-15 02:13:54.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"fd23d93015858e25722d20e0dde5ffc955cca701","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:29:57.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97af0ebcd6aa4b6986c86a246e952b38\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81d973f0703a4d57aaca27498ecdf23d : SUCCESS in 4m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2efc2c4a7f94fe19f5adf1137ac7341 : SUCCESS in 1h 00m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/751f900dd90e41a9b221c918003fc259 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77a8005aec554118aed95256d61b82ac : SUCCESS in 17m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79844986c1de4ac6bffaaaf2dc56a17e : SUCCESS in 21m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1996fd088b064960b6c53230134f6d1a : SUCCESS in 5m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6a8b607216384af3af9285c48e76e29a : SUCCESS in 21m 34s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0dc44eebd1fb42df86a00d2f60fc5c9c : SUCCESS in 21m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/17d7c5fc0e654ecbb1ae715ad29e131a : SUCCESS in 18m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/aba32e76e1724530a5d05df65866c8a4 : SUCCESS in 18m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d3d64d126df34ede82eeff0a733e9000 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6d1de9f09acb4662a5dfec6fba5379b2 : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/007619f68888431dbfc9d50bd76984c7 : SUCCESS in 15m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3c337ae098e94768814608300a56703c : SUCCESS in 42m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40a6fb4ed5324524910045eeac8c835b : SUCCESS in 50m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf377e53bc1d4254b34de1b8900b8568 : SUCCESS in 23m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bca63183e33432c88d0f7b134ae1e4b : SUCCESS in 2m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4f95b65daa154e31b125f069cf6dd870 : SUCCESS in 19m 09s","accounts_in_message":[],"_revision_number":2},{"id":"74df66c6a94516ca6715dbc1b63a33c4b4ef8399","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:30:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d563bcaa4d0b745502b5720c759b810b6c3ef94","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 03:42:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e494ab45a2b4a4d8f56d0a063a88e0c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bfe0d072b86b46f680c32f8fc5cf954d : SUCCESS in 1h 14m 10s","accounts_in_message":[],"_revision_number":2},{"id":"dcf790ecc2fe11d16ba5d555aa9166d17daeffa2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:28:12.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b40551d65e4c25979e982095115ed5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/805ba0aaa40643c7a2623902f8cf9b17 : SUCCESS in 9m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5bfe3cbda26e4e93b6a6fad9ed59d835 : SUCCESS in 57m 49s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/981baa6d966b41fe80d92c43e18ddcc8 : SUCCESS in 4m 13s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1bcd7af82dda4fd7a9be1011ad95f606 : SUCCESS in 8m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/94609f69dd634d73968fb2d067335394 : SUCCESS in 17m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/92726283d35548ac8d9b9d71470f88f9 : SUCCESS in 27m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a3ec569cb014f7bb8ecae0956a23edf : SUCCESS in 7m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/26d98345b5dc4629b6712d8841a5fbab : SUCCESS in 8m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/61f8ff883e15479dad5580798e5d3568 : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/10f1bb7cad874c74b5669fb9cea836ee : SUCCESS in 8m 06s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5d29ed182f1f4a0cb51b02171a628c0e : SUCCESS in 6m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2aef351d601d45ffba00dd2cf88786f3 : SUCCESS in 48m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/98050fa9374c4b6d99e75798176f7ce5 : SUCCESS in 42m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/beaee3b95b004c9690e907c290f55cbb : SUCCESS in 16m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52249d048489468f9a61b208d009d255 : SUCCESS in 2m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8628f79779c649e5a29ff4bb57fd9aaf : SUCCESS in 19m 47s","accounts_in_message":[],"_revision_number":2},{"id":"eea0377ffe2bd69e39bcb3abd1e08a155bd35352","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:28:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"2ecc967c1cd3b7a89cbaea246e3ea8c6a554490b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:29:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4b2686d99c343d29cda74e15d74c21a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/899b6477feb141d6ac8243dc873acaeb : SUCCESS in 56s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/81834ef0cc0b4137b4e936713bfb4bde : SUCCESS in 1m 07s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"020b5a78a537759f1fb99e6b00dc43fe17aeb9c0","revisions":{"f79b0d4476cdac3d53c72b4326b7771a488d24f8":{"kind":"REWORK","_number":1,"created":"2025-04-07 14:34:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/946559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/946559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/946559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/946559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/946559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/946559/1"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-07 14:23:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-07 14:23:20.000000000","tz":60},"subject":"trivial: statsd tests: remove references to delegate methods","message":"trivial: statsd tests: remove references to delegate methods\n\nSome tests previously made assertions on statsd client methods that\nwere implemented by a delegated interface patched onto a logger\ninstance. Those tests have run against a StatsdClient since the\nRelated Change, so the use of the term \u0027delegate\u0027 in tests no longer\nmakes sense.\n\nChange-Id: Ie47fa50a226da6ff7b067d6a367758ac9f12ef4c\nRelated-Change: I3a677bb67c5700da48f89c847f652b4610ba47c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f79b0d4476cdac3d53c72b4326b7771a488d24f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f79b0d4476cdac3d53c72b4326b7771a488d24f8"}]},"branch":"refs/heads/master"},"020b5a78a537759f1fb99e6b00dc43fe17aeb9c0":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-14 10:03:53.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/59/946559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/59/946559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/946559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/946559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/59/946559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/59/946559/2"}}},"commit":{"parents":[{"commit":"c12aa812c9f434d4584e65d48c8f77b6c8c37862","subject":"Merge \"Drop support for py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12aa812c9f434d4584e65d48c8f77b6c8c37862"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-07 14:23:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 10:03:44.000000000","tz":60},"subject":"trivial: statsd tests: remove references to delegate methods","message":"trivial: statsd tests: remove references to delegate methods\n\nSome tests previously made assertions on statsd client methods that\nwere implemented by a delegated interface patched onto a logger\ninstance. Those tests have run against a StatsdClient since the\nRelated Change, so the use of the term \u0027delegate\u0027 in tests no longer\nmakes sense.\n\nChange-Id: Ie47fa50a226da6ff7b067d6a367758ac9f12ef4c\nRelated-Change: I3a677bb67c5700da48f89c847f652b4610ba47c2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020b5a78a537759f1fb99e6b00dc43fe17aeb9c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/020b5a78a537759f1fb99e6b00dc43fe17aeb9c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
