)]}'
{"id":"openstack%2Fswift~946577","triplet_id":"openstack%2Fswift~master~Ib14d6eb653e1fb4129739c52dd7ead99e6c3f660","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Ib14d6eb653e1fb4129739c52dd7ead99e6c3f660","subject":"WIP: only use slow crc ref funcs in tests","status":"ABANDONED","created":"2025-04-07 17:47:27.000000000","updated":"2025-04-12 17:15:24.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a6ea1e7a98d890b9f3a2896770fc41892b479721","_number":946577,"virtual_id_number":946577,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-07 18:51:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-07 18:31:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-07 18:51:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1625ca3f5583eba43b6c89033b8aab00952c678b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-07 17:47:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7660c53fe013b7866336f6cdbb9f23900dbf1bf1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 18:31:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e38c0961b04448f81ba67de8803e8d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4edbc2ca4a25441ba33c7194506f66e2 : SUCCESS in 41m 46s","accounts_in_message":[],"_revision_number":1},{"id":"1afed053758bb73dff0fac722733e6632d5e6092","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 18:51:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f903bef0b6aa45e18859fda9c3aceb6a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e3703593621477b9db011f3b2abc918 : SUCCESS in 9m 55s\n- grenade https://zuul.opendev.org/t/openstack/build/5587aab31a0144abbb2b7ae69df77eaf : SUCCESS in 57m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1dfc01590f5349b2a4e278d11aabc7f5 : SUCCESS in 56m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e140f2b9e13c4680961d8bf0bdb0c743 : SUCCESS in 43m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ba15145064f4b77b2b8bb4d88efa3f6 : SUCCESS in 1h 02m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b5ebdf3d1414c4b808091e765eeb1b3 : SUCCESS in 18m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95ee8e09be144a728b896bf8e90f4850 : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9901a17dcb145008bfd3f341938d2f5 : SUCCESS in 20m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3fa87b108e30496d93f253e1d7cfb0d8 : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/07b754e001da4d06b17399e448623f51 : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b1e502be98c947f987a26ff5fe224f4f : SUCCESS in 16m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2e3c0ba123214f77a8eaf76e52213453 : SUCCESS in 18m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2cdd5672e63c4311825d761394fe5aa5 : SUCCESS in 14m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/73179c3b1ae141b6a867de26b7110aa3 : SUCCESS in 14m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/928c70d3f6ab4de18037d7cb4897c4f2 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/877316cd403f4b86b02c8746b30fe163 : SUCCESS in 5m 08s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/47370d287119498383f4a718880367da : SUCCESS in 13m 59s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee411f1d8ad84dceb0a7a7bf05d061ad : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a27e67d900db415880fe11e81c4b52b7 : SUCCESS in 52m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b37741013e8447b4a37f71d21b594482 : SUCCESS in 47m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/577f78e12a2c4d0f946c5a44fe31e3ac : SUCCESS in 17m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f06cf4858d51488587881faa48faa9ee : SUCCESS in 10m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce15cd2f322641c7b9beb153e56e302a : SUCCESS in 4m 01s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e163eae569e74d639e289f939533367e : SUCCESS in 22m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/788e3c584e3c4d18ae02f27bf4d8aa4b : SUCCESS in 50m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/62c6d8f5f0cb416d9a92fc19c4be8ffb : SUCCESS in 26m 32s","accounts_in_message":[],"_revision_number":1},{"id":"a6ea1e7a98d890b9f3a2896770fc41892b479721","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-12 17:15:24.000000000","message":"Abandoned\n\nPursuing the approach in https://review.opendev.org/c/openstack/swift/+/947011 instead.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e7f9aa8e90a592defb1328c6ccabf5e223ad397e","revisions":{"e7f9aa8e90a592defb1328c6ccabf5e223ad397e":{"kind":"REWORK","_number":1,"created":"2025-04-07 17:47:27.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/77/946577/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/77/946577/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/946577/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/946577/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/77/946577/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/77/946577/1"}}},"commit":{"parents":[{"commit":"39b8e56ab2bbdc24ad152ac2cf5fc2dddc88fe76","subject":"Prefer anycrc to pure-Python CRCs if available","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/39b8e56ab2bbdc24ad152ac2cf5fc2dddc88fe76"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-07 17:45:03.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-07 17:46:54.000000000","tz":60},"subject":"WIP: only use slow crc ref funcs in tests","message":"WIP: only use slow crc ref funcs in tests\n\nTODO:\n- maybe there\u0027s a better way?\n- add a test that s3Request will raise S3NotImplemented\n  if there\u0027s no isal or anycrc crc64nvme available.\n\nChange-Id: Ib14d6eb653e1fb4129739c52dd7ead99e6c3f660\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7f9aa8e90a592defb1328c6ccabf5e223ad397e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e7f9aa8e90a592defb1328c6ccabf5e223ad397e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
