)]}'
{"id":"openstack%2Fswift~946706","triplet_id":"openstack%2Fswift~master~Ic1c948e909c82c9c96994c64a844a011292cf8e8","project":"openstack/swift","branch":"master","topic":"labeled-metrics","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-09 10:42:44.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-08 20:21:15.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic1c948e909c82c9c96994c64a844a011292cf8e8","subject":"Apply a common pattern for abstract common tests","status":"NEW","created":"2025-04-08 19:27:55.000000000","updated":"2025-04-09 10:42:44.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3d4ab531c0a64f858b454a4019741f7bcd74b63d","_number":946706,"virtual_id_number":946706,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-08 22:29:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 20:09:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-08 20:21:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 10:42:44.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"6f7befb79a6f4e57423dc2d17a26aa0cd942ff97","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 19:27:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"007d16657c6d1aab568b37953682cea204053c0c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 20:09:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3152380c4261476ab6ed6db0b4dc28a2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2dca62aee91a46f28a75d0312ac66ac8 : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":1},{"id":"24aae0b8934790b0aeb3faf03b9b999523d60b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 20:21:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83afee4b678a4a17bf5acd8980877e3d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ea1c7a7e7b448a7b7484d7b885e160e : SUCCESS in 10m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3e354fc59cb549b0921797b8df17b19a : SUCCESS in 46m 32s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f9cbfceb692418280caba6be23a4c9d : FAILURE in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/28b7bdff39f342f8ab7058097e1ec119 : FAILURE in 13m 09s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5850fefdab294f64b12a7ff1ca69be5b : FAILURE in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/815b1c2835514cfb8bd90b151547ba72 : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/60169afb05684129aea9e7460adfc942 : SUCCESS in 17m 54s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d096aab6f15943eb8af63d931cdeb99a : SUCCESS in 8m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/33bdc79de0e94c979cca9e138a518247 : SUCCESS in 31m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e7285cfabe54649a37fd9b1c98ef36b : FAILURE in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/69e6ab9070254101a18d5f743389cf6a : FAILURE in 12m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d562ad2a8d384ad6a79732a1af89011a : FAILURE in 13m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/186f437e82b04bdf9ae7a9c59cc864a1 : SUCCESS in 5m 49s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/2a2307569c6c4e3c8beb4f128c7bc483 : SUCCESS in 5m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cf27f8ff9544c0ca7cc11fbe7c7b205 : SUCCESS in 49m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5550cf7564074a6eaca5801f3de3cc61 : FAILURE in 26m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee8030363af04a32aa892b28d29e9a44 : FAILURE in 23m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/03f9771c3d6d49da9007adb1035ffede : SUCCESS in 18m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2acf91d2de3a44a9808662822d2cc3fe : FAILURE in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1cbf9dcecb847a28390171ad4c2c6ce : SUCCESS in 19m 39s","accounts_in_message":[],"_revision_number":1},{"id":"7cadb92d29b196a547d0e16bdee6132fa4eff405","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-08 21:27:48.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"afd443e00fa8cc456f02a83b5ae928dd6daf6ac1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 22:14:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aabd8d977a148dfa8f9f379d166950f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/10e2cd1205b141d68096bfb552fbe33c : SUCCESS in 45m 22s","accounts_in_message":[],"_revision_number":2},{"id":"5720ffa6a86bc12fc6e5d54dcdd797fcd4163eff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 22:29:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c7e21babbaf4491a6816e331d9fa08b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3022fd15769e43f482dada9811ec7b9f : SUCCESS in 9m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a849510c68a946ce82a33b9c6038fc34 : SUCCESS in 54m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/954a39ce31c841efa954e344861cf210 : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c7a17f662839486dae315e3563c0bb4b : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b045c76cb5d84a0d97410c8b8a0637f0 : SUCCESS in 21m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64bfce095e5347deb1f9fa1b20730621 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/be3904b0713b4c05a21d1197e7d834a3 : SUCCESS in 18m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b867205703cd46d2b971c53eff9c9c33 : SUCCESS in 23m 11s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f0cd49a975f41f5a181911e42211771 : SUCCESS in 10m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ab2a823d6a204f8fa4df9951431756a4 : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8f97df21f1d64fc799d426f752f5bd81 : SUCCESS in 13m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/44381aebc7b7416f8e2a13489ac501cc : SUCCESS in 14m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/77cd44affcc543a9a1fc2efe85bab9a4 : SUCCESS in 6m 10s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/adc13d204a3642ae87bb494399d5ac09 : SUCCESS in 5m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9c88c4e901fe4033af408cd9b078a6f8 : SUCCESS in 56m 20s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ad83094976b54c06baec4ade30318f5c : SUCCESS in 47m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a187c853fba4074941a750d47bc85ca : SUCCESS in 30m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3fb785bf35004d0ab735b46eaac20257 : SUCCESS in 20m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf503c6912d54392934a25e1946f7871 : SUCCESS in 3m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6e868c16e9742a48a0556881f1dd01a : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bb0a6a76110bf55c82bc090966a651cad7305fc7","revisions":{"7a9db9234a719aaf7eef71421d1a45c9a7e1b382":{"kind":"REWORK","_number":1,"created":"2025-04-08 19:27:55.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/946706/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/946706/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/946706/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/946706/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/946706/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/946706/1"}}},"commit":{"parents":[{"commit":"ed0ef357135b0baadfabaa67774f00bfd6a7ebd4","subject":"tests: add LabeledStatsdClient sample_rate kwarg coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed0ef357135b0baadfabaa67774f00bfd6a7ebd4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 17:30:16.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 19:24:57.000000000","tz":-300},"subject":"Apply a common pattern for abstract common tests","message":"Apply a common pattern for abstract common tests\n\n... with a helper and a doc-string in test/__init__.py\n\nN.B. there is probably more places we could apply this pattern if it\u0027s\nhelpful and we want to invest.\n\nChange-Id: Ic1c948e909c82c9c96994c64a844a011292cf8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a9db9234a719aaf7eef71421d1a45c9a7e1b382"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a9db9234a719aaf7eef71421d1a45c9a7e1b382"}]},"branch":"refs/heads/master"},"bb0a6a76110bf55c82bc090966a651cad7305fc7":{"kind":"REWORK","_number":2,"created":"2025-04-08 21:27:48.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/06/946706/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/06/946706/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/946706/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/946706/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/06/946706/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/06/946706/2"}}},"commit":{"parents":[{"commit":"ed0ef357135b0baadfabaa67774f00bfd6a7ebd4","subject":"tests: add LabeledStatsdClient sample_rate kwarg coverage","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ed0ef357135b0baadfabaa67774f00bfd6a7ebd4"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 17:30:16.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-04-08 21:27:25.000000000","tz":-300},"subject":"Apply a common pattern for abstract common tests","message":"Apply a common pattern for abstract common tests\n\n... with a helper and a doc-string in test/__init__.py\n\nN.B. there is probably more places we could apply this pattern if it\u0027s\nhelpful and we want to invest.\n\nChange-Id: Ic1c948e909c82c9c96994c64a844a011292cf8e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb0a6a76110bf55c82bc090966a651cad7305fc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bb0a6a76110bf55c82bc090966a651cad7305fc7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
