)]}'
{"id":"openstack%2Fswift~946710","triplet_id":"openstack%2Fswift~master~I6b7cc64262d2f8ef7e3bbce30bab5550899b5211","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-09 15:15:53.000000000","reason":"Change was submitted"},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-09 14:03:06.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-09 15:15:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6b7cc64262d2f8ef7e3bbce30bab5550899b5211","subject":"tests: Display extra test summary info","status":"MERGED","created":"2025-04-08 19:59:37.000000000","updated":"2025-04-09 17:54:38.000000000","submitted":"2025-04-09 15:15:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"946710","meta_rev_id":"d23cf9a38f1e0ffd677d3a83183c4a34af835756","_number":946710,"virtual_id_number":946710,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-09 15:15:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-04-09 09:00:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2025-04-09 14:03:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2025-04-09 14:03:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-08 20:40:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-08 21:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-09 00:54:10.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-04-09 09:00:20.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"c6a1b5150f405a17135c6afe51ea6668e17d558a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-08 19:59:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a626b75bafee7f5d955524dcb64c696ffba3674e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 20:40:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a71c3672b2294179a3c855a34526e665\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd912a67b8ed4c3a8e50ab263cd8b2a6 : SUCCESS in 39m 04s","accounts_in_message":[],"_revision_number":1},{"id":"a9145d8823260fce66f9e51c09c9b5a24c1f7ea3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-08 21:13:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e51c7295ceb140a698b86f7f68cdff76\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/e14085bf1f714206b928ef4e0c4802bb : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45109eb58e81403d940a9efbafc76163 : SUCCESS in 9m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2e03e7d752b4c9d8e2b84659ececb1a : SUCCESS in 2m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/7b8ccb56cf454a378cc698a13f9dfa3e : SUCCESS in 47m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ee1c7923520845d5ae03f19c97e388c6 : SUCCESS in 28m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/606b19bbc1a645cdb3599c62833e6e12 : SUCCESS in 47m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9952a450caca49dd9e0ac6a6fb2eaf19 : SUCCESS in 1h 05m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eca789457fe440be84378b2a1b599381 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dbf67eee7e134716994755e1f4948071 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea62861d268c4b1d975e001eec91570e : SUCCESS in 12m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c07d7bc2b004ddfa87ff18d0fb1fb98 : SUCCESS in 5m 06s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/2fdba4e8805342a4b3837c62b19224ac : SUCCESS in 18m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e2b00b8f108d43599bc3c83387916d74 : SUCCESS in 17m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4b43fcb01ceb4616b5bd093994bad8ef : SUCCESS in 28m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3b962d8feefb4e458a8347cce0c17cd7 : SUCCESS in 13m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5ab830b0abe64617a60a8ab9ec95e07f : SUCCESS in 12m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8d018b66437948349f4732787aa5c841 : SUCCESS in 15m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5fcf06eafafa406da6c7071affb0c585 : SUCCESS in 5m 25s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a292a413f6b34fba99d544d91f8bafbc : SUCCESS in 13m 07s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/74d89d51887142bc8dcb75244230d334 : SUCCESS in 6m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b32429e727f84e6aafd1c29f49a44a5b : SUCCESS in 55m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6ffe5f218ede4d3999fde89708448d88 : SUCCESS in 50m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/df11bcdbf4444bbda8e6b276d6146bf0 : SUCCESS in 39m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/32da92f266b443279418a4c5c71a64eb : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42873e4b9fb54271ba46c5b13dae1cf6 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de4c049611c6408aa5e397c6ab604de6 : SUCCESS in 21m 42s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bc485050f0f54cdcb92958623a41f41e : SUCCESS in 51m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c27bb6acc07846298b894ff2ba2d53ca : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":1},{"id":"3f08fc7ff4b4847114508dae6535fd6834d17a53","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 00:54:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8abb440d72060b2ab094b63db7b88f3b0bfa9e96","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-09 02:44:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d74a9eb10232a226b8483606ee7c49b4588f8e75","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 09:00:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"8ed18ad24fa2a4e5427308022692a2be649562c5","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-09 14:03:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"b0c8f9ce6543abffce804469ae51546ffcb414e5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 14:03:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3bd5c2f6302354b09b8552854a6539163a6dded6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:15:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e596a58aee834e56982cbd71d71c2c5e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d84c73b3aa5045ef910782902f7c0d29 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc5baa4e2e514b88aa282840122f9d35 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a05c5343cefc4fd1bab754725f2664b2 : SUCCESS in 4m 57s\n- grenade https://zuul.opendev.org/t/openstack/build/e52468975aa248909ca1b72f7cb83447 : SUCCESS in 55m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/528aca73222d43898a22c5823a609d5f : SUCCESS in 57m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d209f658d1654c8f9bc77e01ae6e8f73 : SUCCESS in 53m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d67c336bb8954bd3a9a6bfad12411bc9 : SUCCESS in 53m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/eb6425a6452945f084a7426fd75cf288 : SUCCESS in 3m 52s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d33831f6905b4db1ab6fc7681727e338 : SUCCESS in 17m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b66329ab322341bc905b7b81c5f64cc2 : SUCCESS in 19m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9690ab28424b4f2a99adfb3efa587ed4 : SUCCESS in 28m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/050106697f094344bab6ad4fbd59c22d : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/00c503e3eb9743e88a32659593fd2008 : SUCCESS in 14m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/349d74d474f94a819ca770fa1bf0d881 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7a3a574be79840e982bf0b6be44a4446 : SUCCESS in 5m 45s\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/3a426471e4774bacb2d8b16f3d89901e : SUCCESS in 7m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cc156c9726c48b086ecca8119e0ec17 : SUCCESS in 1h 01m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1fd7b3d6bee94a6aa40f81a0658d6607 : SUCCESS in 45m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/40fa21fe79654000aad746c74ea8578e : SUCCESS in 18m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b4abed8c93424c9c8cf09ef3641bccad : SUCCESS in 22m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c808aefa142e4183bccd873a624278ae : SUCCESS in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8c454d4d7070458ca7fee5dcb825440b : SUCCESS in 21m 25s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2b7148fa36ff4268ba7bdc3736161fbd : SUCCESS in 36m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45e474f0622e49f897e0e3cd02e66044 : SUCCESS in 56m 53s","accounts_in_message":[],"_revision_number":1},{"id":"4d26846d0dbba2f1e661cbeaab2e1b74f76224a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:15:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"622fa51e178d5df96224883cfc96724aba1a6ef1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 15:17:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a8b8a1fbb6644c3a0030e3f44474544\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/67d7d205baaa4ecda52d2e349c98f754 : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e426374c990941e99b00f176214192c3 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85062efb868140069a129e103a134ba8 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/dad5c5269f7e4ce7a29000ec8dcd4d44 : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d23cf9a38f1e0ffd677d3a83183c4a34af835756","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-09 17:54:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"405440a2ee1f4c8ddda1651c7aaf070232fecd7c","revisions":{"405440a2ee1f4c8ddda1651c7aaf070232fecd7c":{"kind":"REWORK","_number":1,"created":"2025-04-08 19:59:37.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/10/946710/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/10/946710/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/946710/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/946710/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/10/946710/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/10/946710/1"}}},"commit":{"parents":[{"commit":"769e482ea862e6416f3eff893da7edf18e36f6cb","subject":"tests: provide FakeLabeledStatsdClient for debug","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/769e482ea862e6416f3eff893da7edf18e36f6cb"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-08 19:59:29.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-08 19:59:29.000000000","tz":-420},"subject":"tests: Display extra test summary info","message":"tests: Display extra test summary info\n\nChange-Id: I6b7cc64262d2f8ef7e3bbce30bab5550899b5211\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/405440a2ee1f4c8ddda1651c7aaf070232fecd7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/405440a2ee1f4c8ddda1651c7aaf070232fecd7c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
