)]}'
{"id":"openstack%2Fswift~946762","triplet_id":"openstack%2Fswift~master~I4a276069ed95d0e6baf12cbc3f52c94c5ff137b9","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_object","hashtags":[],"change_id":"I4a276069ed95d0e6baf12cbc3f52c94c5ff137b9","subject":"tests: add more tests for object-server labeled metrics","status":"ABANDONED","created":"2025-04-09 12:24:54.000000000","updated":"2025-04-17 15:09:12.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"3796beee3e2a4828852fbd0c1562d81f64014419","_number":946762,"virtual_id_number":946762,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-09 13:31:59.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-09 13:15:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-09 13:31:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe57fb2c4dff48bddc97ff33f9839c5c33b03b27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 12:24:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"496e4fb101de2573d0dd6a7a7cf89e0ebf1fe06e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-09 12:29:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"73156a9016a8d4ec52d1d99bf4ae81afb1308a72","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 13:15:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4a68ee2d7c84c1cb7620972ac97af17\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/359fa400950e4320b43d3f31c062eac8 : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":2},{"id":"c241414e8d773d482b970659d444895172eea598","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 13:31:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b3e86334235447b9e88b00390b90ccb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/580d00c6c06942d0920f30d8a35c98ef : SUCCESS in 9m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c156fd809284b00981a3f9921a61c95 : SUCCESS in 55m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b9dd02da09544838f4ed8d02f177a89 : SUCCESS in 16m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df8e4c89d7de4392a8e52a6960f6aca5 : SUCCESS in 20m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/721d12d618d245c8b1a3e5e46917bac6 : SUCCESS in 21m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/513210af5fb548c88cd6c50ea62ecd41 : SUCCESS in 3m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8bd801c329d6402b81337bc3df613385 : SUCCESS in 16m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1c1c45c7dade4db0b7076503f1b01b55 : SUCCESS in 17m 09s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/63f18537f4f544319c00e8acff109ed3 : SUCCESS in 27m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c9bfb2caaa794a5abc2bac458fc19e6a : SUCCESS in 16m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d3ed52f7d0bd4f059c6fdc5c3e160121 : SUCCESS in 14m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/62523b9aab374cbe968d7cbb84358d9e : SUCCESS in 15m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/85d552f54b1740da88d3c5a8bcbde5a9 : SUCCESS in 30m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/325eafec792b4f1895b4a8b8547312c5 : SUCCESS in 33m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e4ff02c0a0240eda451e5e46b6e6bc8 : SUCCESS in 18m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/75eff40f27f841e6ad74fb09a981e7b6 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b5391797698647aa93a357e9f029368e : SUCCESS in 8m 43s","accounts_in_message":[],"_revision_number":2},{"id":"3796beee3e2a4828852fbd0c1562d81f64014419","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 15:09:12.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"976ba2a1481d78265901f4f6e96aa443be56c6c4","revisions":{"ced1b69adbb04b6f5c05656a901a1bfd749ef0e1":{"kind":"REWORK","_number":1,"created":"2025-04-09 12:24:54.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/946762/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/946762/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/946762/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/946762/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/946762/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/946762/1"}}},"commit":{"parents":[{"commit":"5286ab2e0004136fd372d2b720804b8c19f50ff3","subject":"Object-server: add labeled timing metrics for object REPLICATE request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5286ab2e0004136fd372d2b720804b8c19f50ff3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 12:23:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 12:24:22.000000000","tz":60},"subject":"tests: add more tests for object-server labeled metrics","message":"tests: add more tests for object-server labeled metrics\n\nRelated-Change: I155572bc5f4d955c918c951d6d2af9dbbae0136d\nChange-Id: I4a276069ed95d0e6baf12cbc3f52c94c5ff137b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ced1b69adbb04b6f5c05656a901a1bfd749ef0e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ced1b69adbb04b6f5c05656a901a1bfd749ef0e1"}]},"branch":"refs/heads/master"},"976ba2a1481d78265901f4f6e96aa443be56c6c4":{"kind":"REWORK","_number":2,"created":"2025-04-09 12:29:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/946762/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/946762/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/946762/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/946762/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/946762/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/946762/2"}}},"commit":{"parents":[{"commit":"5286ab2e0004136fd372d2b720804b8c19f50ff3","subject":"Object-server: add labeled timing metrics for object REPLICATE request","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5286ab2e0004136fd372d2b720804b8c19f50ff3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 12:23:35.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-09 12:28:57.000000000","tz":60},"subject":"tests: add more tests for object-server labeled metrics","message":"tests: add more tests for object-server labeled metrics\n\nRelated-Change: I155572bc5f4d955c918c951d6d2af9dbbae0136d\nChange-Id: I4a276069ed95d0e6baf12cbc3f52c94c5ff137b9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/976ba2a1481d78265901f4f6e96aa443be56c6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/976ba2a1481d78265901f4f6e96aa443be56c6c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
