)]}'
{"id":"openstack%2Fswift~947263","triplet_id":"openstack%2Fswift~master~Ib0330e863ad15e362dfa587517a26109c74eefec","project":"openstack/swift","branch":"master","topic":"p-s3request-translate-read-errors","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-04-15 17:45:31.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-17 09:47:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0330e863ad15e362dfa587517a26109c74eefec","subject":"s3request: refactor error handling while reading input","status":"MERGED","created":"2025-04-15 15:30:02.000000000","updated":"2025-04-17 09:48:34.000000000","submitted":"2025-04-17 09:47:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"947263-p-s3request-translate-read-errors","meta_rev_id":"4c697f97ed56c27a1598672ac15ab1cef8973f47","_number":947263,"virtual_id_number":947263,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-17 09:47:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-04-17 08:41:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-15 16:16:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-04-17 08:41:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 16:16:01.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-04-15 16:18:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 16:34:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b70c706e1a6598be749f7bb96795536439aa1e0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-15 15:30:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"126a7eeef9386c27b8c6942dcd000f4b7443efb4","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-15 16:16:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d895f4aab6d3e12448512a9445d978d451e1c74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 16:18:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0defa9bfa3e04c1aafd38fb97a56a4ef\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2855939f3b7a485da010128996e8b53d : SUCCESS in 47m 01s","accounts_in_message":[],"_revision_number":1},{"id":"80805c50eb4bdd0a15e35bcf29ebca5a4ae4f503","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 16:34:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/737587fabe6c48bcbea5f42446789126\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd81c209e4294d6ca07e64b271af7d5b : SUCCESS in 6m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/3d68a7bcd3a74f7593d607d08151878f : SUCCESS in 31m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b6917a2448e94b819f60105619f15605 : SUCCESS in 33m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/baf5bacb21aa4cad934a6b74a21234be : SUCCESS in 34m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/997ed3a62c3e46a598df117229e31883 : SUCCESS in 35m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1f3c7a813b84fa582cfb9859cfcd223 : SUCCESS in 11m 59s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bac388657ee438594da94f8c2006f22 : SUCCESS in 18m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/661bac975184407992571ce2e6763308 : SUCCESS in 18m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3551b0a73a4145f9a51718e04e0ba8d4 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/29399efd3d294297b9c1e91b86608d3f : SUCCESS in 18m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a6431f96c2b4407e99247625938abdea : SUCCESS in 18m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/24e49edb5d08482da09336ac5fed9071 : SUCCESS in 25m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6f433068d05a485494d156c67d16b9e8 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/79147ce34ced4fd4bdf0320a346e4cc0 : SUCCESS in 15m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/28c848b693fc46869f7acbb9bbcab13e : SUCCESS in 15m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a41f80a79d1c437db48e87b53be30250 : SUCCESS in 7m 59s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3c4fc6d10b7d48ed8da55a13f7a8415a : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/df50805b9f02418b9a00cc507f7c1e45 : SUCCESS in 6m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8ea7944dd37d4082b8773cded5098688 : SUCCESS in 58m 43s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9568c52b2c3e4cb1be4ca193eddf9685 : SUCCESS in 36m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/874dcba3c18d4124bb3c606cac1ce130 : SUCCESS in 47m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f8dc60a0329f43a5a5e7d2f0b79b540a : SUCCESS in 19m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/236f502b72ab49d58c09ac969fb0ceeb : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1b0aff9b3d8c43c7a5d447412ca533e4 : SUCCESS in 16m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/919861cf3b5a489e94a2cc296654c185 : SUCCESS in 49m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/941ceef4da0d44329784f218e823670a : SUCCESS in 27m 52s","accounts_in_message":[],"_revision_number":1},{"id":"c42e4f86ff2948715561b959a08174c95574a1e6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-15 17:45:31.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"cef51a2f411d29fddd0c346786aad05047f05fa9","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 08:41:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"40e756e18398793956e2f0580c5fce334edb17b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 08:42:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"3dce4bd1adce5d01a791174734c22dfe4446b357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:47:39.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab51d0a3dfda425a9ecd9b9e13013649\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fbbf54480c34976ae4ab7c65cf3897f : SUCCESS in 9m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/495fe53127e843edb805c8f58bee5e5c : SUCCESS in 52m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c0f8bcd5a74d4246b1a260819d4aaeaa : SUCCESS in 52m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90610e49af29441ea95a75859ede366f : SUCCESS in 48m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/874b33577c07430c8742c57d65bb3ad3 : SUCCESS in 1h 00m 22s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5c1be61f741f49388695635a2c959aef : SUCCESS in 5m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/e1c57a6e0adf4426910693d9f1ba35c5 : SUCCESS in 14m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5c406c90d0bd4328aaf9bb6d99bebff3 : SUCCESS in 16m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/274f6d6d1b084505bcbfe3849db04f11 : SUCCESS in 9m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8afe4cde9d88496fb58e150f5c16249e : SUCCESS in 14m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0141cbbe8ee744e7b47d7a0198dd0484 : SUCCESS in 8m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5fbc3f1a32f947d19d39ba79ecdf6312 : SUCCESS in 12m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5361291b0a76415086b035d9ceed22e5 : SUCCESS in 7m 45s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0ebdaed9bb0d44ac905f2c5caf6f882c : SUCCESS in 5m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5fc9a3e8cdba43489e5c2240181abd22 : SUCCESS in 37m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/671b9c15305c4704844c4a30f6219928 : SUCCESS in 41m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6fc61e5f5e2b41ed831b46280c2f1c43 : SUCCESS in 17m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/67e34b7f7265426dbe547e47c2524489 : SUCCESS in 18m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d99e6b5c7aed40a6a9ad5299fdeee331 : SUCCESS in 3m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8b92285b15d640c796459d88affa0c19 : SUCCESS in 8m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/77c3eb01a4cb49f7beecae4ee638db51 : SUCCESS in 33m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8f7d92f27c964b8390fed4ed7c57d737 : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":1},{"id":"049f0573abf5efd2e841b9cad61c302d3afa60e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:47:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4c697f97ed56c27a1598672ac15ab1cef8973f47","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 09:48:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/995b329230cb4ebe97b8f611c2ec542e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b5317ae94f40179f33ae700b726bd8 : SUCCESS in 39s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/535dbbaea2ec454689f6291d732ed068 : SUCCESS in 33s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"addcc1e32b42da93978171f4f56f237f930945fb","revisions":{"addcc1e32b42da93978171f4f56f237f930945fb":{"kind":"REWORK","_number":1,"created":"2025-04-15 15:30:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/63/947263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/63/947263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/947263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/947263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/63/947263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/63/947263/1"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-15 15:28:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-15 15:29:38.000000000","tz":60},"subject":"s3request: refactor error handling while reading input","message":"s3request: refactor error handling while reading input\n\nIntroduce a single error handler to catch errors while reading the\ninput.\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: Ib0330e863ad15e362dfa587517a26109c74eefec\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/addcc1e32b42da93978171f4f56f237f930945fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/addcc1e32b42da93978171f4f56f237f930945fb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
