)]}'
{"id":"openstack%2Fswift~947266","triplet_id":"openstack%2Fswift~master~Id5e0c4f597c77cae5742c53f31fad479fa864261","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-04-24 22:56:48.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-24 21:51:56.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Id5e0c4f597c77cae5742c53f31fad479fa864261","subject":"s3api: Remove some useless decoding","status":"MERGED","created":"2025-04-15 16:10:33.000000000","updated":"2025-04-24 22:57:46.000000000","submitted":"2025-04-24 22:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947266","meta_rev_id":"b4c544740a931f3a00fcef2c67760d975f0d4f65","_number":947266,"virtual_id_number":947266,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-24 22:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-24 21:51:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-24 21:51:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 16:26:28.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-04-15 17:06:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 17:15:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"14207d9e1a73d9e17a533706969d65d045526fdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-15 16:10:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1062b953087212b5c5b817e2b7f52d9e117945f1","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-04-15 16:26:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"637d1aa65123eff7b65e72b5bd957d7c5cb4aff4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 17:06:33.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b56c075b812840c8b43bdbad1b0693da\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f05dc77292514736b86889804a535195 : FAILURE in 54m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6fec073e00c0b99c557dc1e881f70c7572a0f09e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 17:15:48.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b8da008f994483ebcbf19c3f79dd091\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/17bc9dfc99244db4ba7589eb8edaf899 : SUCCESS in 8m 23s\n- grenade https://zuul.opendev.org/t/openstack/build/c9f4637106df4260a6dcc6c982714247 : SUCCESS in 28m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/efc1992e520d4c7ea05ac664cfab699f : SUCCESS in 48m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cbf27cf7d82348b991f7eb3244161a63 : SUCCESS in 28m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6f4f22616fe464287b2ffc0433a9698 : SUCCESS in 1h 00m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3381122d62e4a5ab2a5bb2fdd2f7d5b : FAILURE in 33m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15b4c591b66d4d3ea902b81b7a7e3974 : FAILURE in 34m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7df5094b992947319d9b608f98ddda5c : FAILURE in 32m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8e631e4fd7f949a78a9fd066115d2220 : SUCCESS in 3m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d50217b262f64651aa580f8440536bdd : FAILURE in 9m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c41ac64e4033418db0dd94c3e3bfeb45 : FAILURE in 9m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b60d3ed332084381ad866fb35e7ec66a : FAILURE in 12m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5ef80376c8ac4e81b399620bb80054ee : FAILURE in 14m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/86e13e8a90b6413591e9a0d54581c159 : FAILURE in 13m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/55348261d1174db195fcdb925cb14b2b : FAILURE in 17m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eda520efbc9e46f8a3d17a97731ef299 : SUCCESS in 6m 17s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02fdc4fed4054208ba6286473562fd28 : FAILURE in 16m 50s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/1075594b4ed54232b95cb56b175302df : SUCCESS in 6m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8629ae1582d148639faf3643ef8a6c3c : SUCCESS in 56m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2f4e46a6e83c4e27835348b1c4b377b7 : FAILURE in 23m 26s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42076ecce4a348eabcdd02e4e1744169 : FAILURE in 27m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/987c0ac10b6a4a1f9051c3255a9a30d5 : FAILURE in 20m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8906a72b9b64f4792525d5ab3905848 : FAILURE in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fb8d2e54f7b24d36b49287115537f8e1 : SUCCESS in 22m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ea098de27b24962b7a6ab114b2c2eb2 : SUCCESS in 45m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ad8dcee065042d8b6e3d02257f4094d : SUCCESS in 52m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2cecd778b1b0356afc61f4e2ce6dad20df70733f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-15 17:40:24.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"955f9dc8ad953167f4b20111865801cccc7d6915","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:27:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07b3e09c10e04b949e167c84340c8720\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9a6477d267244924a2ac110306f22a22 : SUCCESS in 45m 19s","accounts_in_message":[],"_revision_number":2},{"id":"99f31be89388f67ff914921b9b3e5f100fa01cfc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 18:56:24.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f9ca38e29f2145f9a50e6e04e72d543a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21de259389364a8298c71e283001a74a : SUCCESS in 8m 33s\n- grenade https://zuul.opendev.org/t/openstack/build/acfd3f126f5841f6af7abb2dfa848093 : SUCCESS in 46m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/71b4bd2f8ca34ce9b14366f60022a7f7 : SUCCESS in 53m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a841f465c5b047b59dc68e99f20b4142 : SUCCESS in 49m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98b5535ffa614fb7b310513706be1b8a : SUCCESS in 56m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a8c13ba4fc94b3d9b0749a73e5087b9 : SUCCESS in 19m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6370b2a70a2a4f9f86d73853afc55548 : SUCCESS in 17m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e8a69dabb7da456ea0ea0a9a067f3f75 : SUCCESS in 18m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/050f1b77a362464ca0cd690a7cacde20 : SUCCESS in 3m 14s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dabe982aacbc49ea979e33eca75ce1db : SUCCESS in 8m 50s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cfd0f71c1fd847298778fe8450df9d44 : SUCCESS in 23m 23s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0a17b3106976425c86241764eebdcd71 : SUCCESS in 27m 37s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dc27878facc144c8ad5d07fd26e1ec18 : SUCCESS in 11m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3649a21601e9486d8b44b7a59162a1a2 : SUCCESS in 13m 28s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9a19fcbdecce4747bad82232de06b08c : SUCCESS in 16m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/04e9a96fd8c645bbb7ca2b9555a493db : SUCCESS in 7m 34s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/96c2c83bec17456ab661ae263c7b8add : SUCCESS in 16m 29s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/b695817b41a44a4fb41a322922b4098b : SUCCESS in 7m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3030dce34dcc4b61832278b7dc0bead0 : SUCCESS in 55m 05s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4327c3c2a3914f6ca8bf7ea99a7ca82d : SUCCESS in 17m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cea63de422d4dc995cde569ebdecf3b : SUCCESS in 49m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d7bd73bcb6e34d63a2ad725acd852feb : SUCCESS in 15m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0823833d7a44639b101ca35f072c6c9 : SUCCESS in 5m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f22d9535c301420b8a95caae6f66748e : SUCCESS in 17m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92c7e99f68364c21a4e3ad2033a18218 : SUCCESS in 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1294da34d1a548f0b38345507db87f85 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":2},{"id":"8571f686bd2af409b496060d770174b06d316925","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-21 19:49:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8cc8173d97a493e2971921b6ffcd85316c252455","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 20:38:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a6151a65c8347d0b9e75710152f8a6b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b8abdb4644a142e283bd914e5a2f8d1d : SUCCESS in 47m 29s","accounts_in_message":[],"_revision_number":3},{"id":"a9be1fa8baeef15a762e2a8816eaf66654f6519c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 20:54:51.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/203b37b72c17423995457820c7cfd718\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35c3f46003524604aeb4e13fde792ee0 : SUCCESS in 4m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/dd9e62158dca40db9608b3d021fbae9e : SUCCESS in 27m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8f94657b80974c7a9acecc4df2a29309 : SUCCESS in 42m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e554a6bd90f04869a4c8c4523b4ecef7 : SUCCESS in 50m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e6860ed10742415985ea911b5cf51202 : SUCCESS in 1h 00m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07ba04d679a4414c862bee87c928b2a0 : SUCCESS in 21m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cecf0910de9d4bccb8533e0ae9f7cf48 : SUCCESS in 19m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3942f78712b44f0aa57f13994d02e3b : SUCCESS in 20m 31s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d9060c3e3f2d4760a3bbbe907d41b9ba : SUCCESS in 4m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ed510673455d4f48b71060e473b19830 : SUCCESS in 7m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/29cd9229abdb4f639b4eea3877d3bc09 : SUCCESS in 17m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/47d98cfa49e8488dbcef0c70c1173e2f : SUCCESS in 12m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/193de70003924588927e185189f8e261 : SUCCESS in 7m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f3c3d5fd42d140958391656e99a113be : SUCCESS in 17m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8a734858fc794f70a6cd55c69d4893e1 : SUCCESS in 9m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a610b5dc9cd448f3b79962ca105b6a56 : SUCCESS in 5m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d936969a959f49e09617b1c0df085a1f : SUCCESS in 7m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5523bdb604d843c09d7c17dfce84ba40 : SUCCESS in 5m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bf46a68ac7e241199734be274eed6887 : SUCCESS in 33m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00ee7c35ebf1409bae6a6537498c2aae : SUCCESS in 34m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8f4489a7774e44a2985a25e17084cabd : SUCCESS in 17m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4cb9e110bc1d4c93858afce46f0ecccf : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e85b9545a504ff8affb0cf8bc2d0617 : SUCCESS in 2m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e274807b994a4830add4a81914c096da : SUCCESS in 18m 08s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f1eb545d0db64c47a525a32a2dd47f8b : SUCCESS in 28m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b4a3e118e9e49528e1a18508e3717a3 : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":3},{"id":"d2845cb96cdd323cf44832025a7de04a84cccbb5","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-24 21:51:56.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a6d2b7d397ae6a7b4e5e472d98e0810b469193ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 21:52:18.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"06848de3288db80497fe4989c1a9a57ffab64cb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 22:56:48.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5673fbcc9f24e9c81fa3b9b68d15d51\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/430376e5cdb34d65bdf63f1970655e23 : SUCCESS in 8m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/41457638f5014a06828a8d30a75d3190 : SUCCESS in 53m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1dd2dbbf9df34b429addab2c7e3556bc : SUCCESS in 59m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f2a1bb14fb8748f3a392d609f391af37 : SUCCESS in 48m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52a769dcdd944d83b023183b77ac761f : SUCCESS in 57m 38s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/cf8313586b1b431eba1bb16f6c202bcc : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/6e294def70da4198b7d2739af154663a : SUCCESS in 16m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e27266de9c04f3587ea1a9763d10f37 : SUCCESS in 18m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6a07a0ebd9f64917a98ea683f63251ab : SUCCESS in 33m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/620aa4a6bcd44b808e3a2fb62765174f : SUCCESS in 13m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ecfea70a8a484c8a86dcb266f656b800 : SUCCESS in 14m 10s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/353c4754d3054412ae0052b19e272b89 : SUCCESS in 7m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a94dd5b069c6451e88c1d27ff0a3c851 : SUCCESS in 5m 15s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c59b80673abd41739a9aaddbad1cc3d3 : SUCCESS in 3m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ea4bd8fcbb014d12b3e486da4984769d : SUCCESS in 49m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d22b090ece9d4748ab9a013060cd98b7 : SUCCESS in 17m 20s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e2ac94c8dfdf4b09b7cc174176929e04 : SUCCESS in 48m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/33b83e3e6fc24a699b8701db11cd51db : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f7d31783e2847549c0eed0ed1962914 : SUCCESS in 3m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3d3ccd2d75a24d86bb882cabac5a6e63 : SUCCESS in 20m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/65f7810f8e6f47cb837d1f8ad114a476 : SUCCESS in 32m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/15ae5625fc2848a0a7b3c62c1806d60e : SUCCESS in 33m 54s","accounts_in_message":[],"_revision_number":3},{"id":"0d458e19c2cfb8d0b1e381d6561fba0e3df3bca9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 22:56:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"b4c544740a931f3a00fcef2c67760d975f0d4f65","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-24 22:57:46.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/958fae6473644f7fac25e702f51c686c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4cc4a46976348f9a1447607113c7008 : SUCCESS in 48s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/3c4920fbaf7a400a9b8c0c4361168e8b : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"51c8ea01857cb285a53dc529851157ac3a856949","revisions":{"5e6b2337b455410f188470fddb1c619a19223446":{"kind":"REWORK","_number":1,"created":"2025-04-15 16:10:33.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/947266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/947266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/947266/1"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 15:31:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 15:31:29.000000000","tz":-420},"subject":"s3api: Remove some useless decoding","message":"s3api: Remove some useless decoding\n\nNow that we no longer need to support py2, iterating over bytes always\nyields ints.\n\nChange-Id: Id5e0c4f597c77cae5742c53f31fad479fa864261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e6b2337b455410f188470fddb1c619a19223446"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5e6b2337b455410f188470fddb1c619a19223446"}]},"branch":"refs/heads/master"},"61e3a556866264204cc3b63b11f527bdd250ee84":{"kind":"REWORK","_number":2,"created":"2025-04-15 17:40:24.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/947266/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/947266/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/947266/2"}}},"commit":{"parents":[{"commit":"c7f93f0f63841140d75c21a833647073fb3f4eb3","subject":"Merge \"Add skip reason next to test name in zuul output\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7f93f0f63841140d75c21a833647073fb3f4eb3"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 15:31:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 17:40:24.000000000","tz":0},"subject":"s3api: Remove some useless decoding","message":"s3api: Remove some useless decoding\n\nNow that we no longer need to support py2, iterating over bytes always\nyields ints.\n\nChange-Id: Id5e0c4f597c77cae5742c53f31fad479fa864261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61e3a556866264204cc3b63b11f527bdd250ee84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/61e3a556866264204cc3b63b11f527bdd250ee84"}]},"branch":"refs/heads/master"},"51c8ea01857cb285a53dc529851157ac3a856949":{"kind":"REWORK","_number":3,"created":"2025-04-21 19:49:43.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/66/947266/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/947266/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947266/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/947266/3"}}},"commit":{"parents":[{"commit":"fd9ceecc506f88cd940edd7bb369574c044c5038","subject":"Merge \"s3request: refactor to introduce SigChecker classes\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fd9ceecc506f88cd940edd7bb369574c044c5038"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 15:31:26.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-21 19:48:50.000000000","tz":-420},"subject":"s3api: Remove some useless decoding","message":"s3api: Remove some useless decoding\n\nNow that we no longer need to support py2, iterating over bytes always\nyields ints.\n\nChange-Id: Id5e0c4f597c77cae5742c53f31fad479fa864261\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51c8ea01857cb285a53dc529851157ac3a856949"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51c8ea01857cb285a53dc529851157ac3a856949"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
