)]}'
{"id":"openstack%2Fswift~947283","triplet_id":"openstack%2Fswift~master~I6d4f3ab45c19651a3ea9c4760e07643321c5ed28","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-15 21:04:17.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I6d4f3ab45c19651a3ea9c4760e07643321c5ed28","subject":"s3api: Move more auth_details into SigCheckers","status":"NEW","created":"2025-04-15 20:01:08.000000000","updated":"2025-04-15 21:04:17.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"41df9b57cd40499abcbd22d765c93461bb986600","_number":947283,"virtual_id_number":947283,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-04-15 21:04:17.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 20:47:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-15 21:04:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9a51f4028a6d0301d16e338a30426c7ab3397850","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-15 20:01:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"264fd9780a401c07486f8f5d97ad53e0a0545d95","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 20:47:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05bf596986d54fd9b02280daff83e5f9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9ad3dfc1900f4c708b5faf4cdd775876 : SUCCESS in 44m 04s","accounts_in_message":[],"_revision_number":1},{"id":"41df9b57cd40499abcbd22d765c93461bb986600","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 21:04:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cdbdbe4ddf65450a86eb7e4676485173\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ed4001898b74714a93362833106ca0a : SUCCESS in 8m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/aeb60d55a4c344bfa16fd532f3f40994 : SUCCESS in 49m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/db48f865a6984ac5a8652c05dc481742 : SUCCESS in 55m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/39e267ad0944490a85cbefc08adf5645 : SUCCESS in 50m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/541bdb4f4f924a71a41201645c67e689 : SUCCESS in 1h 01m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/15bffb84937c45448987f579eac3817b : FAILURE in 21m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df2ad6607fb14c6e8ee203e0fa422af0 : FAILURE in 23m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41cc756ccfe54d41afcc67ced722db8e : FAILURE in 16m 09s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e1f5bf030108471aa1f93b6af2e96638 : FAILURE in 3m 58s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/06c3c0e0c2794b609412c2ea6133bbc0 : SUCCESS in 7m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a51658b3153471084c5a89785cb30c6 : SUCCESS in 9m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c1d0eb921f9947738b7f5970911c76e1 : SUCCESS in 19m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b89a5c2b10e848289046b30d89b1e615 : FAILURE in 14m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4d5e24573c454db49d97537635a07193 : FAILURE in 12m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2806e418acd04e82980d32f7fed57d0e : FAILURE in 18m 49s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a0d695cd05d24bfc8f85640956eae40f : SUCCESS in 6m 18s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b425759094f24080b48a2634853c65b2 : SUCCESS in 16m 45s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/476891a049594089a11915593f9ee203 : SUCCESS in 4m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/19bfad5c273b42ad82e2749efab11a20 : FAILURE in 34m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2cda656e885e452791f20f0a1b029e08 : FAILURE in 20m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/350adc27cbb048008fcf639254d91011 : FAILURE in 28m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a336a92445424cf38770f055101f0d95 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7c6f210c337452c90e62ffccad3ab76 : SUCCESS in 2m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21871687513b4624ba835fd093748959 : SUCCESS in 22m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81219a1b87464aa1afff9a7a952407ea : SUCCESS in 52m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec64aeaf9a5c466c90f76d67a226e9d1 : SUCCESS in 53m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5eddabee934c79516ce08727aa83a3e360129d57","revisions":{"5eddabee934c79516ce08727aa83a3e360129d57":{"kind":"REWORK","_number":1,"created":"2025-04-15 20:01:08.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/83/947283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/83/947283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/947283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/947283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/83/947283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/83/947283/1"}}},"commit":{"parents":[{"commit":"a93e420d32ecebb28f111e7fcf9e0be7163747af","subject":"s3request: refactor to introduce SigChecker classes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a93e420d32ecebb28f111e7fcf9e0be7163747af"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 20:00:01.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-15 20:00:01.000000000","tz":-420},"subject":"s3api: Move more auth_details into SigCheckers","message":"s3api: Move more auth_details into SigCheckers\n\nChange-Id: I6d4f3ab45c19651a3ea9c4760e07643321c5ed28\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5eddabee934c79516ce08727aa83a3e360129d57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5eddabee934c79516ce08727aa83a3e360129d57"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
