)]}'
{"id":"openstack%2Fswift~947449","triplet_id":"openstack%2Fswift~master~I4d130736a8e82319d8cd2982b6cb50c8c6ecd90d","project":"openstack/swift","branch":"master","topic":"label","hashtags":[],"change_id":"I4d130736a8e82319d8cd2982b6cb50c8c6ecd90d","subject":"sq: add s3request and utils tests","status":"ABANDONED","created":"2025-04-16 11:49:16.000000000","updated":"2025-04-22 10:11:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aca2f210494f4ac63fe3b32c38289f969656ef44","_number":947449,"virtual_id_number":947449,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-17 15:10:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 12:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-16 12:51:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7738b527cb96f164e1d5cb4a138e16a017c3f4ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-16 11:49:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33fe08d65f00251e1722b7d6bb223ab293ab9b2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 12:49:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01bfe40578c44acda732f7541f4f4a57\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8041c7f85b2344cfb79fffdc7602723c : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"376b40f8e5fdad809cfa4039f0d207747f273911","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 12:51:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e12d7f0246e4504801f2304a1e4133b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4edec9638e5a43d492ad3c654c0af32a : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/b61205caced94085a614bed7eefefd19 : SUCCESS in 47m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1ff56982dcc48e2be77199ff03120bd : SUCCESS in 26m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ca474b4c79604b3aa7a4fa7753bf9cd6 : SUCCESS in 52m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72f9f90dc5754a9eafd0f462c1efd45b : SUCCESS in 53m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/459fdac6474b4b6c8ba6548283b477ac : SUCCESS in 23m 40s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d08746f5e26f42eda5214c81b41bcbb0 : SUCCESS in 21m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c60e1a5a8c3a4e74ae1e90b625367331 : SUCCESS in 13m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3584dd920814462d9c9f511e78f4eca3 : SUCCESS in 6m 15s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d9475cd8f2a64a52bfae45794810b8d1 : SUCCESS in 7m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7cb6bc843e68403ab8d22be83af9b895 : SUCCESS in 19m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/60b6eb67636b45d0a822939d1d6ce805 : SUCCESS in 19m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c11ca9dc4827422fab8104e78d7c91cc : SUCCESS in 16m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ed3891238859463ab2b7a86eacf80d8c : SUCCESS in 15m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/406788435fca40819785847eb0ae6e5c : SUCCESS in 8m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58f5f14da86b4a229ac91f1a329738b0 : SUCCESS in 7m 29s\n- swift-tox-func-s3api-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/014393e2441c488baa6e6a69fda272ee : SUCCESS in 19m 11s (non-voting)\n- swift-tox-func-s3api-tests-tempauth https://zuul.opendev.org/t/openstack/build/44374a0ec5c84fb5bf66c0c81b149493 : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/50d0960cbb114f3684551e25b89567a2 : SUCCESS in 54m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e67c34cd78114517ac507abdef133e3f : SUCCESS in 45m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/194ae1696d7244a899d123efd83e5263 : SUCCESS in 39m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/41cb3c5e62b9486b843cd8da94e1dd31 : SUCCESS in 9m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c11462479ae84e94bd365a1fcdfcd514 : SUCCESS in 5m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/05120193e6504f84acf9f8ede30989a0 : SUCCESS in 13m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/61746efa9cfe47af960a81c03d97423e : SUCCESS in 53m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/685b26e4993e48d6b06b24b57b848c2a : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":1},{"id":"935438bc6020d3a77a70b9154040a1ff2ba08d54","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-17 13:44:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a9c7804264f73ff102a4e71b75291306fefccb3d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 14:37:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f048ef13e7546b9b3f78b066a0446d0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/42b7fa22a6514494af36111c6dec65f8 : SUCCESS in 51m 44s","accounts_in_message":[],"_revision_number":2},{"id":"ee91429c41d39f242134cc0e6e47d8f29c854f96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 15:10:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e7cc1ce79bf49f8a41f192bd143dfca\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb024c2aa5ce44349ff23bc0d007b3e3 : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/e06a776022e84f79b556f97233b394b5 : SUCCESS in 30m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e693277342de4b5babd119fe7bc992de : SUCCESS in 44m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4a4f037424a94b6bbf2bfa1ce1895117 : SUCCESS in 45m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e1f7fa91d454739af347f827c0807cf : SUCCESS in 55m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9912c8d93c7f4eb4b09a7163841f0484 : SUCCESS in 22m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/874835e3c98343948688925015a9f1c2 : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39d2de045b1643728660bb4c6936548f : SUCCESS in 20m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f6e16ba451a54005afaad2576976fd66 : SUCCESS in 6m 31s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/7c1f3e323c174bd684645330bfd1e634 : SUCCESS in 21m 18s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9e5185a2badf4958a325df2f5ff64662 : SUCCESS in 21m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/028d54b1ec24439592fd3368902b8607 : SUCCESS in 34m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/aa1f2b78d6294655b0fb737c49d385ce : SUCCESS in 15m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c2f0c7deb9094ccd96967200e89cd570 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d128c57adcbe4649bdd5fadd8830e20f : SUCCESS in 16m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4cb7e196c8524fcbb95da3c986bb4b7e : SUCCESS in 5m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ec8b5c98f74146c494874241a9cb299c : SUCCESS in 16m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9278f0a64f2c4709bd3afebd97ad3c15 : SUCCESS in 8m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9673f14d3f1b4c5790619f0c81ee75cd : SUCCESS in 1h 00m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/8d85c3d0792c499d8c718a01f1df40c4 : SUCCESS in 43m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ee6c8aa33e0d424f8ca6c5e0824f4bcd : SUCCESS in 46m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3bc9b846e611480198e0d6f559f5387b : SUCCESS in 23m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b74aa5a36744669f95856a4a9e18d9 : SUCCESS in 2m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/63a3c2ee3a7a492885256b1942980246 : SUCCESS in 22m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b428ba1f71140c598db4909df571010 : SUCCESS in 45m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8182b5412f644385be4c44d45abd1c12 : SUCCESS in 1h 01m 35s","accounts_in_message":[],"_revision_number":2},{"id":"aca2f210494f4ac63fe3b32c38289f969656ef44","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-22 10:11:39.000000000","message":"Abandoned\n\nsquashed into https://review.opendev.org/c/openstack/swift/+/944073/23","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"00aaeded50ad09258b7b4c82f14e79d6e5bf43c2","revisions":{"b233cbc4d20e342031085cb707cf1f43a74e3d61":{"kind":"REWORK","_number":1,"created":"2025-04-16 11:49:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/49/947449/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/947449/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/947449/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/947449/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/947449/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/947449/1"}}},"commit":{"parents":[{"commit":"87ff58d549d34f03a71cf38f10828e3596abc7c4","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/87ff58d549d34f03a71cf38f10828e3596abc7c4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-16 11:46:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-16 11:46:10.000000000","tz":60},"subject":"sq: add s3request and utils tests","message":"sq: add s3request and utils tests\n\nand cleanup unused aws_sha256 variable\n\nChange-Id: I4d130736a8e82319d8cd2982b6cb50c8c6ecd90d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b233cbc4d20e342031085cb707cf1f43a74e3d61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b233cbc4d20e342031085cb707cf1f43a74e3d61"}]},"branch":"refs/heads/master"},"00aaeded50ad09258b7b4c82f14e79d6e5bf43c2":{"kind":"REWORK","_number":2,"created":"2025-04-17 13:44:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/49/947449/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/49/947449/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/947449/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/947449/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/49/947449/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/49/947449/2"}}},"commit":{"parents":[{"commit":"7bc04ce503892b06cfc685040465a18052dd84d2","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7bc04ce503892b06cfc685040465a18052dd84d2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-16 11:46:10.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-17 13:36:59.000000000","tz":60},"subject":"sq: add s3request and utils tests","message":"sq: add s3request and utils tests\n\nand cleanup unused aws_sha256 variable\n\nChange-Id: I4d130736a8e82319d8cd2982b6cb50c8c6ecd90d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00aaeded50ad09258b7b4c82f14e79d6e5bf43c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00aaeded50ad09258b7b4c82f14e79d6e5bf43c2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
