)]}'
{"id":"openstack%2Fswift~947666","triplet_id":"openstack%2Fswift~master~I5c292c5af9f22d79cfa10f9ffc787843b87f2182","project":"openstack/swift","branch":"master","topic":"appcred-validation-no-catalog","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-04-18 18:45:04.000000000","reason":"\u003cGERRIT_ACCOUNT_25468\u003e replied on the change","reason_account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-18 18:45:04.000000000","reason":"\u003cGERRIT_ACCOUNT_25468\u003e replied on the change","reason_account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}}},"removed_from_attention_set":{"25468":{"account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"last_update":"2025-04-18 18:45:04.000000000","reason":"\u003cGERRIT_ACCOUNT_25468\u003e replied on the change","reason_account":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}}},"hashtags":[],"change_id":"I5c292c5af9f22d79cfa10f9ffc787843b87f2182","subject":"Fix issues with large catalogs which are required when processing an application credential acl","status":"NEW","created":"2025-04-18 12:47:19.000000000","updated":"2025-04-18 19:13:36.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a68662da46a02f4853325a074d416bb0e7429f6b","_number":947666,"virtual_id_number":947666,"owner":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-18 13:54:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":-1,"date":"2025-04-18 16:25:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"date":"2025-04-18 18:45:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-18 13:19:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-18 13:54:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-18 16:15:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-04-18 16:25:23.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"0cf9913f22f522c936cb9600d9318a03efba7768","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2025-04-18 12:47:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52da45a4b3e8b0a79be6f66ede136dc343aba5f6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 13:19:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c0fb1385aa4c59a5bbf4cad267ef90\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23bce09251f0403cb5a477136f2e57a3 : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"5dd07cb7908243737040b4397a714ba4ed8e5f8d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-18 13:54:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2810d6606a7a41bfb7108acf8706415f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce7965dcfedb48c49e2ac345bcc00867 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/2c863f0fdbf64a61b2aedba8ee46e192 : SUCCESS in 54m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8aeb8f7e9a844c90a00737d27d1bdba4 : SUCCESS in 56m 24s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4865d9bf55364e17ae40c802a0eb11bc : SUCCESS in 1h 02m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6397b9ecdd44f5d965a7bf51023873f : SUCCESS in 54m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5177b150b8664df4a6df10a8721d9e2f : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cebed8b442384bd08724c6c9d86a84db : SUCCESS in 19m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/087c7845faca43c3b3de16ceabccfebd : SUCCESS in 19m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f63589131b4741d89969c4eff75ef7c5 : SUCCESS in 5m 54s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/d02dedba346e4c31b938bb07ccac8392 : SUCCESS in 17m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8846be894b644437aabbf3b0e8e59035 : SUCCESS in 18m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f52497744417492fa346e4e2476a9bb3 : SUCCESS in 17m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/711d357c11a94a3cb2acd091fde727b0 : SUCCESS in 14m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/639e20e179a54ec9930756894c895fc9 : SUCCESS in 7m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3859cdbf11574537b18ef3504a9f437e : SUCCESS in 7m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cbce12468d92435486418f2abadc8e3e : SUCCESS in 4m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4dbb4c378c914657a912307b902167a4 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/665a641b285248089ac1ca4164e035ad : SUCCESS in 4m 23s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/41896e151fb54d62954bc424a2f0d762 : SUCCESS in 56m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b59e2a7465b946acad642465ccfbf6eb : SUCCESS in 17m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9b6ec34f6eae4112b4b813042c6412ee : SUCCESS in 35m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0afeb57ee2fb4453b4e2e8059e27f2af : SUCCESS in 19m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab3b609a208d4142bf6140fbd3b63a14 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd75535904dd4b638cf7181a9592ca20 : SUCCESS in 20m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14aa7e46cf164b36892bdd2300dc1985 : SUCCESS in 40m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f2d0e80343ef45469d83e2ea3208b454 : SUCCESS in 52m 50s","accounts_in_message":[],"_revision_number":1},{"id":"e5bb4b4d9203d565fc65a0d2c6e3104c46d82dc7","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2025-04-18 13:58:45.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"810aed545a2f8035a2e91933e3b02cc51026d5a7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-18 16:15:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8e64a1b5930d9409cf5e2379d0ecf8bd94093b1d","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-04-18 16:25:23.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f5561daf8019e67aab4c574fdff698d16df3d10","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2025-04-18 18:45:04.000000000","message":"Patch Set 1: -Code-Review\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"e2753272b698ac42767cd5452a947548c55c630d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2025-04-18 19:13:31.000000000","message":"Topic bug-1657390 removed","accounts_in_message":[],"_revision_number":1},{"id":"a68662da46a02f4853325a074d416bb0e7429f6b","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"date":"2025-04-18 19:13:36.000000000","message":"Topic set to appcred-validation-no-catalog","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d841f9fa83f67584774120db5ee2803c9f1126bc","revisions":{"d841f9fa83f67584774120db5ee2803c9f1126bc":{"kind":"REWORK","_number":1,"created":"2025-04-18 12:47:19.000000000","uploader":{"_account_id":25468,"name":"Michel Nederlof","email":"michel@nederlof.info","username":"pellucid"},"ref":"refs/changes/66/947666/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/947666/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947666/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947666/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/947666/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/947666/1"}}},"commit":{"parents":[{"commit":"e3e43ae2a696a2c40c590ab42608ccdd0b5ca13d","subject":"manage py36 constraints for swift","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e3e43ae2a696a2c40c590ab42608ccdd0b5ca13d"}]}],"author":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2025-04-18 12:47:11.000000000","tz":0},"committer":{"name":"Michel Nederlof","email":"mnederlof@cloudvps.com","date":"2025-04-18 12:47:11.000000000","tz":0},"subject":"Fix issues with large catalogs which are required when processing an application credential acl","message":"Fix issues with large catalogs which are required when processing an application credential acl\n\nRelated-bug: #1657390\nChange-Id: I5c292c5af9f22d79cfa10f9ffc787843b87f2182\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d841f9fa83f67584774120db5ee2803c9f1126bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d841f9fa83f67584774120db5ee2803c9f1126bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
