)]}'
{"id":"openstack%2Fswift~947838","triplet_id":"openstack%2Fswift~master~I410abe176ea06973fb335aacaa2717b4a535cfa6","project":"openstack/swift","branch":"master","topic":"label","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-12 18:44:06.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I410abe176ea06973fb335aacaa2717b4a535cfa6","subject":"sq: add readline method to StreamingInput","status":"ABANDONED","created":"2025-04-22 15:08:44.000000000","updated":"2025-05-12 18:44:06.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"91f7603767eb6113b8e91f5e84300d36702afb69","_number":947838,"virtual_id_number":947838,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-12 16:34:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-12 18:27:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-22 16:20:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-22 16:24:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-22 19:27:15.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-05-12 18:27:40.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"730cc76333ce578f669986a8a2c23de235598af9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-22 15:08:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8dc0e5a1daef00c4b13219b693b99bfeed01b36f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:20:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6e04692b9f54c86aaa0d37ae02190ab\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e13d259634224f9a964e45e019ec834c : SUCCESS in 1h 09m 19s","accounts_in_message":[],"_revision_number":1},{"id":"ebe5535e4700b16997955fd6fde1ff5c0d61d843","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 16:24:43.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4404cde23814d29b1ede1a148405964\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a173e61ac041e2a7849e4c157b92b4 : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e54c4fc487c746d8aaa9ca5f8d9b56a2 : SUCCESS in 58m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d514058bf839440cb2f396a535800cbe : SUCCESS in 58m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/51caf0045486442b98a63cd953027fe9 : SUCCESS in 55m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/79da6ac5d6a0484495c28200771c282e : SUCCESS in 33m 27s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7c1901e18aaa44de8a00e9a1196596ff : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c5c339055264f1db35f3d05321c0c21 : SUCCESS in 18m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/95c5d49cef4d402c8630b21b410a3f13 : SUCCESS in 21m 04s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/47de0fcf6309443b8e737a21f0ff1e7e : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a8854a0f31d64c318cf8c22e85e28a9f : SUCCESS in 7m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/57cafd59752143b9b058b75fbefc3475 : SUCCESS in 21m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3487c1bcd5c54d2fa93b9021f4ecacea : SUCCESS in 10m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ff14a7028ddb4821bc22c964fff4d0e4 : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/823c989e5e56420e912686b418217d54 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/52b4ba5eb6b64aa2962dee9781aca4bc : SUCCESS in 9m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8da6a65af68446bb80c3264f7604cca5 : SUCCESS in 7m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b2204b4274244bddbbbbca5b0465b253 : SUCCESS in 15m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2313320aa9dc4e8b933c566407e095e6 : SUCCESS in 7m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6e3f6d2be8443df97210fd5ed72205e : SUCCESS in 43m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/92c8c49825a54d828faa20d8ef8f8712 : SUCCESS in 36m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0fe89ea693554c5b9b90e9b05af5def2 : SUCCESS in 48m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91e4de3354c642b08efedcbe28fce574 : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edec913c390c47ca97f4dbf4d7306654 : SUCCESS in 4m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bedb0c2fa2af4395b26266d38be6db59 : SUCCESS in 17m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3901f651a6484dbeb97833a0eb486948 : SUCCESS in 41m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/242342a29a7f49ba80d6220db21199f4 : SUCCESS in 46m 14s","accounts_in_message":[],"_revision_number":1},{"id":"c7fad745fe0424a6e69c825cb2e6f96d7ae789f9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-22 19:27:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cb10d60e6bcd8b1b9496ff2787862fe192bf3065","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-04-28 08:55:11.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5fb7afe31b65b536c8e7a69e8390750d6c8c7ca5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 09:54:19.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c4aa482637a468d9e51f931ca54e8d0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4353510fa29846ffa49be1a18167c01a : SUCCESS in 56m 46s","accounts_in_message":[],"_revision_number":2},{"id":"ebc99d975cfe21f229f3c5537ebc10c9067f3ddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-28 10:01:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/259bc00fad7a427ba61cfa86da16a263\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c0b2a3fc14f24d9ab1458deb4b5e715d : SUCCESS in 10m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/e8774b13b5c04e869857fd1ae633f31b : SUCCESS in 51m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/43f4d5c5bb3c415caa3a324cc5adfd32 : SUCCESS in 49m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/421ac78b60704cbe92e0503b10d18b47 : SUCCESS in 50m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80436431cca7427a98dc0bd3fe2efd0c : SUCCESS in 59m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89d200d6ef434f179222cf244f127d3f : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3cde3a033d9b44b59a45136510ed4de9 : SUCCESS in 19m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a93f2aa30d9c43d78a7ca5a5f9707561 : SUCCESS in 19m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2aa925379c79482bbc978896942030b9 : SUCCESS in 5m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/18b35820f38d40459778ed29fe575c8c : SUCCESS in 17m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f7ed915f5a94a3db2442a94e56e571d : SUCCESS in 21m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b40c5ca6efdd4bd2836e7151cb92b826 : SUCCESS in 24m 29s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/24f7ff7196b24311ba166298b86d5827 : SUCCESS in 25m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/35086106534643e2875ed87c7a56fd0d : SUCCESS in 12m 50s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/018688f80d8f4d36a1666db34f9287b7 : SUCCESS in 13m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8771a75b28f24f2891a0ab03eff9eea6 : SUCCESS in 13m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aa3b85d080cf44fc88db074f8794787a : SUCCESS in 5m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/db418e691e13419faef4b990d87d7def : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c575cf15627a414babb83781e8763fae : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/530631ec6ffa4cecb94294e0601ab420 : SUCCESS in 55m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/40ae92b1bc80446f8dcef04b08aea80e : SUCCESS in 45m 41s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a9b45a317c724c99bb6b5b69995db650 : SUCCESS in 43m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83415b2205024369ba4f5fb6e58ac46f : SUCCESS in 19m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58afa3c6c03c4ccd8fa1d4a3cbab5701 : SUCCESS in 4m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/00317a45cab1407db11f8cc1e554bc15 : SUCCESS in 22m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/518244779b1a43bfa87ba7f9a91b80f9 : SUCCESS in 43m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2532a245a4bc41e5887506f09582c4eb : SUCCESS in 51m 50s","accounts_in_message":[],"_revision_number":2},{"id":"2025d9f53316c66edc4d97d78605faf238d1140e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-12 15:32:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"62cabc48363e3324e9cbd7d9ea8cd4ed925d5bda","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 16:09:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8253cf99c8f45ffa519357b2c387b96\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3c4151c9890349fe9f34cbd30e8639c7 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":3},{"id":"700a2535b805da63bf82e8d842da7b746ccc6f83","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 16:34:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa2062e907bb445c8b62a88dced9298b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/817123ac8e0444738c2a6968fddd3f4f : SUCCESS in 8m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/55f543e593eb46cea24104bb97878f4f : SUCCESS in 49m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3c520a9e7da485d91e49efdcb53f6bc : SUCCESS in 53m 14s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0151cb26d9e54ae28d17800ea03099f3 : SUCCESS in 43m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c72fc05bc3e41a4a84b96808f5cd4d1 : SUCCESS in 56m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d9600546f911466fad50474aede026ac : SUCCESS in 18m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3567518993ec4c86a6c5fa7fbc4a078f : SUCCESS in 18m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/288d3046b3fc4827a9eb9206f7a74d9c : SUCCESS in 23m 17s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/241cbc49ae5348f3a0de94005ca87408 : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/ee1706b26d6e4625b366aa6dbe7b955b : SUCCESS in 18m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2a38d21ff3cd440ba75a9cf0c49ee75a : SUCCESS in 19m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e9ea15ebbd99437db282b6fe7475505a : SUCCESS in 17m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d89752bc88184254b6cbfddf6d0e377b : SUCCESS in 21m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/88f890e6f7cc446d97ca4e35fa78f93c : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ee838c1af3f142eabceb1f91987cc8d1 : SUCCESS in 14m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3828c674014445798d6ff26b0016b3d4 : SUCCESS in 14m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d689bf3a3e8b4c6e90cfeaf6b7e9af24 : SUCCESS in 3m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/46541d88c2d04854913fad972c88cca2 : SUCCESS in 13m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/137ba0088bf3499f963bfbc5275c2591 : SUCCESS in 7m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c7d0c8d156ce415db98c19192e5224d5 : SUCCESS in 49m 51s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/368e455ba9244f679891c48c20c64320 : SUCCESS in 46m 53s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/65b98a0b782146b78dc83e4a6cb0faab : SUCCESS in 48m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a26df88906b434aa67d1f74692d62a8 : SUCCESS in 15m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6d1a581d5a74befbb0f77e2325a7d48 : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4a860ffec0e24628a94feefbecd39a9f : SUCCESS in 19m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/05a3aff2439247ed8ddaedd0137b3992 : SUCCESS in 42m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/78751b3334f948f995fa8a47a8043356 : SUCCESS in 54m 39s","accounts_in_message":[],"_revision_number":3},{"id":"9cbaf635a0cfec5a4bc91ebc2e2bacf88ed051c1","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-12 18:27:40.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"91f7603767eb6113b8e91f5e84300d36702afb69","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-12 18:44:06.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a28550c0d5cc1b609350eee1220323cfe7b255dc","revisions":{"a29bf72c8dd0f46d21af51adb0f9c15c5d1b4ab1":{"kind":"REWORK","_number":1,"created":"2025-04-22 15:08:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/947838/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/947838/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/947838/1"}}},"commit":{"parents":[{"commit":"1e806ca917318e2ec19764ffb4dd539d86696e53","subject":"sq: bugfix: truncate chunk reading at requested size","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e806ca917318e2ec19764ffb4dd539d86696e53"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-22 14:56:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-22 14:57:46.000000000","tz":60},"subject":"sq: add readline method to StreamingInput","message":"sq: add readline method to StreamingInput\n\nChange-Id: I410abe176ea06973fb335aacaa2717b4a535cfa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a29bf72c8dd0f46d21af51adb0f9c15c5d1b4ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a29bf72c8dd0f46d21af51adb0f9c15c5d1b4ab1"}]},"branch":"refs/heads/master"},"09820d113b8757b01f3b752349fb5295301c1e90":{"kind":"REWORK","_number":2,"created":"2025-04-28 08:55:11.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/947838/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/947838/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/947838/2"}}},"commit":{"parents":[{"commit":"a2312eb9e9aa5372843cb1c6587ebce77fc94cf9","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2312eb9e9aa5372843cb1c6587ebce77fc94cf9"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-22 14:56:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-28 08:53:39.000000000","tz":60},"subject":"sq: add readline method to StreamingInput","message":"sq: add readline method to StreamingInput\n\nChange-Id: I410abe176ea06973fb335aacaa2717b4a535cfa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09820d113b8757b01f3b752349fb5295301c1e90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/09820d113b8757b01f3b752349fb5295301c1e90"}]},"branch":"refs/heads/master"},"a28550c0d5cc1b609350eee1220323cfe7b255dc":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-05-12 15:32:17.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/38/947838/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/38/947838/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/38/947838/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/38/947838/3"}}},"commit":{"parents":[{"commit":"b236cd5fe95e49a52e5b169994f1c7898e2475af","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b236cd5fe95e49a52e5b169994f1c7898e2475af"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-04-22 14:56:20.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-12 15:31:53.000000000","tz":60},"subject":"sq: add readline method to StreamingInput","message":"sq: add readline method to StreamingInput\n\nChange-Id: I410abe176ea06973fb335aacaa2717b4a535cfa6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a28550c0d5cc1b609350eee1220323cfe7b255dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a28550c0d5cc1b609350eee1220323cfe7b255dc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
