)]}'
{"id":"openstack%2Fswift~948279","triplet_id":"openstack%2Fswift~master~Ifec53e00b760a7c3ee82b7ef3d54b8f856c55aba","project":"openstack/swift","branch":"master","topic":"drop-tags","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-04-25 23:36:05.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifec53e00b760a7c3ee82b7ef3d54b8f856c55aba","subject":"Remove tags from README","status":"MERGED","created":"2025-04-25 22:53:14.000000000","updated":"2025-04-25 23:36:59.000000000","submitted":"2025-04-25 23:36:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948279-drop-tags","meta_rev_id":"2226640c6b454595654976db6ad76a1ef045882e","_number":948279,"virtual_id_number":948279,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-25 23:36:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-04-25 23:28:18.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2025-04-25 23:28:18.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-25 23:12:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-25 23:28:18.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"38903d6ccda82bdb7ae13f7e8558cdf93794daac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-25 22:53:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"185cec53313bf5235b7465ea77dcc7a8f8364eac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 23:12:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4317f01e392940d5bfe2e64fd583e212\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3d874d67532341b4ae9cfdceee480f2a : SUCCESS in 2m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd9453ac7bf640f98843167824516213 : SUCCESS in 4m 29s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a5358423fadd43f3b0a0d6fffc5b4008 : SUCCESS in 6m 17s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/9c328692308344b09ffff82b2a640c5e : SUCCESS in 7m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8f27e2d26476458eb011dd10cc969720 : SUCCESS in 14m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8cf1d92e92b14bf19fccd95b39e86e62 : SUCCESS in 9m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ae0f6ef8daea4e2f82409c9ff20198ce : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1c98f01b16dd4bcb860c36f6193ca6cb : SUCCESS in 3m 11s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9270c512c62445199458873c723558e6 : SUCCESS in 14m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5844f031e9e24094a38e9f798ccbe8fc : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":1},{"id":"3acedf6db31a0c10dbddc7d834224a9e229eda4b","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-04-25 23:28:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1613fc93ba82d3662d64a6bacd623f4be5121c71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 23:28:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fbc41789e998ffdb9c233092bac68a47670c49ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 23:36:05.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b58f959205894673bc90a25e328115db\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0c194c24c12340a894ab71328272d9fa : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9eb5431afa764be1b25c2d8d940ad47b : SUCCESS in 4m 25s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6fb0d1534e384f9d93b14fef6156c887 : SUCCESS in 3m 39s (non-voting)\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/67f1a31b168a446cb2865061c4162397 : SUCCESS in 3m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e915e41e6ecc4d4db566c3505e66c9fd : SUCCESS in 3m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16346006122d4d6396056cb2c09bd950 : SUCCESS in 7m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/174951c28df94d02bb0683c3b0edb399 : SUCCESS in 2m 15s","accounts_in_message":[],"_revision_number":1},{"id":"5c03308f74911b36702edfc4c9c6c2e799a01597","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 23:36:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2226640c6b454595654976db6ad76a1ef045882e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-25 23:36:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/843cde07198343eea34d1e8315a62233\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee62641c075c4c3598a8051b3a7d1c64 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/35228a98916f4558a88bd8bc4b8bda14 : SUCCESS in 32s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f8efa83da369030517cc924cc4fc63ad8c736077","revisions":{"f8efa83da369030517cc924cc4fc63ad8c736077":{"kind":"REWORK","_number":1,"created":"2025-04-25 22:53:14.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/79/948279/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/79/948279/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/948279/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/948279/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/79/948279/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/79/948279/1"}}},"commit":{"parents":[{"commit":"64bd1acb9ec0b6ecc9a2f1c56c053c0de6f5efe6","subject":"Merge \"Clarify that get_ppid is Linux-only\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/64bd1acb9ec0b6ecc9a2f1c56c053c0de6f5efe6"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-25 22:51:59.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-25 22:51:59.000000000","tz":-420},"subject":"Remove tags from README","message":"Remove tags from README\n\nThe tags framework has been discontinued for a long time; see\nhttps://governance.openstack.org/tc/reference/tags/ and\nhttps://governance.openstack.org/tc/resolutions/20211224-tags-framework-removal.html\n\nRelated-Change: I78815177dba9d765587d9a0aee4e0f76e14a043e\nChange-Id: Ifec53e00b760a7c3ee82b7ef3d54b8f856c55aba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8efa83da369030517cc924cc4fc63ad8c736077"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f8efa83da369030517cc924cc4fc63ad8c736077"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
