)]}'
{"id":"openstack%2Fswift~948457","triplet_id":"openstack%2Fswift~master~Iaa2ce6b47c2771f44582fa12b572b142348e3478","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"Iaa2ce6b47c2771f44582fa12b572b142348e3478","subject":"swift-manage-expirer: sub-commands for summary, details","status":"ABANDONED","created":"2025-04-29 16:07:49.000000000","updated":"2025-06-26 23:26:25.000000000","total_comment_count":1,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"8cf99417b16a395decabf2f609498b4b0b2dba48","_number":948457,"virtual_id_number":948457,"owner":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-26 21:07:30.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}]},"reviewer_updates":[{"updated":"2025-04-29 17:21:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd4cc75a8ddeb9bd43206f7829eefa040eab24b1","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-29 16:07:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd394f0f13950ab2013cab982b2273d4b0b78974","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-29 16:09:27.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dd1d5b43e4ea2fbd535ac24cd45ae6664a5767d3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-04-29 16:13:03.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"0dab12b14be26de2dc253fe6ace3162cd6bb7560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 17:21:29.000000000","message":"Patch Set 2: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0ae05dbca4445d89637804030e2465d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afad50d961f244db8e7465d28b38833e : SUCCESS in 10m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/5652c10ff4404396922274d89c4f79c3 : SUCCESS in 56m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a6df806849f40bab6f4a0485876da33 : SUCCESS in 54m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f1a8adb244b645e187e67b3295a4d3a0 : SUCCESS in 50m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/333834d1f81f42a385f162ef3bdfb8aa : SUCCESS in 1h 03m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/610aacf5bc2b4959ab336e6bd9ad3be1 : SUCCESS in 18m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70ee69237edf4d8db1ae40ba9ab00ec4 : SUCCESS in 18m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ed29f0f962b466eb5ec46466b7c8059 : SUCCESS in 11m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/97b7d8636d0b4f589f2889f926be5e2d : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4cdac7d8d0e94f8796cfb626778c3ad0 : FAILURE in 17m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/705875c4b4a4410fafc7e02ff69b898a : FAILURE in 21m 50s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/675ea77489a0469fbd2fe870ce0fc127 : FAILURE in 23m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/459ec93e0cfb4f17ac25c95a56fdb67e : FAILURE in 22m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cffc984aec2c4b9080773c4e1b31cf69 : SUCCESS in 13m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ebbbb34381d411c9592af23a4159f47 : SUCCESS in 7m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1036df761691404284187449f9f00e1b : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c5abc50c8924278bc17626de7dfce52 : SUCCESS in 3m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e38f3309ee6b4fd98e6c747eb9e15fbf : SUCCESS in 15m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3015c4f3112143098179a8a6ecffb8e2 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ca87b62a273448e8bdc1b1ba830005f : SUCCESS in 33m 19s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/479a985f5a1e4adf88ff84df7433d1c5 : SUCCESS in 43m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0c77e5998ed46df9a2377d947be2fc8 : SUCCESS in 32m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0e22a763c758469b9cd95b5d3f6b1f91 : FAILURE in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f53770b754284ebdbee2b1abb52ba67d : FAILURE in 3m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a0774dca7634aa587477f943b928c23 : SUCCESS in 9m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0c2d77ac9b5d45e6899e75d1cdb621f0 : SUCCESS in 51m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0734c48da4554906acbfc63831c95d47 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":2},{"id":"59ff5557ad7627923a7b0089bf8baf74701d4bfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 17:41:43.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3eaaac49f65c4744b76fd8c0a15b50a1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c78a2f13762a460aa4a83ad75902798f : FAILURE in 1h 25m 50s","accounts_in_message":[],"_revision_number":2},{"id":"084abb21b1fc9595717f826b48e419fd1f464bfa","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-06-26 19:57:45.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"d8835ee40eb460dd74323c15d7d10d8b8b4efea8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 21:07:30.000000000","message":"Patch Set 3: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e61de44eae694b4aac17299449bc167b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/014eee040afe49a4b0988c816b9af7da : SUCCESS in 9m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/7370a62d5a5f4ce9ad6ee0bdc16f183f : SUCCESS in 28m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3558a587a97e43e9a323783fa58f8fec : SUCCESS in 1h 04m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/edb6209a4cb540d29e3b37ea008825d4 : SUCCESS in 55m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a792cd18f804450a9fd5b99466ea84c : SUCCESS in 32m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ca77d3a92dd4b14b3eab771092ea4de : SUCCESS in 19m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98ef77bfd36644a7b1451af81a95ee0f : SUCCESS in 21m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1c583455b1104d93bde57c453f0e81ba : SUCCESS in 10m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6d7f49f74c974669b44acdfdcbd31670 : SUCCESS in 4m 53s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7bc6e78e3a2e44e39e02f24102b80613 : SUCCESS in 21m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/663157d66eb0499294a9d48e43c19bae : SUCCESS in 17m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1f21e17d2a76445b8433107e63c52510 : SUCCESS in 31m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/643ca3ccc22b4ca3ae19fb1e32c1aa1d : SUCCESS in 21m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/326d498251974c5ea25d29b43942d8a8 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/abf6d0e064794564836306245d6fec85 : SUCCESS in 14m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/3b3d239db57d431a8d9c1dfa10b424df : SUCCESS in 14m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bafbd45c70494b9e854a4a0707dbe2ab : SUCCESS in 8m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3adcf8506d2a47399595ecb7b256e4d0 : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cb6f7cadaef54f64ba7cb12d2fd9b2f2 : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5628267bdf64cb3839b46df9d411b7a : SUCCESS in 54m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c79e573c70fe4fd494e4236b563ad978 : SUCCESS in 38m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f4d540b347b54bf7bad2b68a701a968e : SUCCESS in 27m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15dc7955b5e849a08b887532095fa417 : SUCCESS in 21m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b600bcb467a74d11b4378bde5574e57e : FAILURE in 5m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6aead603bc9457f9db555e10f99d896 : SUCCESS in 13m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/70f4d38edd4a48b5ad928eb01b480233 : SUCCESS in 36m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/679f1da32ed944cda8b04bd92cfbe80a : SUCCESS in 38m 52s","accounts_in_message":[],"_revision_number":3},{"id":"585e8a4e2f794d45de2748950594aa34b8d20c35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-26 21:14:22.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d90d80119fb44c5ae792180cc2493a5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/affbf6cb47df41d79e3cca565c8c2ba1 : SUCCESS in 1h 14m 20s","accounts_in_message":[],"_revision_number":3},{"id":"8cf99417b16a395decabf2f609498b4b0b2dba48","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"date":"2025-06-26 23:26:25.000000000","message":"Abandoned\n\nduplicate of https://review.opendev.org/c/openstack/swift/+/953486","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c4b9d34a7d8f6c1d1c0e707582a84bbed77c0441","revisions":{"c41b6479c4b84f6287da4a5e9caf51d0c047f402":{"kind":"REWORK","_number":1,"created":"2025-04-29 16:07:49.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/57/948457/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/948457/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/948457/1"}}},"commit":{"parents":[{"commit":"6f2ef63214a6be385c567a8829dc5cbc429f34a0","subject":"swift-expired-status: tool to inspect the expiry queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6f2ef63214a6be385c567a8829dc5cbc429f34a0"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-23 20:48:58.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-29 16:07:30.000000000","tz":-420},"subject":"swift-expired-status: sub-commands for summary, details, metrics and export-json","message":"swift-expired-status: sub-commands for summary, details, metrics and export-json\n\nChange-Id: Iaa2ce6b47c2771f44582fa12b572b142348e3478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c41b6479c4b84f6287da4a5e9caf51d0c047f402"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c41b6479c4b84f6287da4a5e9caf51d0c047f402"}]},"branch":"refs/heads/master"},"b068d2983f7ddc0ef45fade04ffba44cd3f18e9b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-04-29 16:13:03.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/57/948457/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/948457/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/948457/2"}}},"commit":{"parents":[{"commit":"a034a5d377d829728671241db932677e6cce86b4","subject":"swift-expired-status: tool to inspect the expiry queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a034a5d377d829728671241db932677e6cce86b4"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-23 20:48:58.000000000","tz":-420},"committer":{"name":"ASHWIN A NAIR","email":"ashnair@nvidia.com","date":"2025-04-29 16:13:03.000000000","tz":0},"subject":"swift-expired-status: sub-commands for summary, details, metrics and export-json","message":"swift-expired-status: sub-commands for summary, details, metrics and export-json\n\nChange-Id: Iaa2ce6b47c2771f44582fa12b572b142348e3478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b068d2983f7ddc0ef45fade04ffba44cd3f18e9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b068d2983f7ddc0ef45fade04ffba44cd3f18e9b"}]},"branch":"refs/heads/master","description":"Rebase"},"c4b9d34a7d8f6c1d1c0e707582a84bbed77c0441":{"kind":"REWORK","_number":3,"created":"2025-06-26 19:57:45.000000000","uploader":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"ref":"refs/changes/57/948457/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/948457/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/948457/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/948457/3"}}},"commit":{"parents":[{"commit":"ca9c450b028a84b5b59f7f47864adb0e3a1c9982","subject":"swift-expired-status: tool to inspect the expiry queue","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ca9c450b028a84b5b59f7f47864adb0e3a1c9982"}]}],"author":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-04-23 20:48:58.000000000","tz":-420},"committer":{"name":"ashnair","email":"ashnair@nvidia.com","date":"2025-06-26 19:57:27.000000000","tz":-420},"subject":"swift-manage-expirer: sub-commands for summary, details","message":"swift-manage-expirer: sub-commands for summary, details\n\nChange-Id: Iaa2ce6b47c2771f44582fa12b572b142348e3478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4b9d34a7d8f6c1d1c0e707582a84bbed77c0441"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c4b9d34a7d8f6c1d1c0e707582a84bbed77c0441"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
