)]}'
{"id":"openstack%2Fswift~948570","triplet_id":"openstack%2Fswift~master~Idfb4974a87274fd3774a95b4188b9866451a560d","project":"openstack/swift","branch":"master","attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-05-02 16:09:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-02 16:09:58.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Idfb4974a87274fd3774a95b4188b9866451a560d","subject":"sq: rename _do_get_updating_namespaces back to _get_updating_namespaces","status":"NEW","created":"2025-04-30 18:21:01.000000000","updated":"2025-05-02 16:09:58.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":4,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b6faec8246c57ab64b51d9ad581d49a823272490","_number":948570,"virtual_id_number":948570,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 19:22:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 19:22:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-01 23:57:09.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"CC"}],"messages":[{"id":"c279126b991f8e028fefc8ecb16f5932ac02ffb9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-30 18:21:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b2a7aa64161f4b06117d02f1a48cafe0d91fb5f9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:22:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fb1466a9c514123abf0c546983eb817\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e858f14c47b45ad877956c98332cc17 : SUCCESS in 4m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/096824b40ca54f31854e6e05ce33c23a : SUCCESS in 54m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e13ce8aaf904770bde1bed7550a012d : SUCCESS in 53m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/879b2d57b6cc4ab5b58506039d334eaa : SUCCESS in 50m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/39893e24642c490daf287886e3ef77b4 : SUCCESS in 50m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/207ae88972fd4d529815c7fee10996da : SUCCESS in 22m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c97b826573ce46d3938e0773cb9a8e65 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed11070a8a4949cf8dcadf2be4cfbf29 : SUCCESS in 19m 00s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/82330874383b44038bd45853a9dcafe5 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a18160e8d1d04f69bfbf85dcf517113a : SUCCESS in 20m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c876621ec59e4f8896f3533986a904d6 : SUCCESS in 19m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/31f31758df68456888d71cfbf04fa55b : SUCCESS in 33m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9097875fd7e04a70889d592c71d29a07 : SUCCESS in 27m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a6a1ca98df8b405ba97b5f4b1bdf4163 : SUCCESS in 14m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/573b5515fb3747e9b97d1bd2f948337b : SUCCESS in 13m 12s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ae4faa5c86d74265933056ac5a1d355b : SUCCESS in 14m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c95378efa7c340f4a804cb04552877ec : SUCCESS in 5m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd57ebea05e24e66bfea719c96cc0b08 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b75ce7963a39425ea5316d7971c29b48 : SUCCESS in 6m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/db30b0940a8b483b903aee42fa4a8264 : SUCCESS in 55m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/edac13d0876c45df9b9103ce095b2f93 : SUCCESS in 41m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ab5d1f49e154a3ab3cb7124abb4dc6d : SUCCESS in 30m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/730efc18ee5c443894def41acf7c4672 : SUCCESS in 20m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f00636b8e4d43daa7c08d2ba4cb75ed : SUCCESS in 2m 31s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7edb0e26765f405da1daeef333451198 : SUCCESS in 15m 29s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/303ce0a9fd264d5bb378bb47ead11240 : SUCCESS in 49m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5b43f4c65eb43d4b5f559d4add7069c : SUCCESS in 56m 13s","accounts_in_message":[],"_revision_number":1},{"id":"94347f66e696bebb4541e8b30b430f6c57a92466","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:57:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2a5a2036159a4e649085bf2b790a70be\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e511a6491e44e488b97fb6339f3ae56 : SUCCESS in 1h 18m 51s","accounts_in_message":[],"_revision_number":1},{"id":"cf2ba0ea7323f8741409e5669deda5aee7766537","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-01 23:57:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97344898dd4193b28931ce68ea65ba1458bda05b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-02 15:22:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"879711235565021b7a0ba86aa3c81ccee46bcd36","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-02 15:57:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6faec8246c57ab64b51d9ad581d49a823272490","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-02 16:09:58.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02d2a7447c87c58a71dedc13760e11565f1fb2ef","revisions":{"02d2a7447c87c58a71dedc13760e11565f1fb2ef":{"kind":"REWORK","_number":1,"created":"2025-04-30 18:21:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/70/948570/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/70/948570/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/948570/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/948570/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/70/948570/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/70/948570/1"}}},"commit":{"parents":[{"commit":"3136ea74d3bc9a03b0553fec387cc1411e4e80a9","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3136ea74d3bc9a03b0553fec387cc1411e4e80a9"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-30 17:53:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-30 18:20:32.000000000","tz":-420},"subject":"sq: rename _do_get_updating_namespaces back to _get_updating_namespaces","message":"sq: rename _do_get_updating_namespaces back to _get_updating_namespaces\n\nChange-Id: Idfb4974a87274fd3774a95b4188b9866451a560d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d2a7447c87c58a71dedc13760e11565f1fb2ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d2a7447c87c58a71dedc13760e11565f1fb2ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
