)]}'
{"id":"openstack%2Fswift~948571","triplet_id":"openstack%2Fswift~master~Ia5f10e308341d07c21edca53b3e1d2d9089f4dc3","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-05 19:36:26.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia5f10e308341d07c21edca53b3e1d2d9089f4dc3","subject":"sq: Pull NamespaceCachePopulator off controller","status":"ABANDONED","created":"2025-04-30 18:21:01.000000000","updated":"2025-05-05 19:36:26.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"58337dcf1efb02d22389ca523851e1a274f6b82a","_number":948571,"virtual_id_number":948571,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 19:26:48.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-05-02 15:59:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 19:26:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-02 15:59:40.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"d4f785f2118eb6b5bca76f0f2562b8921b421a0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-30 18:21:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1588b8e9aa8bc1cfa5cb8e45c4a99cecf346445","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 19:26:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f30784cf5bb945faa9b161562ef71f52\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64b9da6d9b7241688b215d1546bf25f2 : SUCCESS in 10m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/fa542452a70446baab5904e745ffef68 : SUCCESS in 53m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e7fc124296b43fd95d28dc7aca8ef43 : SUCCESS in 56m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2c812a9b92714bc09b29b568fcf11ee2 : SUCCESS in 51m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/304bdb0e1aed4f84bef6efaf6c2b87ec : SUCCESS in 44m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a138c702ee604a99bdb2fc367f8000b8 : SUCCESS in 18m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fed9cbc3dce54c82b4439227f11489f9 : SUCCESS in 18m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4a93756044c943f18a2d8fa2318e6998 : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/647abc40e96041a89d1c1873777879cf : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/dfdeed421b2b4c6586599f2b2bb71ec3 : SUCCESS in 19m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3e2cb378be124b02ba2bf9170c4f7888 : SUCCESS in 22m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cce8029a249140468df98b0593034f27 : SUCCESS in 35m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/35438e3dea344ddbbbc90fadefff8ebb : SUCCESS in 25m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/769711e019e644e5886659cc474efef7 : SUCCESS in 15m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5017f83d049e42f8906a6747faa3bd13 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/bff84340010b47c684fc3f5db0681c37 : SUCCESS in 13m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1d79e1bd6b1249b895dd458bbcf8ad22 : SUCCESS in 5m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cc9706ddfa4e4915a7d65230d9617f35 : SUCCESS in 14m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/98bbbe6c5dd94cbf952af59cafd99d81 : SUCCESS in 8m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/942c5a98705947b0969e05956ae08129 : SUCCESS in 53m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f72df7b4bfaf436da3fda6c4f3ddcb1c : SUCCESS in 38m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92fca3d2e06e49cb95ca02056d761a62 : SUCCESS in 38m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f848afafd694dfc9f26dbf1b8665ba6 : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e8a082113b148c5899c6ce2803f2de7 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2634c5a427a645dbbc20dc00445a62c0 : SUCCESS in 18m 16s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/92cc676fecb5435296997c6bc59959fe : SUCCESS in 41m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53aae9aaf29d42ed8ee1534f8506708a : SUCCESS in 52m 23s","accounts_in_message":[],"_revision_number":1},{"id":"72d73e040def51b95e9f544651b269ec1924dc06","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 20:03:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad732f7e45a14237b74843cc1e75be35\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9c48da68201741b784843a3965fced35 : SUCCESS in 1h 24m 06s","accounts_in_message":[],"_revision_number":1},{"id":"03647db7d6767fe502d389381e22a939d8d51566","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-02 15:59:40.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58337dcf1efb02d22389ca523851e1a274f6b82a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-05 19:36:26.000000000","message":"Abandoned\n\nSquashed into https://review.opendev.org/c/openstack/swift/+/908969","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6350556a9ed5e6d2224b76805b201a04a1b9bcce","revisions":{"6350556a9ed5e6d2224b76805b201a04a1b9bcce":{"kind":"REWORK","_number":1,"created":"2025-04-30 18:21:01.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/948571/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/948571/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/948571/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/948571/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/948571/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/948571/1"}}},"commit":{"parents":[{"commit":"02d2a7447c87c58a71dedc13760e11565f1fb2ef","subject":"sq: rename _do_get_updating_namespaces back to _get_updating_namespaces","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/02d2a7447c87c58a71dedc13760e11565f1fb2ef"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-30 17:59:45.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-30 18:20:32.000000000","tz":-420},"subject":"sq: Pull NamespaceCachePopulator off controller","message":"sq: Pull NamespaceCachePopulator off controller\n\nChange-Id: Ia5f10e308341d07c21edca53b3e1d2d9089f4dc3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6350556a9ed5e6d2224b76805b201a04a1b9bcce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6350556a9ed5e6d2224b76805b201a04a1b9bcce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
