)]}'
{"id":"openstack%2Fswift~948585","triplet_id":"openstack%2Fswift~master~Iaff8b69894595df9f9cb666b3bcb3225e2e2ac2f","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-01 21:43:49.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iaff8b69894595df9f9cb666b3bcb3225e2e2ac2f","subject":"sq: common: add memcached based cooperative token mechanism.","status":"ABANDONED","created":"2025-04-30 20:19:05.000000000","updated":"2025-05-01 21:43:49.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"40b4dddc7a21bba9609d3bf09f482f8bd2080dba","_number":948585,"virtual_id_number":948585,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-30 21:40:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-05-01 21:04:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-30 21:18:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-04-30 21:40:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-01 21:04:57.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"9bb96c4e47ed9f74fb4da00f5cbc26eac91982d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-04-30 20:19:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84b14d5ab54646bce4473ca16070d61ef4c0bb85","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 21:18:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e010154318c44238207af83024fbdd6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/da29625851ee4183a4edf2eafedd7acc : SUCCESS in 55m 12s","accounts_in_message":[],"_revision_number":1},{"id":"571e6229a395bfc9c46a55618b8f56be755973bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 21:40:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e53119722ae4623876d8271510136f8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49e9437a39884f56a5042e97f9de3735 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/f207efff098a4c2eb33f852e68a7febe : SUCCESS in 51m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/dc0a54d55b04494392778ca5d7c8ded6 : SUCCESS in 52m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/11ffcbb5492c4b8bb964d8c69648ed35 : SUCCESS in 29m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/16d2113b78444161b217666622128030 : SUCCESS in 54m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9b36c04721bf492385b62d56ea7b3d01 : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f706de75b9b148599aa7eb97ef1f70d0 : SUCCESS in 9m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/85f593dd1a7746e0902d8623ee290d7e : SUCCESS in 18m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fd41f730ae864dd8abd52df4cba08870 : SUCCESS in 5m 12s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/4312b83a1fa840c18da5d5f4bcd92730 : SUCCESS in 16m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/07f5f76af00a4d26952393402a1ac233 : SUCCESS in 18m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b99e436cf8f842839249127f57857d55 : SUCCESS in 24m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9cbc061413574e68aa3273d54f784a2e : SUCCESS in 22m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5360708302a64c90a14e71e389e07167 : SUCCESS in 12m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4555ee6fe4fe475ea57e85507c52aeac : SUCCESS in 13m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f21f9bea93b244588f88fb22b8080243 : SUCCESS in 14m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9472c1fc7824c3c92a91e2f52cfdaad : SUCCESS in 7m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/603c8ae07fb54bab93e3f875737f9f58 : SUCCESS in 15m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ce2a174ff67a443697a619f58e107fde : SUCCESS in 6m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/067a60e86a5d473a9cd09b34c21a7ba3 : SUCCESS in 55m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4902f06e6bf2404abb5d5e38e035fcb0 : SUCCESS in 42m 51s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ed9524c36cb456a9d8c7cd88326d1ff : SUCCESS in 46m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/efed4f8f85e94662a24ba98eb8ad7d3c : SUCCESS in 19m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5ec3375e21489a86a3228c43ed26eb : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3f7798f0f12b44b8ba23ebab34c3de0f : SUCCESS in 21m 34s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f61dc3b6a1434cac8a934a4aa29465a2 : SUCCESS in 53m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bdc35f566eef4296bf9fd939114cae21 : SUCCESS in 50m 42s","accounts_in_message":[],"_revision_number":1},{"id":"612b5697021e4bd7e5b40f78e8aad715174a918a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-01 21:04:57.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40b4dddc7a21bba9609d3bf09f482f8bd2080dba","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-01 21:43:49.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"37547d495de4c476ca85118d492a5c9921a6bdc0","revisions":{"37547d495de4c476ca85118d492a5c9921a6bdc0":{"kind":"REWORK","_number":1,"created":"2025-04-30 20:19:05.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/85/948585/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/85/948585/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/948585/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/948585/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/85/948585/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/85/948585/1"}}},"commit":{"parents":[{"commit":"13c72a81b1cf362615a4a1cfed7aa24d4848fc9e","subject":"common: add memcached based cooperative token mechanism.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13c72a81b1cf362615a4a1cfed7aa24d4848fc9e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-30 20:18:06.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-04-30 20:18:32.000000000","tz":-420},"subject":"sq: common: add memcached based cooperative token mechanism.","message":"sq: common: add memcached based cooperative token mechanism.\n\nChange-Id: Iaff8b69894595df9f9cb666b3bcb3225e2e2ac2f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37547d495de4c476ca85118d492a5c9921a6bdc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/37547d495de4c476ca85118d492a5c9921a6bdc0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
