)]}'
{"id":"openstack%2Fswift~948833","triplet_id":"openstack%2Fswift~master~Ia52238912f57f98f0640beeb8771995d0cd34d48","project":"openstack/swift","branch":"master","topic":"cooperative_token_proxy","attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-05-07 04:52:23.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ia52238912f57f98f0640beeb8771995d0cd34d48","subject":"num_token \u003d 0 is go slow button","status":"NEW","created":"2025-05-05 21:11:15.000000000","updated":"2025-05-07 04:52:23.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fa2d49140bb01678000d3d26bd9f4d9f1c24e5af","_number":948833,"virtual_id_number":948833,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-05 22:23:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-05-07 04:52:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 22:08:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-05 22:23:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 04:52:23.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"bc1d0ff274138fd18a7ddf9042b2b0773d17243d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-05-05 21:11:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e10ba466f5e00612f605e4b0062392d67076c500","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 22:08:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be7e0bb6c3934c6e8ed437fd28778e99\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b7a2d053c0d423995fcf52bdf2e1d0c : SUCCESS in 55m 18s","accounts_in_message":[],"_revision_number":1},{"id":"6b32b28254faa6c84463dc4c0f2af36a7af2fb26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 22:23:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2e5bca3ff034cad9d8ddbcc819ca0a8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2eae3f362be46118c1b4807e8b9b2dd : SUCCESS in 7m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bcae88ebbcb740b29e46b95916acfb06 : SUCCESS in 40m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c2e6d5c4ccfd4c0f976d219c19767166 : SUCCESS in 17m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4593f715baec417193126900d8dfbc61 : SUCCESS in 9m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33b8d43f703b46d9890f57b7816f87ea : SUCCESS in 11m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/664d29af6c9f4f91a52db46c06713b20 : SUCCESS in 3m 13s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/8f6217894f044767b9dfb3021fe62610 : SUCCESS in 18m 04s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2c7944c68c874ec897f96b8b120cc3cc : SUCCESS in 21m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9f2afc3f2c8c4c19a50623660e712d43 : SUCCESS in 31m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/37334772c6cc43b0bfa37dc7c74a442a : SUCCESS in 31m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/84906831ad28412d9db7524881398555 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87aa073639e4471594c57157fff7b747 : SUCCESS in 14m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/93609cdc8fd7479493597d9e8a5490ec : SUCCESS in 14m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3928c3f2649d4f38af2d0f1fd654c359 : SUCCESS in 17m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3cb6bc0e39fd463fa5f85aaed17ff868 : SUCCESS in 45m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be595860ba284d64893ddc83c166cb37 : SUCCESS in 20m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed7409632af4a1a8d2805ee6a158ea5 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/351ea50dd6564426b2a26abe3436f822 : SUCCESS in 19m 32s","accounts_in_message":[],"_revision_number":1},{"id":"fa2d49140bb01678000d3d26bd9f4d9f1c24e5af","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-07 04:52:23.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf34b06e99683d4743b153e188e5f4dab5fdac3a","revisions":{"bf34b06e99683d4743b153e188e5f4dab5fdac3a":{"kind":"REWORK","_number":1,"created":"2025-05-05 21:11:15.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/33/948833/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/33/948833/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/948833/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/948833/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/33/948833/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/33/948833/1"}}},"commit":{"parents":[{"commit":"16d6894d66acef49f21b5783e22a1d545e24f7fd","subject":"common: add memcached based cooperative token mechanism.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/16d6894d66acef49f21b5783e22a1d545e24f7fd"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-05-05 21:10:47.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-05-05 21:10:53.000000000","tz":-300},"subject":"num_token \u003d 0 is go slow button","message":"num_token \u003d 0 is go slow button\n\nChange-Id: Ia52238912f57f98f0640beeb8771995d0cd34d48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf34b06e99683d4743b153e188e5f4dab5fdac3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bf34b06e99683d4743b153e188e5f4dab5fdac3a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
