)]}'
{"id":"openstack%2Fswift~949108","triplet_id":"openstack%2Fswift~master~I3239e4c43ed18c02c94effc951cc1ad6066afc61","project":"openstack/swift","branch":"master","topic":"fix-s3-list-bucket-creation-date","attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-06-22 21:50:02.000000000","reason":"\u003cGERRIT_ACCOUNT_37337\u003e replied on the change","reason_account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-07-05 20:58:35.000000000","reason":"\u003cGERRIT_ACCOUNT_37337\u003e replied on the change","reason_account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}}},"removed_from_attention_set":{"37337":{"account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"last_update":"2025-07-05 21:08:10.000000000","reason":"\u003cGERRIT_ACCOUNT_37337\u003e replied on the change","reason_account":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}}},"hashtags":[],"change_id":"I3239e4c43ed18c02c94effc951cc1ad6066afc61","subject":"Fixing constant creation-date for s3api list-buckets Closes-Bug: #1856938 Change-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61 Signed-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e","status":"NEW","created":"2025-05-08 01:21:23.000000000","updated":"2025-07-05 22:06:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":26,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"52f581ac72891f199c7e06d46ce43e0c49d9a432","_number":949108,"virtual_id_number":949108,"owner":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-05 22:06:38.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":1,"date":"2025-07-05 21:08:10.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 02:06:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-08 02:44:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-22 21:50:02.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-06-22 21:50:02.000000000","updated_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"b0f12304e546886f8f0f8ebb0905b97da29ce155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-08 01:21:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3f48adfc8e4a3e66d6cc28c3d8411362cc665b08","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-05-08 01:25:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b8a4cfcf95134eb551335b8005f78c6b60f96df2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 02:06:23.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc8045f7d97f4579bb193354b17a130c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/60f2b25ba35f4bfd9b3206401a311182 : FAILURE in 43m 22s","accounts_in_message":[],"_revision_number":1},{"id":"3143fd5d63e928625952f7858a18883b01077cbf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 02:44:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22e275358d2c4e8aba596142bb73c030\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4046f9441e244986b871a328c2b93cae : SUCCESS in 9m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/9e6dab758f3f4e43841180411c3594b3 : SUCCESS in 53m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b21daa2d96d042bd9d84d79ba6361eda : SUCCESS in 51m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e3d7378666a84e67b56a0c83847f665d : SUCCESS in 49m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e60c80b64107442b8b8370950ffbcd87 : SUCCESS in 1h 06m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0145d3e8579c4465b59c41ecf1a55742 : FAILURE in 23m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f9b65f858d664efa9c5ad92c55dfb2be : FAILURE in 25m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d1bf2e59048e4216a9da9d69c38db355 : FAILURE in 22m 47s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1713fcb143eb41e696c822b7615af7fe : SUCCESS in 5m 42s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/f2df46b45baa47ab90d16541cbbf4448 : FAILURE in 7m 06s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4c7fa02d01404693ae58a8e0ad23b278 : FAILURE in 14m 31s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7168c928f0c3450c909f9e469287376b : FAILURE in 21m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1c343b3121d0453ea084a4c17e5f815e : FAILURE in 26m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c1c6235add2c48f3a29aa4be3179d2eb : FAILURE in 17m 44s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/35e57efaa80a49ef9a46e5fde0ad26c3 : FAILURE in 21m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c478de5d31034fb6b453dde787105262 : FAILURE in 19m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/0c713e2519be4982924e36e3e536cacd : SUCCESS in 6m 58s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/770902e7d7c04f13b7d93191194532ae : SUCCESS in 7m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eaf0b74c25b64e728bad706ffddaec18 : SUCCESS in 5m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/70c014952f9b4e70812f4d76a2832cb2 : SUCCESS in 59m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dc3240870d1b40fc8e3c65b837247837 : FAILURE in 25m 13s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/284f615178ea4a7f94600d09dcfa4cd8 : FAILURE in 25m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/04347d6ceb7949648cb47806afeebf07 : FAILURE in 14m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8dbc84441a549dd941b9d1c1d43a817 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/38fb751eeda24f4b886f31556e10d02b : SUCCESS in 19m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/09bef7be94e74cfc8a84986582bc9c0b : SUCCESS in 37m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f3eb0051ed194e82967f3607327ba140 : SUCCESS in 45m 53s","accounts_in_message":[],"_revision_number":1},{"id":"79f8c4e55ef891dbb849d885a2f8c834206dbe29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-06-22 18:45:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3cfb101cf46d2caa1a3f753f975ddeb5969d4274","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 19:37:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80d34b851294a6c829ef86226b257e5\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/81459354128a4c0abd0ed22ad6feb970 : SUCCESS in 50m 23s","accounts_in_message":[],"_revision_number":2},{"id":"d068b88c1c5de26e6ed9798988ad7c24d4da1ed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-22 19:59:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0914143fbe4a278f11655fc477be05\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad9ed3ad31dc477884cde38830b6931f : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/59e7ca4ab8c84ef9bdd9498bc8a37063 : SUCCESS in 45m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/337a6a10fff04e0da8daeac02b306bb7 : SUCCESS in 1h 03m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/95c91c3022194ae29da14aacbb361557 : SUCCESS in 54m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa48dcebcae14f69aea3f95163e04da8 : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d973507f58514ead82723f7459c84b26 : SUCCESS in 18m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6234cd3c9d942538c4d8d646319b4b6 : SUCCESS in 20m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66b82db4fe434cc089c5e52e78a6e37c : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/363ae736468048599d0e8e4395dfd291 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e57c27b2b7c94abc9f5f3a7f3bd5beb5 : SUCCESS in 23m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/977ef3533d864f61b5561d22d5229bc5 : SUCCESS in 18m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/efed71df8b064d8292ba3c44b5077bd7 : SUCCESS in 31m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c3696abba9314fe0a9a133ef71658511 : SUCCESS in 21m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1cbd1fb7bcc941ef9559116c84aa6766 : SUCCESS in 12m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/843d00304a1d42dcad3864da989bc80b : SUCCESS in 14m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4c344574b21b4ffcb34a12ad718a1798 : SUCCESS in 15m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ee2ed7b4177f4257beb4bc1b6bb9b8cf : SUCCESS in 5m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/700a882413994571a7512aae6f87bb59 : SUCCESS in 15m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/440f5280b38446d5be0a84ac981712c1 : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b3eca8f7719b48f6a146e61feafc2745 : SUCCESS in 48m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ca83cfad3b674174979ad5bf9db84e2f : SUCCESS in 47m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/150ea23992964894ba232ed4989fedea : SUCCESS in 46m 27s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/340df37d5fae47b888e66dc24bd0318a : SUCCESS in 16m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d935e8a945548c7bbff48fc2f454c09 : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e6734a3e28cd43e9941e6bb758518568 : SUCCESS in 15m 44s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e9d7ac734877414f80d6359164e0db5d : SUCCESS in 42m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9598c842e7d34372aecb2dd453967dc7 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":2},{"id":"1930bd863a91b41b5dc16edb854d06d84e602074","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-06-22 21:50:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"97d2a28b921849b2868f94d41f34ca19fd2b8804","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-06-22 21:50:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"330e04f0c5072648654efb57fd3d6adf77b1fee3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-25 01:30:14.000000000","message":"Patch Set 2:\n\n(13 comments)","accounts_in_message":[],"_revision_number":2},{"id":"54ae415a0a7b6c611e331f9157cd84b39b002231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-07-05 20:56:37.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"66418b1b30a2e5ca3716e610cf326fb802bd5e9d","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-07-05 20:58:35.000000000","message":"Patch Set 3: Code-Review+1\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"bd15f5270f8cebdbbdc8d35d50cc391a9e7a0974","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-07-05 21:00:57.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"d7453836a3a0de6acb45260acc8eaf32058e15f6","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-07-05 21:08:10.000000000","message":"Patch Set 4: Code-Review+1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4},{"id":"2a2c250fd6aa4d47fa78d2c0d7a2d03ca89c7071","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 21:52:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63048a87ab4b44b995e723979ff80bfd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c6532b3c63af4e8e99ae4b1d58169d7c : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":4},{"id":"52f581ac72891f199c7e06d46ce43e0c49d9a432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 22:06:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/400a3cbec7ce4fff9b55ef9dc1e277db\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8bd010a600b2482fa1393966ce277930 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/5f0388f957284d1e9e2b02d827a162da : SUCCESS in 40m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6448770aac7942cd91500aafcbefed35 : SUCCESS in 44m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/49d5eaba07834bcba5ba6c171c61ac5e : SUCCESS in 36m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/222307ad7ece48ff906923f2fd62eeca : SUCCESS in 57m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d49c1c1548fe4340838f0d269a249435 : SUCCESS in 11m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5538c60a922d48c58853c9642db1aebb : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b536abd409a74c2ca6ca68b3919fe835 : SUCCESS in 20m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2ec51de6a53e43f6b04291f53308ce38 : SUCCESS in 4m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a12996b2be764e28a2c37405171da6c8 : SUCCESS in 14m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/164099bc20f3465a95a764a6212b8692 : SUCCESS in 13m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d82bce4ea917414796204531538fe85d : SUCCESS in 16m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/96753fa003a94d2b9260ea18c8292414 : SUCCESS in 21m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/193d3b1df0c442feaca201002b5a0e53 : SUCCESS in 13m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/aebe59c96c0a4145b237c2ed664fd1cf : SUCCESS in 15m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/13d827e9e8694e2f940ae62f3d32196f : SUCCESS in 8m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/85ec1d7a000945859653a956dcfb7950 : SUCCESS in 8m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/15d2f0fab7de40318c0c395fa58d48d1 : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ea962a5dadc04bc2be78dcd89d206db9 : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/40f4427888a44bcbba606575a6b25cd6 : SUCCESS in 55m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/00255997117c4499a416303fb7cd5d34 : SUCCESS in 44m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3efe83008ddf413d833f454066b09889 : SUCCESS in 51m 59s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/435c59f767104bb8bc104f778bc68668 : SUCCESS in 13m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8370383d71e44617b54882b84c6125ec : SUCCESS in 3m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b871db875cc1489ea2424ff8086c34ee : SUCCESS in 19m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a6ba3304a52f4334afd6538b0e2cca50 : SUCCESS in 38m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f5285b3b9b04213a383d23bae492549 : SUCCESS in 51m 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"293b5c6e19eb86e234149e37812284237e90d544","revisions":{"7a55ca722cb06271d315206452c2bfcbfda99297":{"kind":"REWORK","_number":1,"created":"2025-05-08 01:21:23.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/949108/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/949108/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/949108/1"}}},"commit":{"parents":[{"commit":"076771462cf5333644170b069ad95e8bd999d024","subject":"Merge \"Fix the inconsistent use of printing errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076771462cf5333644170b069ad95e8bd999d024"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 01:20:44.000000000","tz":-240},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 01:20:44.000000000","tz":-240},"subject":"Fixing constant creation-date for s3api list-buckets","message":"Fixing constant creation-date for s3api list-buckets\n\nChange-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a55ca722cb06271d315206452c2bfcbfda99297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a55ca722cb06271d315206452c2bfcbfda99297"}]},"branch":"refs/heads/master"},"24902b17f8c072f965988fd0bdfaf08b3221e246":{"kind":"REWORK","_number":2,"created":"2025-06-22 18:45:50.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/949108/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/949108/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/949108/2"}}},"commit":{"parents":[{"commit":"076771462cf5333644170b069ad95e8bd999d024","subject":"Merge \"Fix the inconsistent use of printing errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076771462cf5333644170b069ad95e8bd999d024"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 01:20:44.000000000","tz":-240},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-06-22 18:45:17.000000000","tz":-240},"subject":"Fixing constant creation-date for s3api list-buckets","message":"Fixing constant creation-date for s3api list-buckets\n\nChange-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24902b17f8c072f965988fd0bdfaf08b3221e246"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/24902b17f8c072f965988fd0bdfaf08b3221e246"}]},"branch":"refs/heads/master"},"3bd171c62f8eefe73ae6b86b074a428fcdb1fd57":{"kind":"REWORK","_number":3,"created":"2025-07-05 20:56:37.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/949108/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/949108/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/949108/3"}}},"commit":{"parents":[{"commit":"076771462cf5333644170b069ad95e8bd999d024","subject":"Merge \"Fix the inconsistent use of printing errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076771462cf5333644170b069ad95e8bd999d024"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 01:20:44.000000000","tz":-240},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-07-05 20:56:24.000000000","tz":-240},"subject":"Fixing constant creation-date for s3api list-buckets","message":"Fixing constant creation-date for s3api list-buckets\n\nChange-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61\nSigned-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bd171c62f8eefe73ae6b86b074a428fcdb1fd57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3bd171c62f8eefe73ae6b86b074a428fcdb1fd57"}]},"branch":"refs/heads/master"},"293b5c6e19eb86e234149e37812284237e90d544":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-07-05 21:00:57.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/08/949108/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/08/949108/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/08/949108/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/08/949108/4"}}},"commit":{"parents":[{"commit":"076771462cf5333644170b069ad95e8bd999d024","subject":"Merge \"Fix the inconsistent use of printing errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076771462cf5333644170b069ad95e8bd999d024"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 01:20:44.000000000","tz":-240},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-07-05 21:00:57.000000000","tz":0},"subject":"Fixing constant creation-date for s3api list-buckets Closes-Bug: #1856938 Change-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61 Signed-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e","message":"Fixing constant creation-date for s3api list-buckets\nCloses-Bug: #1856938\nChange-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61\nSigned-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/293b5c6e19eb86e234149e37812284237e90d544"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/293b5c6e19eb86e234149e37812284237e90d544"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"}},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
