)]}'
{"id":"openstack%2Fswift~949121","triplet_id":"openstack%2Fswift~unmaintained%2Fyoga~If724485e1425d1481d10b9255436301e346f07e8","project":"openstack/swift","branch":"unmaintained/yoga","topic":"fix-head-content-type","hashtags":[],"change_id":"If724485e1425d1481d10b9255436301e346f07e8","subject":"Ensure correct content-type in container HEAD response","status":"MERGED","created":"2025-05-08 06:34:13.000000000","updated":"2025-05-08 08:57:49.000000000","submitted":"2025-05-08 08:57:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949121-fix-head-content-type","meta_rev_id":"807f78e8c803e6ad1a79275dbc792149cfdbd4dd","_number":949121,"virtual_id_number":949121,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-08 08:57:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2025-05-08 08:04:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2025-05-08 08:04:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 06:34:13.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2025-05-08 07:54:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2b5e9fec2bfb9062b7b832bc591fc85e4538070","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-08 06:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0f1c7f6844a2dddddf431308ede0ce103c5da01c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-08 07:00:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a41204df1dcb9bc27d09cccb8e0ac34524d29ba5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 07:54:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24f8f9d4d7d449bcb1ca438fb9a86f8c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a8f449288df40bcba541c9117588aca : SUCCESS in 6m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ec62d51c4cb14374bb42121504705944 : SUCCESS in 16m 56s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af527337b0ef4627adc819fdb7bd0792 : SUCCESS in 17m 24s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dac2c0244edc4e59b3d13ecb90eacb8e : SUCCESS in 17m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d7e5c13218a64d85a7099ae118a688c9 : FAILURE in 5m 28s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/82880f53a43b47a19f4692eed1490e2b : FAILURE in 3m 29s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/b71e63feec634507a529df699f678fed : SUCCESS in 15m 56s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/bffe27cc01444e00bf28ab0d523ea084 : SUCCESS in 15m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e9aa3450ac8846c38d750c1d01344968 : SUCCESS in 13m 40s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/b43021e0eca14feba269ba5316157e68 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/bdfc860987db43baab980c8bb381e415 : SUCCESS in 16m 34s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/b1939a667ab34b0eb1f7138edde220c9 : SUCCESS in 17m 16s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b8b03bee132b43d38a4b134ccdbfdf96 : SUCCESS in 13m 54s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/349f586212934d9eb8f0f84e69c1a9d5 : SUCCESS in 14m 48s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/b4bb6c9aa7a3443bbaa7ca0621f933c5 : SUCCESS in 16m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/682f96d8559b4d348449d867438355b8 : SUCCESS in 6m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/522c7eaacda64daabcc12a9cbe82a904 : SUCCESS in 15m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/23875a8e23e64f6faf1e478c080b7881 : SUCCESS in 41m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ae66b0c37ec94ae595940a8dc36bd2c5 : SUCCESS in 17m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44a33ae7359e45baac1f01aece682ff2 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6a16f2f918804d46b5d9f3ff69b3a766 : FAILURE in 5m 53s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fb2f7f6082194f6ea7d37957060b121c : SUCCESS in 48m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6b0295b0222c41fdacd92a5d58c0039c : SUCCESS in 47m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/dfce6e81b1ba45b783bc8b91f770f26e : FAILURE in 5m 40s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2058f49b3da24804840cb93298c50961 : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":2},{"id":"1f11fc2e6424c4fcaaaac0d1000d74145a1793d7","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-08 08:04:49.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5f0d0d3278c3e3b1f17f636e61ca57f8e0937c7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:05:08.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"86bb83d9bc73e6297b200c0faa3fd3f95de64bd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:32:20.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8046dca7f1654a939eab40260eed1b5e\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/22e242f7854842fbb0ccf3de07d96107 : SUCCESS in 26m 17s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/92d8e0c3e8924f63b70bd49978d78cba : TIMED_OUT in 1h 30m 51s\n- swift-probetests-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/a4fc01f77e374a3cb41d8ee59e192e6c : SUCCESS in 1h 26m 20s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/576fd2d13b624434b4ac2eddc28294ea : SUCCESS in 33m 05s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/45dd3b6c166d44d9be469ee2bb05351c : SUCCESS in 28m 39s","accounts_in_message":[],"_revision_number":2},{"id":"dd4ea2efd096a73d5e452f07e749c600fb463192","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:57:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a04093ee534f416fb8d79d72f4addb7d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f270d33b1a8402da76681343591fb6c : SUCCESS in 7m 17s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/f4b81bd106154450b9462e0385395382 : POST_FAILURE in 4m 51s (non-voting)\n- swift-upload-image-py3 https://zuul.opendev.org/t/openstack/build/2664b06240eb403294324f0912655b6e : POST_FAILURE in 6m 46s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/78073c77eb944ccba8108a7005735c62 : SUCCESS in 16m 11s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/a5ac2fbace274fa5807365d19f038776 : SUCCESS in 8m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e4f3b9b78074ddeafd2dd1dde02ef45 : SUCCESS in 16m 10s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/d67e3a539a8c40b5a60f895229656565 : SUCCESS in 10m 39s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/07d963472eeb45478e3c034ac9d4f936 : SUCCESS in 16m 36s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/172eb02aac764b328b36b0f7e4d642ed : SUCCESS in 11m 08s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/4a9bcda4466e4d87b22ecdd5b5c77853 : SUCCESS in 8m 48s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/ec119c5b32b44790ad57fcbf462e557d : SUCCESS in 15m 31s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/bc8978773ee143ad9a8382168898e6a0 : SUCCESS in 14m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7bddb04058f441d5b49976e040671f23 : SUCCESS in 6m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/52fa7f9e842346aeaf68765305b444a3 : SUCCESS in 51m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b08df8b29a7147b6b076e03b281040fc : SUCCESS in 37m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a1d66222a69b479fa4eebdbf9dbbf14a : SUCCESS in 41m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5312b9997ebb470d83709f1a0cfc1675 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a24a55dce274b39ac234d33f55eaab6 : SUCCESS in 4m 09s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b6380f0969f24ba190950bf001abd2e3 : SUCCESS in 50m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/504bb1e96275439bbd3f42c643079edb : SUCCESS in 34m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83a0f338cc6e49cf88901eff296d6a26 : SUCCESS in 35m 03s","accounts_in_message":[],"_revision_number":2},{"id":"b729e12d2802e9d908ec40fa2e2749239d08ba48","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:57:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"807f78e8c803e6ad1a79275dbc792149cfdbd4dd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 08:57:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c92cea20a97549bea70e1280be86b820\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d7d7c11c370647d19af321beb335358d : SUCCESS in 36s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d0b1b4e29904fa98fe3b7f9341dd6f6 : FAILURE in 27s (non-voting)\n- swift-promote-image-py3 https://zuul.opendev.org/t/openstack/build/9755305f57954eb492ff6dad68672376 : FAILURE in 27s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"47c4a1df5385133dcd9c764a1e288427c7c6d2de","revisions":{"9f26f4a42a5347250a1a5c0f986eeb7804344472":{"kind":"REWORK","_number":1,"created":"2025-05-08 06:34:13.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/21/949121/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/949121/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/949121/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/949121/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/949121/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/949121/1"}}},"commit":{"parents":[{"commit":"d8a309c756a3d9ef96812b7012f0a6a4c60f6e82","subject":"[CI] Use defined version of probetest job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d8a309c756a3d9ef96812b7012f0a6a4c60f6e82"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:59:43.000000000","tz":-360},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-05-08 06:33:36.000000000","tz":120},"subject":"Ensure correct content-type in container HEAD response","message":"Ensure correct content-type in container HEAD response\n\nA failing CORS test in the gate discovered that we were responding\napplication/json to ?format\u003dtxt requests (which is maybe not even a\nvalid value for that qs param?), but only when running with\neventlet\u003d\u003d0.38.0\n\nThis avoids the problem of backend container server HEADs no longer\nhaving \u0027Content-Length: 0\u0027 by fixing the client HEAD resp headers before\nwe check for chunked-transfer resp.\n\nDrive-By: refactor listing_formats to use HeaderKeyDict and always set\nContent-Length explicitly\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: If724485e1425d1481d10b9255436301e346f07e8\n(cherry picked from commit fa889358acf7675efba2de095777886b4c1ff7f8)\n(cherry picked from commit 6aa424f181bcdb7f49db9b13c02bba1b8c03dd7e)\n(cherry picked from commit b8c1c223612f78b55658dea8498783f0b4ab69b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f26f4a42a5347250a1a5c0f986eeb7804344472"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9f26f4a42a5347250a1a5c0f986eeb7804344472"}]},"branch":"refs/heads/unmaintained/yoga"},"47c4a1df5385133dcd9c764a1e288427c7c6d2de":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-08 07:00:00.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/21/949121/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/21/949121/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/949121/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/949121/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/21/949121/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/21/949121/2"}}},"commit":{"parents":[{"commit":"1a18d3c16bbec95da54aa6b2a8c3e1469181d799","subject":"CI: Move off CentOS 8","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1a18d3c16bbec95da54aa6b2a8c3e1469181d799"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:59:43.000000000","tz":-360},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-05-08 07:00:00.000000000","tz":0},"subject":"Ensure correct content-type in container HEAD response","message":"Ensure correct content-type in container HEAD response\n\nA failing CORS test in the gate discovered that we were responding\napplication/json to ?format\u003dtxt requests (which is maybe not even a\nvalid value for that qs param?), but only when running with\neventlet\u003d\u003d0.38.0\n\nThis avoids the problem of backend container server HEADs no longer\nhaving \u0027Content-Length: 0\u0027 by fixing the client HEAD resp headers before\nwe check for chunked-transfer resp.\n\nDrive-By: refactor listing_formats to use HeaderKeyDict and always set\nContent-Length explicitly\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: If724485e1425d1481d10b9255436301e346f07e8\n(cherry picked from commit fa889358acf7675efba2de095777886b4c1ff7f8)\n(cherry picked from commit 6aa424f181bcdb7f49db9b13c02bba1b8c03dd7e)\n(cherry picked from commit b8c1c223612f78b55658dea8498783f0b4ab69b1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c4a1df5385133dcd9c764a1e288427c7c6d2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/47c4a1df5385133dcd9c764a1e288427c7c6d2de"}]},"branch":"refs/heads/unmaintained/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
