)]}'
{"id":"openstack%2Fswift~949157","triplet_id":"openstack%2Fswift~unmaintained%2Fxena~If724485e1425d1481d10b9255436301e346f07e8","project":"openstack/swift","branch":"unmaintained/xena","topic":"fix-head-content-type","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2025-12-19 15:00:48.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"If724485e1425d1481d10b9255436301e346f07e8","subject":"Ensure correct content-type in container HEAD response","status":"ABANDONED","created":"2025-05-08 11:30:54.000000000","updated":"2025-12-19 15:00:49.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"d7553b6ec7fc6704704b5f7b354b69a9aad219d7","_number":949157,"virtual_id_number":949157,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-08 12:08:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 11:30:54.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2025-05-08 12:08:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6e3f6416ce3f5e1181840dd2051f21be27153ca7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-05-08 11:30:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e9be27042c9df7a6dc9115a57f6f36333dd0157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 12:08:40.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2ee2f6912f9c47bda7c4a6b44cbbde63\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c15e0ccc2bac4cd9b3eac2798578113d : SUCCESS in 7m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/612d8a710daf45ea99fa0f5b48ad2f5c : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c9cc89215b814bd1bfe2f53d8f2a3a26 : SUCCESS in 16m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b3dcf7d59a314ba08dd0ad4841f447b4 : SUCCESS in 23m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/76da61e41f394e44aed8d3bad617d31c : FAILURE in 3m 39s (non-voting)\n- swift-build-image-py3 https://zuul.opendev.org/t/openstack/build/6a2e28b60b4a48bcab0f1a26eff1a409 : FAILURE in 5m 00s (non-voting)\n- swift-tox-py27 https://zuul.opendev.org/t/openstack/build/69fc48b03f04439ebd4479564d00439a : SUCCESS in 14m 07s\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/795c9b4ac0344637b3f5cfcc46d2bd14 : SUCCESS in 14m 41s\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/07f556bf51184907b17967bfc835cb65 : SUCCESS in 15m 51s\n- swift-tox-py38 https://zuul.opendev.org/t/openstack/build/c52c12e93a0f41afb20a4cb3326f113d : SUCCESS in 16m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5e849b0561e84528bc12c1cd00b1df7c : SUCCESS in 14m 17s\n- swift-tox-func-py27 https://zuul.opendev.org/t/openstack/build/cb218596d49c4d77874dfdcbafcf965f : SUCCESS in 8m 10s\n- swift-tox-func-encryption-py27 https://zuul.opendev.org/t/openstack/build/ae19002f89c04355baa02dd9ef5035cb : SUCCESS in 8m 44s\n- swift-tox-func-ec-py27 https://zuul.opendev.org/t/openstack/build/bf467b2950704f3b93056dd7628fd144 : SUCCESS in 17m 23s\n- swift-tox-func-py38 https://zuul.opendev.org/t/openstack/build/b18170f7ff25446f9c903ccc3b083f1c : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py38 https://zuul.opendev.org/t/openstack/build/88fd363775f54af5bec33fa25d8e1fec : SUCCESS in 8m 37s\n- swift-tox-func-ec-py38 https://zuul.opendev.org/t/openstack/build/cf83dcea7e6048588060f4e85f56ffc7 : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e6a11d6c2c8442ebaa5176974aac9740 : SUCCESS in 7m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b659e38a2dc647469dc7a867fe8351a0 : RETRY_LIMIT in 9m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8ef60286d2f045a3a8a1a72f18d413b5 : RETRY_LIMIT in 7m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/91b0965777fc403e9f422044a612b144 : SUCCESS in 12m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b09b5a45d8c4d8e9181b6556de97085 : SUCCESS in 4m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c68f28213c6e400aa2df411ee7d24fa3 : FAILURE in 6m 38s (non-voting)\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1dfb06e0ac96408894baf471882239a2 : FAILURE in 5m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7e9c6427e2f646f188be1afdd0a85848 : FAILURE in 12m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/587904b235904134b633f11414cdf311 : FAILURE in 11m 00s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b798ca1d7ef344d1805e77e3877c7149 : POST_FAILURE in 11m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b6412c92bca1bb5b446e7213954d0da6ee719ed","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 13:05:13.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8eb8204400d14170891a456ce96fbb48\n\n- swift-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/165144ebcdd340fdb6b97618af759b43 : SUCCESS in 39m 56s\n- swift-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3c0fafcc8ec64a469c299a9b17b1aacf : SUCCESS in 28m 10s\n- swift-probetests-centos-9-stream-arm64 https://zuul.opendev.org/t/openstack/build/bd1309cb33a849f5992d2cd556b8637b : FAILURE in 1h 25m 48s\n- swift-tox-func-encryption-py38-arm64 https://zuul.opendev.org/t/openstack/build/d434d308fcd74cd78655e06ea520b78c : SUCCESS in 37m 34s\n- swift-tox-func-py38-arm64 https://zuul.opendev.org/t/openstack/build/964c192f00e649059c3eb1161a1f9ea2 : SUCCESS in 48m 00s","accounts_in_message":[],"_revision_number":1},{"id":"d7553b6ec7fc6704704b5f7b354b69a9aad219d7","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2025-12-19 15:00:49.000000000","message":"Abandoned\n\nunmaintained/xena branch of openstack/swift transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5a62897136e899c9124b60fd7030054307d1d500","revisions":{"5a62897136e899c9124b60fd7030054307d1d500":{"kind":"REWORK","_number":1,"created":"2025-05-08 11:30:54.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/57/949157/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/57/949157/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/949157/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/949157/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/57/949157/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/57/949157/1"}}},"commit":{"parents":[{"commit":"4e35060889940f237b2136b7c5291fada9c875e5","subject":"Merge \"Update .gitreview for unmaintained/xena\" into unmaintained/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e35060889940f237b2136b7c5291fada9c875e5"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2024-11-20 16:59:43.000000000","tz":-360},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2025-05-08 11:30:19.000000000","tz":120},"subject":"Ensure correct content-type in container HEAD response","message":"Ensure correct content-type in container HEAD response\n\nA failing CORS test in the gate discovered that we were responding\napplication/json to ?format\u003dtxt requests (which is maybe not even a\nvalid value for that qs param?), but only when running with\neventlet\u003d\u003d0.38.0\n\nThis avoids the problem of backend container server HEADs no longer\nhaving \u0027Content-Length: 0\u0027 by fixing the client HEAD resp headers before\nwe check for chunked-transfer resp.\n\nDrive-By: refactor listing_formats to use HeaderKeyDict and always set\nContent-Length explicitly\n\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-Authored-By: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n\nChange-Id: If724485e1425d1481d10b9255436301e346f07e8\n(cherry picked from commit fa889358acf7675efba2de095777886b4c1ff7f8)\n(cherry picked from commit 6aa424f181bcdb7f49db9b13c02bba1b8c03dd7e)\n(cherry picked from commit b8c1c223612f78b55658dea8498783f0b4ab69b1)\n(cherry picked from commit 47c4a1df5385133dcd9c764a1e288427c7c6d2de)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a62897136e899c9124b60fd7030054307d1d500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a62897136e899c9124b60fd7030054307d1d500"}]},"branch":"refs/heads/unmaintained/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
