)]}'
{"id":"openstack%2Fswift~949517","triplet_id":"openstack%2Fswift~master~Ia7ca818b510f65a6dc045d0ac9af69f0b85acd8d","project":"openstack/swift","branch":"master","topic":"s3-chunk-sig-validation","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-12 18:45:02.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia7ca818b510f65a6dc045d0ac9af69f0b85acd8d","subject":"S3Request: move more chunk validation into ChunkReader","status":"ABANDONED","created":"2025-05-12 15:29:42.000000000","updated":"2025-05-12 18:45:02.000000000","total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"f367035ce5ce89ae3c5b13fadb5718c2d7e4899e","_number":949517,"virtual_id_number":949517,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-12 16:32:34.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-12 18:26:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-12 16:19:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-12 16:32:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-12 18:26:06.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"a5ccfa8784c8d71f53416f5f691d072b5bb1a7e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-12 15:29:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"399832e68d1ad3efca6ff143880bb4eb6a5b01ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 16:19:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8df14874d5de4335877bf56597f3b1a9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/186af260661846d78d0159d9ed4958dc : SUCCESS in 48m 01s","accounts_in_message":[],"_revision_number":1},{"id":"e7ce5e14c0499f9d0b5732d0d2b8e5f0dfd212e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 16:32:34.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c65919f10345d2b0db8088f672f014\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1697fbd057474e69b0ec8350c50ef715 : SUCCESS in 10m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/6534e305a0e24e5f900d5caf70f26c5b : SUCCESS in 41m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1793ca7a77c348a4bdb7d67f4d5ff415 : SUCCESS in 55m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e770bb54720469ab10d92469e40d16f : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5e5832d1578c43929b18ab1b9b3e01ba : SUCCESS in 38m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/90429b5c6fec499fa54d48afa2f998ad : SUCCESS in 18m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de2896308b854c3da71eff9df5fa73e4 : SUCCESS in 12m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70fca959952c440bbf68913437afa76f : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b2a4431d31b84b4597ae8bea09efb29c : SUCCESS in 4m 23s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/289ab16a88b34ad69453ca2dd1ff3820 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c120fbe550c24524b66ee44edacd3e04 : SUCCESS in 9m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b5b378a8aa174d03a9562e1b1fcc33f1 : SUCCESS in 31m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/391d68205c484bab8d5177ca95a84fd8 : SUCCESS in 10m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c4a3e0c3e79c42da9b121c0ac3e7cebc : SUCCESS in 7m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f9790a9f974b43399ed1a8ca73777a7b : SUCCESS in 14m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6f87a202220949a09a3ae267967375ce : SUCCESS in 13m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f904058137234459a3b68def20815ca0 : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/11763ef209ed48c7b5adf04e3251f508 : SUCCESS in 14m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/53dfb3187a074485bd3e02f9b0aa8648 : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9d0e484cbc4457683e196eb5644ff48 : SUCCESS in 48m 42s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4ce26167264e4760a5bf7075bb5ecb81 : SUCCESS in 38m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1bd74129be834176a497b1b94246c600 : SUCCESS in 33m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1f1f22a1e0ad4e23bf96a49f26951043 : SUCCESS in 18m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11b6d956f4fd450cb3db2c8c383f7d61 : SUCCESS in 4m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d4d1f1a34199418ca5d44e9b318ab54d : SUCCESS in 20m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ca817f055b654678b416a8390185d0e6 : SUCCESS in 52m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bcc4a834feee4887be09423dd9e324ee : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"b0a77ac2fec4afcd42c22c087c06543b0fe262ef","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-12 18:26:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f367035ce5ce89ae3c5b13fadb5718c2d7e4899e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-12 18:45:02.000000000","message":"Abandoned\n\nSquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0e5aff9723fc81c3001e02b1be9bc20a4d0ecca7","revisions":{"0e5aff9723fc81c3001e02b1be9bc20a4d0ecca7":{"kind":"REWORK","_number":1,"created":"2025-05-12 15:29:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/17/949517/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/17/949517/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/949517/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/949517/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/17/949517/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/17/949517/1"}}},"commit":{"parents":[{"commit":"b236cd5fe95e49a52e5b169994f1c7898e2475af","subject":"Add support of Sigv4-streaming","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b236cd5fe95e49a52e5b169994f1c7898e2475af"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-12 15:26:57.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-12 15:26:57.000000000","tz":60},"subject":"S3Request: move more chunk validation into ChunkReader","message":"S3Request: move more chunk validation into ChunkReader\n\nRefactor so that chunk signature validation is performed by\nChunkReader. This encapsulates all interaction with\nSigChecker.check_chunk_signature in the ChunkReader class.\n\nChange-Id: Ia7ca818b510f65a6dc045d0ac9af69f0b85acd8d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aff9723fc81c3001e02b1be9bc20a4d0ecca7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0e5aff9723fc81c3001e02b1be9bc20a4d0ecca7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
