)]}'
{"id":"openstack%2Fswift~949671","triplet_id":"openstack%2Fswift~master~Ife37041c1e83eea3e108ad80101d84a5ebf70f2b","project":"openstack/swift","branch":"master","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-11-06 01:20:34.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-06 02:25:47.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ife37041c1e83eea3e108ad80101d84a5ebf70f2b","subject":"s3token: Ensure check_signature gets called when caching secrets","status":"NEW","created":"2025-05-13 19:37:47.000000000","updated":"2025-11-13 00:34:37.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"ee0a9f6813354dc5f1dcdb22031daa3e3c9bfe74","_number":949671,"virtual_id_number":949671,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-11-12 22:45:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"}],"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 20:38:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-13 20:56:23.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 18:34:50.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2025-11-06 02:19:04.000000000","updated_by":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"reviewer":{"_account_id":34892,"name":"ASHWIN A NAIR","display_name":"indianwhocodes","email":"nairashwin952013@gmail.com","username":"indianwhocodes","status":"Nvidia"},"state":"CC"}],"messages":[{"id":"f946a17b0eef980fa38da0a7e722df6f5f127a63","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-13 19:37:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7e57f2fd4e5c9d0b951bcad0232d582d155939d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 20:38:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c42c8073fc42c38c4b1c54368359dd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6e768416d29144f79b833ccee55708ac : SUCCESS in 59m 00s","accounts_in_message":[],"_revision_number":1},{"id":"2db96a9e1467f9e0b8c83bf65bc677991b913283","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 20:56:23.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/45234e63e7634126b125cb4e7bfa0fad\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47f862eb2a0d4de38f73108f8e712a0a : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a85e9c1e2ceb46219f966d508a89720e : SUCCESS in 32m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d6eae263cbd461a95f3319a115a95b7 : SUCCESS in 29m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c91ce48518a846268ab43f2df7ba3e20 : SUCCESS in 48m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1928713d9e604d11a326411e682cab9f : SUCCESS in 52m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80dd08fc40924de0b7782044a86643a9 : SUCCESS in 15m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cbb437d933464bf5a214ab924e4f15fd : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d5f2360ce3b747579e0538f930bf7a84 : SUCCESS in 17m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/449b21c5ed4347428f250ef3b184cebe : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/286624e23f364cff9feeef586a317ee1 : FAILURE in 2m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e8ab40c947b947e483b8557c62eab3f6 : SUCCESS in 22m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a7c1fc06594b44209de1cbe4689078a3 : SUCCESS in 29m 01s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d75fb701038d4fe295b30637954552a3 : SUCCESS in 25m 38s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/17dac6aaaf014faeb551e5ff1ec8350f : SUCCESS in 12m 41s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e9da344eb5a84c2fa7e93e83c0b3ef5a : SUCCESS in 12m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d0b1cf6c49394f6ab393132201efbdba : SUCCESS in 13m 39s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6e20f63d564646aa9d06ac23f621aeec : SUCCESS in 4m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/334628fa060843939a937d621293cfab : SUCCESS in 13m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b0b6fa6008447369eda252a08a61a04 : SUCCESS in 7m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fdd1897571d4d5b882ff2765ce873f6 : SUCCESS in 49m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9e211b0c7a5c42109e0b93057c85d58f : SUCCESS in 34m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6fe65abe79454063b9f308e99372510c : SUCCESS in 39m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c4b1ac09596c4479ab1bc3b11012fc95 : FAILURE in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3014409dd914dad80b9461e99f257b5 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3a15728c249a4b76b8f79a0095e1b712 : SUCCESS in 21m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/39e472ee6eee45fea31778f118f3a43f : SUCCESS in 50m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2268fc44933449c8f619eaae82131d9 : SUCCESS in 51m 06s","accounts_in_message":[],"_revision_number":1},{"id":"804ce75175dbe3387eb751367adf178a8c045ce9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-15 03:56:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c2cad68405765819029d39f4118225ce60e7a94f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:50:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d5d52b23fc94900a9aa30b8863f60cb\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e4a149dd410f470f8c839c95bf07e9e7 : SUCCESS in 51m 30s","accounts_in_message":[],"_revision_number":2},{"id":"6542983a0482649c07550bc0894e1ce8abd1c378","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 05:02:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/918ecb2425f049aba382ac7fed00f56e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd387de77f4a417cb7a229c279176d36 : SUCCESS in 4m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/a8a6a29618324dd68a7b2db313e627d1 : SUCCESS in 49m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb4447d128834363b4eae9bc3bcae03a : SUCCESS in 1h 05m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/02200f46797742059d4fed195602cf56 : SUCCESS in 43m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8fcd0fab9df94ba4bf9544c31ffad6fd : SUCCESS in 57m 11s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/00bda74597a9486ea9064d3736cd39cc : SUCCESS in 21m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9248598049d14d379d9e430edb335d6f : SUCCESS in 18m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/060821515a3d4155ad84a4fc74f4ff66 : SUCCESS in 18m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/573f3c13929048b298d29cef4ee9907a : SUCCESS in 4m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fee08d1de7314cd693969bf7d4dfc0ee : SUCCESS in 20m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f732bee4acea49e0b777640bb98014f3 : SUCCESS in 16m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/98513df6414f44ae82d1c0c9a0cbb787 : SUCCESS in 29m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/66be49514f484872b31eb7af8bedb377 : SUCCESS in 24m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/810b0263eda6410eac3b2e5d6cc78185 : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/27bd143cb5aa4206a112cd1616411c0c : SUCCESS in 9m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f7b0848b70ec4bc6bdf7b8a8bf2e8531 : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/226304b6398b47bfada253a68923212e : SUCCESS in 7m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fff82def16646a28f6fb137621c002f : SUCCESS in 9m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ae9d02d9f2844c92ac40092fdc1feaa7 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1f3efa1fa5e245b4849c6befcb75d864 : SUCCESS in 46m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1828780aac174e859699ff1fc9eca28d : SUCCESS in 48m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5607180d6f25429c8a167ffe09d6b6e7 : SUCCESS in 43m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cef1a64bd82a4bd3b80dc97e750da773 : SUCCESS in 16m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db76ab975cd0498b891d548105ad0ca7 : SUCCESS in 3m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce0ce8b1a1344b42abb25058145783a4 : SUCCESS in 18m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d6afcab7c51b4becbb5cc5c76174162b : SUCCESS in 35m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11b146d9b75f4f8d802dfda1912edc3d : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":2},{"id":"f790727c71de8f74c531d22621c2b37368c8218c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-28 18:34:50.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfde0e65c5bf9c59ca1c62255814dbac829dc977","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-29 10:06:05.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"94758928dcdb1f9d1fb53f249cec670473b480c0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-29 16:40:05.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"64d2af659f7bab979cde97253893c86b85b1fcd3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-29 16:52:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5ffba1c42538e6c0e6a117f39fd3539b4913383b","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-29 21:40:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1d541e7625079d2da91da2252b7a790faa6e76c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-06 01:20:34.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ce9b3b55fbe6e4aad3c58932b2d09e36f44db0a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-06 01:20:41.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"b1eebd919239a31550ee2e5d85bf79ec464a24f1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 02:21:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b413b8a0ea245ec8a1f45b9a83b728a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/90bc8055de904b0f8126997e65cf0fac : SUCCESS in 59m 27s","accounts_in_message":[],"_revision_number":4},{"id":"d43aa0263a5945975303fadcfc14f0cb9650c834","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 02:25:47.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd888e9c64084346a535ea4e5717cccb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4520d4aabb7a44c4b7485b60e07cc31b : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/e0f1b3116a5440d7bc08ef6844e224c9 : SUCCESS in 1h 00m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/547f70c074124b66807f348d8fc212cc : SUCCESS in 45m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/099988089a174cfcadf0ca66ff8c818d : SUCCESS in 40m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/32adc42b47de49bcaf49086afe32ca38 : SUCCESS in 59m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d46f468480e54051a2a5be6d1dcc499b : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e4f90b897e9e48c48ed0298d96ac5791 : SUCCESS in 21m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ba0dec09af340b0ab9335c5c9122c2d : SUCCESS in 14m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/623dd0571d8c46a4bb874931736483d5 : SUCCESS in 6m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a8b80bb89224729baf4fb41b25bc43e : SUCCESS in 17m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97d86b9030314a2a80acd806a3dd82e1 : SUCCESS in 20m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dacb4af617534729af2907240ffee6be : SUCCESS in 19m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dd0b4f81e4e746f48eaf5ddcc076760c : SUCCESS in 13m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1519b2a072344f669ea483ef4f4e3a7a : SUCCESS in 13m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8139230fc0574ff0b47bfdb0aab6a225 : SUCCESS in 13m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0561e9c2bd5d44229cc1cbb89aac26ae : SUCCESS in 15m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/31afafd1eaef4a878b0b09ef7377d48c : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/94cb3e38238e4774afff8012686acf65 : SUCCESS in 17m 12s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b9ae1c6fd8ad4f03b533dbfb9fc53b11 : SUCCESS in 8m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6be65b3d85d43a69cdeb43140709159 : SUCCESS in 58m 02s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4a4aa0428a294b59b816c37ebc9906c5 : FAILURE in 33m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/db9cf01fe7c5491186619f898896e910 : FAILURE in 32m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/680c107a32db4386b340b57b15382757 : SUCCESS in 26m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5eae895626b406fa6fb9cd75ca50c75 : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ce7440c7cc5a4b168273da9bab837fd5 : SUCCESS in 10m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/10c9261142fb4132a2e94ec6bff4bffa : SUCCESS in 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f674a45f61154bc9bf9fa4ffced4a915 : SUCCESS in 56m 56s","accounts_in_message":[],"_revision_number":4},{"id":"402f140323a14a875b1ec608ce7938b0a14f3fe3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-12 21:41:10.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"00abde72a454333db6beecbdf3639fe14686faef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 22:45:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b7fd31754e24992b1f45c4bd9060a95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00a61ca8baf54216bd1027f051af6b8d : SUCCESS in 10m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/73b209a5b3bb44f6b68055f7e5fb5b24 : SUCCESS in 34m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/eeb9565491a14846add6f370519ba265 : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/253719d8afc84c9389a0bb7a744a1e8c : SUCCESS in 55m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ca5fc0235514f6fa09fa6ed0b087cf3 : SUCCESS in 37m 03s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4cfb8485eb3a4a209eafdea0b6f51aef : SUCCESS in 16m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fafcbeb30c3d44c6899be87aebed0f3e : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7844bc4051544c81a38e3396eb8a7acf : SUCCESS in 16m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8da8df17eb894d5880d82e349d3a29eb : SUCCESS in 6m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/633c4d2c47da4d7d8b59f305a6218914 : SUCCESS in 12m 17s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dfeb567864844dd39ac5eb08635793c9 : SUCCESS in 21m 55s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/79bb5fd2c84b4f4298a7566f46ea7021 : SUCCESS in 28m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cd193407e9394af88ba7a451969c1d46 : SUCCESS in 27m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7392a2fbda5e4076839cd9c329cd741e : SUCCESS in 9m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87b41e1e8bf7458daad7057a0b851207 : SUCCESS in 13m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/95a8f245ff9a4f7aac57cc783098cdc7 : SUCCESS in 9m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/34c5708b89e2401994bfedb217cb53f2 : SUCCESS in 4m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9bedb7babb9f4cd298d986a69f7dceda : SUCCESS in 15m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/80b064c6302345238dc92ac27ea72fee : SUCCESS in 8m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7e3975866c34eeca246512ab3eb842c : SUCCESS in 38m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/45686a2a83eb42818c808676eeac441d : SUCCESS in 24m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/18a8e92624c6443781f73fe321836957 : SUCCESS in 30m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c7993771ac984522997ad77f897121dd : SUCCESS in 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0f238586ed948808e4838d86ffe831a : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/20956ce9754d4c80abfa5d12fef9ef14 : SUCCESS in 24m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/db1ef4856a524de48258546494188669 : SUCCESS in 53m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f6d858e1ca1248c7adcb3a16cf7e30f8 : SUCCESS in 32m 03s","accounts_in_message":[],"_revision_number":5},{"id":"ee0a9f6813354dc5f1dcdb22031daa3e3c9bfe74","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-13 00:34:37.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5023ffe141994f6296616373dec0f39d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bb96651a3c9546b2bdffd12b59ef3842 : SUCCESS in 37m 47s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"13b0597ffb5f7187ccaf14ff627c495d06146d32","revisions":{"ddb8b3beab65551c35febd7c803d115c5d477e2d":{"kind":"REWORK","_number":1,"created":"2025-05-13 19:37:47.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/949671/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/949671/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/949671/1"}}},"commit":{"parents":[{"commit":"3cf1f56a73a1a558a1bda302fcf2bc713747954f","subject":"Add labeled metrics to s3api","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3cf1f56a73a1a558a1bda302fcf2bc713747954f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 18:59:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 19:37:35.000000000","tz":-420},"subject":"s3token: Ensure check_signature gets called when caching secrets","message":"s3token: Ensure check_signature gets called when caching secrets\n\nChange-Id: Ife37041c1e83eea3e108ad80101d84a5ebf70f2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddb8b3beab65551c35febd7c803d115c5d477e2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ddb8b3beab65551c35febd7c803d115c5d477e2d"}]},"branch":"refs/heads/master"},"5d8a12e998df4924b178a74099ba51df839cee64":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-15 03:56:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/949671/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/949671/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/949671/2"}}},"commit":{"parents":[{"commit":"6212869399d943305c8b1f2e0a20d9cd2cd71b75","subject":"Merge \"proxy_logging: remove unnecessary condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6212869399d943305c8b1f2e0a20d9cd2cd71b75"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 18:59:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-15 03:56:19.000000000","tz":-420},"subject":"s3token: Ensure check_signature gets called when caching secrets","message":"s3token: Ensure check_signature gets called when caching secrets\n\nChange-Id: Ife37041c1e83eea3e108ad80101d84a5ebf70f2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d8a12e998df4924b178a74099ba51df839cee64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5d8a12e998df4924b178a74099ba51df839cee64"}]},"branch":"refs/heads/master"},"a6f5b71a50e0180327f0b39e943efef8795f7dce":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-11-06 01:20:34.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/949671/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/949671/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/949671/3"}}},"commit":{"parents":[{"commit":"6212869399d943305c8b1f2e0a20d9cd2cd71b75","subject":"Merge \"proxy_logging: remove unnecessary condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6212869399d943305c8b1f2e0a20d9cd2cd71b75"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 18:59:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-11-06 01:20:34.000000000","tz":0},"subject":"s3token: Ensure check_signature gets called when caching secrets","message":"s3token: Ensure check_signature gets called when caching secrets\n\nOtherwise, we aren\u0027t guaranteed to have the secret on hand in s3api\nto validate aws-chunked transfers.\n\nChange-Id: Ife37041c1e83eea3e108ad80101d84a5ebf70f2b\nSigned-Off-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6f5b71a50e0180327f0b39e943efef8795f7dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a6f5b71a50e0180327f0b39e943efef8795f7dce"}]},"branch":"refs/heads/master","description":"Edit commit message"},"b5b64270beea9d21d8ce706a355cf7ac24f5ebd9":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-11-06 01:20:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/949671/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/949671/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/949671/4"}}},"commit":{"parents":[{"commit":"4c83b5b88075aaacddad7b9a50739acb2017d456","subject":"CI: config Keystone users and s3token with creds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c83b5b88075aaacddad7b9a50739acb2017d456"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 18:59:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-11-06 01:20:41.000000000","tz":0},"subject":"s3token: Ensure check_signature gets called when caching secrets","message":"s3token: Ensure check_signature gets called when caching secrets\n\nOtherwise, we aren\u0027t guaranteed to have the secret on hand in s3api\nto validate aws-chunked transfers.\n\nChange-Id: Ife37041c1e83eea3e108ad80101d84a5ebf70f2b\nSigned-Off-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5b64270beea9d21d8ce706a355cf7ac24f5ebd9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b5b64270beea9d21d8ce706a355cf7ac24f5ebd9"}]},"branch":"refs/heads/master","description":"Rebase"},"13b0597ffb5f7187ccaf14ff627c495d06146d32":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-11-12 21:41:10.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/71/949671/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/71/949671/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/71/949671/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/71/949671/5"}}},"commit":{"parents":[{"commit":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","subject":"s3token: Enable secret caching by default","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 18:59:18.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-11-12 21:41:10.000000000","tz":0},"subject":"s3token: Ensure check_signature gets called when caching secrets","message":"s3token: Ensure check_signature gets called when caching secrets\n\nOtherwise, we aren\u0027t guaranteed to have the secret on hand in s3api\nto validate aws-chunked transfers.\n\nChange-Id: Ife37041c1e83eea3e108ad80101d84a5ebf70f2b\nSigned-Off-By: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13b0597ffb5f7187ccaf14ff627c495d06146d32"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/13b0597ffb5f7187ccaf14ff627c495d06146d32"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
