)]}'
{"id":"openstack%2Fswift~949684","triplet_id":"openstack%2Fswift~master~I6f0cbde23f0a0db75bdb3e598ee862af0bad0dba","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-05-14 05:37:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6f0cbde23f0a0db75bdb3e598ee862af0bad0dba","subject":"Drop support for py36","status":"MERGED","created":"2025-05-13 21:52:55.000000000","updated":"2025-05-14 19:05:29.000000000","submitted":"2025-05-14 05:37:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949684","meta_rev_id":"a875e11e4c83188e980b6909aa837dbecc3849da","_number":949684,"virtual_id_number":949684,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 05:37:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-05-14 04:31:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2025-05-14 02:40:08.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-05-14 04:31:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 22:55:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-13 23:19:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 02:40:08.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"},{"updated":"2025-05-14 04:31:43.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"}],"messages":[{"id":"1ab850bbb2a245b6d8ab4eaf69e3ac3462dacb2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-13 21:52:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b19839bde2948f20641b184ae4c1180c4f25b402","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 22:55:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f78015c1a5ad4c15a097e7fe63890c9c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6cbae5e0a3414904b7d08d1d33ef9163 : SUCCESS in 1h 01m 12s","accounts_in_message":[],"_revision_number":1},{"id":"71be27e9b9607af1f39511665d535244afcc2f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 23:19:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4042d6f69b484476aff8dd5657d13f23\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/370ae221eb994645bf1d477d2c07e2c1 : SUCCESS in 2m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfe3e2feb9094fc487c59e6c88bec82c : SUCCESS in 7m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/e15b8fc898334efbba4bbab872b720b0 : SUCCESS in 48m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1edadc3a78434dfabdb143e4b14be849 : SUCCESS in 50m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8b156733b9d145b989c6e0cb92525474 : SUCCESS in 51m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d486cbf409bd4062a8cdf3ebef437d4f : SUCCESS in 52m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a49ce0e6d5f460ca57a147baa1a5924 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/153a069e4a514091a4a8bca48aef5426 : SUCCESS in 20m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fbe26cf1a0f4077933ed8cfc4112657 : SUCCESS in 21m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c571a3b367bb4fcba89d69cd4f10e3d5 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51c087a06d0d4d6fac88d7a0c4307c25 : SUCCESS in 19m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f17d364e727343deabb3027dffe55c07 : SUCCESS in 18m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7f11f1f3690d47a7b874526c93853cbe : SUCCESS in 27m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ce58c454e0804ff29e9450ecc43e1399 : SUCCESS in 27m 22s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6e9cf20a49ed4586bf4ddeb38a59cca8 : SUCCESS in 13m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c3d3889cb27745f2aa7949536b24e73c : SUCCESS in 8m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6caa2b18d03e4104b64adee79bf52ece : SUCCESS in 7m 59s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd799f5f876643b68b9c09fce8f9113c : SUCCESS in 5m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8eeebe876dac44cfbb40273fc0d93532 : SUCCESS in 18m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cbea626c142944258c541036c5ad1d53 : SUCCESS in 7m 09s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/94d2d344ada54c62832b030d667c70ee : SUCCESS in 57m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ecbd5359162744c28dd1d5c097bad4ce : SUCCESS in 18m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9cd817f1ba394360af16402961fa4410 : SUCCESS in 49m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/943ddefec08b4cd2aa50a7ede99dffeb : SUCCESS in 19m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8afda9d94cc45819473cbffdb8175fc : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/74868fa73f0a46528fc5a228b0be8b8d : SUCCESS in 11m 10s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/534d5f24e0f74d86b7838c5a215ff3a3 : SUCCESS in 58m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/927f73338d134093bb3a02d7a44dea99 : SUCCESS in 51m 18s","accounts_in_message":[],"_revision_number":1},{"id":"a42089d9beb82a30deacb9f08050915867d7cdf8","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-05-14 02:40:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"746d2fc9f38bfa81bf3c8ae01330d05c2715a3a9","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-05-14 04:31:43.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d5ce614c6aea59a4dd51a755050c768e278daf0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 04:32:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e4d99aac735b651244297de70ed3a28090ca73fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 05:37:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b046066ac0f4c908302ab856567f8d3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/74f8c6458cbf4dc1bba5b7a24a3b72fb : SUCCESS in 2m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2ca7903dd84c019a5f0739748016f8 : SUCCESS in 10m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/0d097903034f455595affeb8811e7304 : SUCCESS in 56m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/733e8b9dee3a465a8a8bd0be5319f3c4 : SUCCESS in 52m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4fab848aa64c48c693e7ea2ca34b3833 : SUCCESS in 53m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2a101ad04384d7dbaa22629ddf560ab : SUCCESS in 1h 03m 47s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/806ed5748ab746959936ba72f0650e97 : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8ab17f4d27b24bc49085d1b7bd0d0adc : SUCCESS in 15m 43s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/78ec285f30e243f8a38a86b0f057ddcd : SUCCESS in 15m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ca14894b50c455c81e7f348d5155fa7 : SUCCESS in 9m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/332cb8d42ca54229bf2f3ea00307ac65 : SUCCESS in 12m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/85192eed98704593a0a79772f506b5d7 : SUCCESS in 7m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ec2d6cb8676142e887be3ed656d5e55a : SUCCESS in 7m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3af4d45b7220482f8c90291a93023177 : SUCCESS in 5m 39s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/daee1462021346d98686607edf7f3a7a : SUCCESS in 7m 20s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/691e710f6b1d467b99129d28be5016cf : SUCCESS in 56m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e2ffdc95751f400ca5fcd1e53bdad021 : SUCCESS in 22m 04s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/176a95a77fff4344adec570168e3adca : SUCCESS in 43m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8fb7703ae025471f88c28f3481f1ded5 : SUCCESS in 9m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2e9d9fff9ad42c3860c22b5e37c5050 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/992f91893d6f43b28d0ea438e799bd95 : SUCCESS in 20m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2fed9798f8424c32a5978401ec6616ae : SUCCESS in 35m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dbba8c3abfdb4521a0742d82e4f0fd9b : SUCCESS in 45m 52s","accounts_in_message":[],"_revision_number":1},{"id":"e1facdada5cebae3d3b439e0ecb30f6d87ce3a10","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 05:37:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"113d4967e62673bc1ac9520102060ff91db37403","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 05:38:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c606b2779c0247e2865f97ea5f96d16e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/620c4c493baf45628bdde275d3953e95 : SUCCESS in 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/551e825efb4947beadc9117d7d43a73c : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a875e11e4c83188e980b6909aa837dbecc3849da","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-14 19:05:29.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0714557aa8ea312dd38ff8df0eff925ca3d829e8","revisions":{"0714557aa8ea312dd38ff8df0eff925ca3d829e8":{"kind":"REWORK","_number":1,"created":"2025-05-13 21:52:55.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/84/949684/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/84/949684/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/949684/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/949684/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/84/949684/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/84/949684/1"}}},"commit":{"parents":[{"commit":"6212869399d943305c8b1f2e0a20d9cd2cd71b75","subject":"Merge \"proxy_logging: remove unnecessary condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6212869399d943305c8b1f2e0a20d9cd2cd71b75"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 21:52:44.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-13 21:52:44.000000000","tz":-420},"subject":"Drop support for py36","message":"Drop support for py36\n\nChange-Id: I6f0cbde23f0a0db75bdb3e598ee862af0bad0dba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0714557aa8ea312dd38ff8df0eff925ca3d829e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0714557aa8ea312dd38ff8df0eff925ca3d829e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
