)]}'
{"id":"openstack%2Fswift~949689","triplet_id":"openstack%2Fswift~master~I92f8e3fd5b53086093c17846f5bdb8382a39f977","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I92f8e3fd5b53086093c17846f5bdb8382a39f977","subject":"metrics-injection","status":"NEW","created":"2025-05-13 22:27:15.000000000","updated":"2025-05-13 23:22:38.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"42dac6e7395ddd9f24b4f1de0ccf653a3a7e767e","_number":949689,"virtual_id_number":949689,"owner":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-13 23:20:26.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 23:20:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b242645e1ec59cc3202033738018a15c022dee1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-05-13 22:27:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3a0396a93397e17b6dd85f378c699edecbb1b3b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 23:20:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5e39cf07a804f478ce5ef7f38247ab7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd7ecde3eaeb4d1ab3e30ec20ad0e83e : SUCCESS in 10m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e9e30a9d470e41cc9b8c2d478b15922b : SUCCESS in 29m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c3cd1897854046aa8af57b5c9847657a : SUCCESS in 13m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9a7b3a75f06a42f38de4545a20ac7e6f : SUCCESS in 18m 54s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9dabf7467fdc41e8a1bf911b3e3529cf : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aff62c68155f4bd7ba0579584018470c : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py36 https://zuul.opendev.org/t/openstack/build/0c7e48ed37434e1b84f9a4557cc18d6c : SUCCESS in 9m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d7e5902a85f74afc9e3505a11a24efba : SUCCESS in 8m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b246217e99894ff1b01f48468b84cdb7 : SUCCESS in 25m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e2f7dedd147148a0a35f984bf409be76 : SUCCESS in 25m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/83ba6a0bbeb74a1180a7b6dd3bca2b3d : SUCCESS in 14m 13s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/54208faa11e7435a82193e55873a4e33 : SUCCESS in 17m 42s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/66e54a8b76cf4a158ef4b32ee10f3fa4 : SUCCESS in 16m 13s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b844ad39c1474739852d782e7c5a62e8 : SUCCESS in 47m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64044e1870d54977be92882ffccc3322 : SUCCESS in 19m 07s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8967723137f446f29d72397c5dc0418c : SUCCESS in 16m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a6030e40cb040d7b531e116b68ed144 : SUCCESS in 3m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7b84d4cc643c4f32a85d1ce7c290f55d : SUCCESS in 19m 40s","accounts_in_message":[],"_revision_number":1},{"id":"42dac6e7395ddd9f24b4f1de0ccf653a3a7e767e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 23:22:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30f23629e47640b0bdf3bd512c38ffa9\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbde1a9a199a4189b003af7622866ffb : SUCCESS in 53m 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ef2c8160854a4f8e10eac588d7c0eafc48fa50e","revisions":{"9ef2c8160854a4f8e10eac588d7c0eafc48fa50e":{"kind":"REWORK","_number":1,"created":"2025-05-13 22:27:15.000000000","uploader":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"ref":"refs/changes/89/949689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/89/949689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/949689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/949689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/89/949689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/89/949689/1"}}},"commit":{"parents":[{"commit":"6212869399d943305c8b1f2e0a20d9cd2cd71b75","subject":"Merge \"proxy_logging: remove unnecessary condition\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6212869399d943305c8b1f2e0a20d9cd2cd71b75"}]}],"author":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-05-13 22:27:09.000000000","tz":-300},"committer":{"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","date":"2025-05-13 22:27:09.000000000","tz":-300},"subject":"metrics-injection","message":"metrics-injection\n\nChange-Id: I92f8e3fd5b53086093c17846f5bdb8382a39f977\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ef2c8160854a4f8e10eac588d7c0eafc48fa50e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ef2c8160854a4f8e10eac588d7c0eafc48fa50e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
