)]}'
{"id":"openstack%2Fswift~949716","triplet_id":"openstack%2Fswift~master~I811642fccd916bd9ef71846a8108d50a462740f0","project":"openstack/swift","branch":"master","attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-22 15:47:18.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I811642fccd916bd9ef71846a8108d50a462740f0","subject":"Add config option for whether to skip s3_acl-requiring tests","status":"MERGED","created":"2025-05-14 09:54:04.000000000","updated":"2025-08-22 15:47:18.000000000","submitted":"2025-05-28 18:05:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"949716","meta_rev_id":"41f0ca08c5ab6d4c6e9bb21db71e7fd52a54c29f","_number":949716,"virtual_id_number":949716,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-28 18:05:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-14 15:57:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-28 16:39:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 10:59:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 15:57:46.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-08-22 15:47:18.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"}],"messages":[{"id":"5e9a7f3ccab81bb993d045f78859705b1d306d36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 09:54:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb6934780839ae90d6c772612f07da68a5087ebb","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 09:56:40.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09fbfab7ab9770911cf284a344a545c2d11f60c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:59:04.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8c9d8cdf661941db84825b8691a9d3c3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5351f2dd9a5444ea829f0ab2d8b591e0 : SUCCESS in 10m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/34df0bdcbcfa420eb08b59581755a134 : SUCCESS in 52m 25s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee30c87f16414a1095ea4c40c0081487 : SUCCESS in 11m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2611324a11ec45fca7963a68fcd2f7a4 : SUCCESS in 17m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0e1eb8580a204993ac8423d30df2fd99 : SUCCESS in 23m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ddf5e08ff5b84487ad48277e747163c2 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb5693d2ca804275a114e88f3657aaf6 : SUCCESS in 10m 01s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2bd65315b47e4fd79601a3ab828e2cac : SUCCESS in 21m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12fbd49182c043598182087458bc42c6 : SUCCESS in 27m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/46bd83f4eeca4233be98fbd2544d5090 : SUCCESS in 20m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/678589d584a941c0b361c9b537555fc1 : SUCCESS in 16m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/58049ab1f2244d1689e70f3a0f70db75 : SUCCESS in 17m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e5f10a1ac6dd430f914c439ae6846f7e : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7655bbadadd04d04bd34fcf8f0edc0cd : SUCCESS in 7m 14s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d4dcaa617cf74f19bd2f6a13181e329b : SUCCESS in 9m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f41f98ea4dff4838b8cee668527a4ab9 : FAILURE in 57m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/69922ec445ae4e12b7d0038242251026 : SUCCESS in 45m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/246e8f82de1b4178b37f61b982bbfc34 : SUCCESS in 40m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/522674a6fc304b20b6f89fe34f6cd400 : SUCCESS in 19m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/831d20049b784a4c9c7573488272d299 : SUCCESS in 5m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8ceb277dcd4246a78916cbeea05a60c1 : SUCCESS in 20m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d7affa44accad35f05cd66bd80263a5190ddc4ec","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 10:59:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94858313cec246a68fa0b0d93d3c1db1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c9d59268b42a4af4abb05d602b1b5c76 : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":1},{"id":"38fd82703bfcf5f6c018c736135ecf1cb35a228d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 13:06:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bcac64cdfa7fea50ce75c7f9ba44a8018481ddc","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-14 15:57:46.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2bf9af801cdd9ff19553b4f92cedf54bf86c9eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 16:14:22.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8dd52855f17d4211a6538e807b4ba67d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97a2c24c56bd4ea7a2da20ef6e5e5862 : SUCCESS in 11m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/91575f351d424e9ebb36f74e00dc3267 : TIMED_OUT in 3h 02m 34s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b69a6bb5b20b4f1da4ec1140efae99d6 : SUCCESS in 20m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/caa6b905d8e149ddbbf96013e99b7439 : SUCCESS in 20m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/aa71c29fe2dc4f5c845c9b4caaba8653 : SUCCESS in 20m 24s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a283a37b2ad6488bbb841761362ad771 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b07b307e704e4a218547e13d8b4d6562 : SUCCESS in 9m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f008fbd17f6a47a489543f2e51fb940d : SUCCESS in 27m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a666076372734210adb34ecb09bedc33 : SUCCESS in 10m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/04fc25176c72455e9d77346ae4441a7e : SUCCESS in 10m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2b8ff38ee17e43fb93c24f73428cda67 : SUCCESS in 14m 05s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/81368c0172cb4958badf0f29f7a9e6c4 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/54bba024e5dd45aea962acde4fad532a : SUCCESS in 9m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/769fd2a5f8e74274b38f020730a1072f : SUCCESS in 4m 29s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/0706adb4b048423586394352a71bec8a : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f98a9d6849534d97b03d522d949cb515 : SUCCESS in 56m 01s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cc6e0b0ce5d4d5b889fb95017cd51e1 : SUCCESS in 24m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/09fb3d7a9a22474d9b2154f8c4774723 : SUCCESS in 49m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7dbdbb1f8f364a4db0a62dfd785283e0 : SUCCESS in 19m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d142dabecdef4aad9e405e612b4c013e : SUCCESS in 5m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c4e8698690034f10abefd63aae93fd2b : SUCCESS in 21m 27s","accounts_in_message":[],"_revision_number":1},{"id":"739177e5e69d07af989cc4c92799efda9d3c89ad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 17:04:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd76f08ab71a4786bb2a8195fbe41458\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1c61072a0de241178574e02a28cc11d6 : SUCCESS in 1h 15m 58s","accounts_in_message":[],"_revision_number":1},{"id":"fb15f6531851d51b7be30dc24caf337c52b60e94","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-15 03:54:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1ed797a5c34e4628c087ad83b2ad0a12e8c6af8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:40:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1fd903ff5814d9bbbba98571361fa19\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af984652940948b98d626014d394ee95 : SUCCESS in 9m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/af58684c67e14d8881ca99caa21d4296 : SUCCESS in 45m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b50c6eeeb2c4403942ea25d32ded2f6 : SUCCESS in 14m 42s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f33d9f72bb94a259dd49333be3da5c0 : SUCCESS in 19m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2bf23e6c1330439a9167e1675ae51c11 : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f97be190582048fb830aeb7b8edc05cb : FAILURE in 4m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/999353ce50694763a5316b3dda1c5e12 : SUCCESS in 17m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0e97a6fd7fa44562b49397767e449709 : SUCCESS in 8m 14s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d52992c35071447ba4fa9afc03b3e6cf : SUCCESS in 21m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/966ab9085779465d98a77c5f58339f2e : SUCCESS in 10m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e69de83a00164181bfcec8bca8143efc : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/30e7ea49970541b69afe0ec135a689b3 : SUCCESS in 7m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2abb0a14f35b410ea3fb322200e5697e : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3d340d9edd32409180052b5ae257ceaa : SUCCESS in 3m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e53b76d3bf0c4ff6a4bf26e884beaed4 : SUCCESS in 3m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbb2891e83714677bf65398fa1dc9860 : SUCCESS in 33m 17s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cba220eaa17e435a92edae68aad82106 : SUCCESS in 18m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/42d0ba0837bb4949a99ed5cd3f176f7a : SUCCESS in 22m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a4ffeb7282564c29825d5a45849238da : SUCCESS in 17m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527a0913d3254294b174c956731af4c1 : SUCCESS in 3m 56s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2954f8c1611a41418e600300762ca8f7 : SUCCESS in 20m 14s","accounts_in_message":[],"_revision_number":1},{"id":"6b9323ec6ab5d433d6cfb22251afcfc66ef4734e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 04:50:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0060771588884f1da0eaff1a757f1ea3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/411613804cbd48edb90b08d7eab10577 : SUCCESS in 54m 38s","accounts_in_message":[],"_revision_number":1},{"id":"ff6b074175adbe8050d63cc54172b39609d2403e","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-28 16:39:04.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f9fc1bcf5217225580b2c89f48ead782f7b3327","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 16:39:35.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"37e239c588bddc8f1316e6d0558bfea7802d69ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 18:05:43.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7da4e3e7ea548e6be023cfe5bca0ae8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7dafc6136fe545d9b7febfdf77fb1543 : SUCCESS in 9m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/23a9e963f1ef4daf9aa99d14ff790546 : SUCCESS in 53m 36s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/bad8fd187a994daaacdb46377a6bac67 : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1abbc5eb8c3246dc8d39fc937fefd01f : SUCCESS in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/770da231eeba48ebb575a89f30eae3bf : SUCCESS in 22m 58s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/10104c0e28c3478eb0db7b0b780fcd87 : SUCCESS in 19m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/52b84a3c61e442818c378de2196729a0 : SUCCESS in 16m 35s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6e9d43248250463ab2c0dc248d023721 : SUCCESS in 16m 00s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/819b8333d55148cd9331a2fdc788ea18 : SUCCESS in 14m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e4c53a2ee11345d3a3c46957550f174f : SUCCESS in 5m 47s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/963c290462014550b4be318614d1674f : SUCCESS in 8m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3142c60ff4d2453a94b69115ad4ca2e4 : SUCCESS in 56m 08s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d69e9e04b8e74f72862b1fa045057375 : SUCCESS in 47m 06s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ae1d9d2762f34d95ad5a23a06c6fa2d8 : SUCCESS in 37m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3b902637e7684f348b3d1240fd8c58b2 : SUCCESS in 17m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ddf666a0fb7c4d44ad540115edd68401 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4ca8341b8bc423aa15654d7597b8bdb : SUCCESS in 17m 14s","accounts_in_message":[],"_revision_number":1},{"id":"7aca092128a35b2b18be0bf144aa1ebaf4b336d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 18:05:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"40344bc6c5103cef6671c25a29d33d711f6c15a0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 18:07:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48ea83754b244604b19619509a77fcc4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58a0ef61b52241b4ab2e8024d85f2725 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/6d2c655f364440bd851b5aeb8a699dd7 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41f0ca08c5ab6d4c6e9bb21db71e7fd52a54c29f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-22 15:47:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a3b7373bcf211ff1d7a29b8d7cd8220f6430e80","revisions":{"7a3b7373bcf211ff1d7a29b8d7cd8220f6430e80":{"kind":"REWORK","_number":1,"created":"2025-05-14 09:54:04.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/16/949716/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/16/949716/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/949716/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/949716/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/16/949716/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/16/949716/1"}}},"commit":{"parents":[{"commit":"c12aa812c9f434d4584e65d48c8f77b6c8c37862","subject":"Merge \"Drop support for py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12aa812c9f434d4584e65d48c8f77b6c8c37862"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 09:53:18.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 09:53:18.000000000","tz":60},"subject":"Add config option for whether to skip s3_acl-requiring tests","message":"Add config option for whether to skip s3_acl-requiring tests\n\nCo-Authored-By: Tim Burke \u003ctim.burke@gmail.com\u003e\nChange-Id: I811642fccd916bd9ef71846a8108d50a462740f0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a3b7373bcf211ff1d7a29b8d7cd8220f6430e80"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7a3b7373bcf211ff1d7a29b8d7cd8220f6430e80"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
