)]}'
{"id":"openstack%2Fswift~949766","triplet_id":"openstack%2Fswift~feature%2Fmpu~If7ba922fb6f432ae970d12cb484fabeb2bc9015b","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-14 20:45:29.000000000","reason":"Change was submitted"},"6476":{"account":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"last_update":"2025-05-14 20:45:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If7ba922fb6f432ae970d12cb484fabeb2bc9015b","subject":"merge master to feature/mpu","status":"MERGED","created":"2025-05-14 14:50:39.000000000","updated":"2025-05-14 20:46:49.000000000","submitted":"2025-05-14 20:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949766-merge-master","meta_rev_id":"24a0c00b394c71a2149b6c6c963a0cadfc0297a6","_number":949766,"virtual_id_number":949766,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-14 20:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-14 17:53:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-14 17:53:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"}],"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-14 15:14:38.000000000","updated_by":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"reviewer":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"state":"CC"},{"updated":"2025-05-14 15:58:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-14 17:53:02.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"2a47a940bb025d2e8aa9362934ce3d83a364789a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 14:50:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3971898730c918603b59cdf600fe0f6db119d5b2","author":{"_account_id":6476,"name":"Thomas Goirand","email":"thomas@goirand.fr","username":"thomas-goirand"},"date":"2025-05-14 15:14:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"08e39bf3cb71c29085c348126667b9cc41bbf851","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 15:24:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bec6cae3e53c0dac5fb592faf8150b1eb844ad51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 15:58:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a723ae96af2d447f8745d41e8b3d22cf\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/87cb46b1e9c447b4b287bfee8e3f4027 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af8ec6cc59734e9c9ed39e577d38ff41 : SUCCESS in 10m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb9bb675042847f78b8eeae12875b0da : SUCCESS in 3m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32e73b839668473390b2c5949f5618b9 : FAILURE in 18m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f248a5f7f3ec4e65bd90e3cc62d9146d : FAILURE in 19m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fc27fc425864b14bdfa8870e91374ae : FAILURE in 17m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/15696be793be4e368e27b29c6dfed6e1 : SUCCESS in 4m 25s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/65c02ddb777d433bb63ccc7a6ec2ec89 : SUCCESS in 19m 53s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/921a141765474d00bf6f2df460e468a6 : SUCCESS in 17m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/399e83f5dd49400ea25c7bfe72d5b204 : SUCCESS in 10m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/43d864ab14f84f1481cbd547a3f9e3f1 : SUCCESS in 22m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/acbad0a3a2b04e2bb00dafaa68ce8818 : FAILURE in 12m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2c18bf01c0c248af8e96b56eb4f2d262 : FAILURE in 17m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4f551f5366d445d0a130144a18e60bf9 : FAILURE in 14m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/7e3ab41f5770410581c55d4286da637f : SUCCESS in 6m 47s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/80ef6bcfecd14e36b66397b8cd95fd2b : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/4822c372ec9e4c9fb5c0999ce0530a51 : SUCCESS in 6m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1b6c1733c3bd49ada2cc7c006bfd26e2 : SUCCESS in 1h 05m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90d9bba4e9074a7f89ebd19556289056 : SUCCESS in 20m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79ba13e956e4fcabe0e779922505dbd : SUCCESS in 4m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/740ac49dbf8344248c605e3ae00f0d59 : SUCCESS in 21m 25s","accounts_in_message":[],"_revision_number":1},{"id":"da2338d49025003d6a77028da9705eac7e141154","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-14 16:18:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c57c11d1301d8c14d3d354f347ae7e8f853048d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 17:25:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f730c3af3a64fd494ee72aa17133e7a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/328948dc6d6c424db72585e01abed56e : SUCCESS in 2m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41901ba7c607425f91db5f468d499011 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1a55d75c706428cb3f3024a9f85b770 : SUCCESS in 6m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c94a31850cf4b799cfd0d07e7def394 : SUCCESS in 20m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/629db45e8da345b99189704aebd38d21 : SUCCESS in 21m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/906eff9c5df948328dfc6a4bcbfeacb7 : SUCCESS in 20m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a5797bb325a4d8ba88df93ca00a77c7 : SUCCESS in 6m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/add470809f114c4995f44a5f069211f0 : SUCCESS in 20m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f5167e850a7f455cb71a65fdf55dcec3 : SUCCESS in 23m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/683fae06a87847b880320520f341d323 : SUCCESS in 28m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2c01f8812a36417a9d5d9043329861f7 : SUCCESS in 31m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5f054fef7456480d93e1008d3181f285 : SUCCESS in 14m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/142fd913fcb147d597e4088c6fbb30d5 : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f6de4967734a47aabcac97bebda1d44b : SUCCESS in 15m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c49a9eaf06b6494e954bf350668b6a5b : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0185473bd85f421d969dc43ff4e49e83 : SUCCESS in 14m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ac101c299b714eed80cc3e56ca8afa31 : SUCCESS in 3m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d7035de7ebfc4b4585a2476764b39218 : SUCCESS in 47m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a67a8b2ee6a7474b8810c5805975c694 : SUCCESS in 7m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42fb785a12a6463b9d05a3a80b3c5fef : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/70a185640254427fbcffd03654287193 : SUCCESS in 10m 56s","accounts_in_message":[],"_revision_number":2},{"id":"0944bc7b66fbb7494eec054c991756f3a86633b6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-14 17:53:02.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c5726f1ed4d4daf96480e93ef0c8812bf0aa619","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 17:53:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"89a3e6ff440f1e56450b96c78637c6b2fd55cdad","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 19:28:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/876e55cf137e4b98aa06b12cdf0eb024\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e54a7af2a92a4d4693af2e4b61bb890b : SUCCESS in 1h 00m 33s","accounts_in_message":[],"_revision_number":2},{"id":"75e2fc84eea4e6ab8138c057c9c9dc96503ba6a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 20:45:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eee4665ec99848acbe87c87c508faa35\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d6aa8967543b4893af982adb78ae5d58 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f12f549fcec41228bc5c9338abf1b28 : SUCCESS in 11m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97d4292152a34dd7a6c1d2c98642d7ac : SUCCESS in 3m 08s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ac39cdfb0cb247f5bff63cf2824ce93b : SUCCESS in 5m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/948fb1a2cd2a40c0aff3f1980ced7db6 : SUCCESS in 8m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a85ce9a1446d412dad81b72ea23aaed2 : SUCCESS in 20m 08s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/42e26029492f4d46a5e50d1ca0de286c : SUCCESS in 19m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/722c83ed4b8a4a169f2a90e3a372f75a : SUCCESS in 17m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/bbae319d4c5447188971f1c5d26d5c95 : SUCCESS in 16m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9456b08d819f46339720f2206eb5ace2 : SUCCESS in 17m 20s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/07e26982d2c745bd9c4a85b942e50275 : SUCCESS in 5m 55s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/43699fb532184c85885f249e11ee35fc : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/28c831b066124d218ac897298331262b : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/befd71f1e8714667a538d7c58bca3d60 : SUCCESS in 14m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cca030b23f90404b8dcd5e9da63e2e11 : SUCCESS in 3m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/512f729745554dc887025c5216d946cd : SUCCESS in 21m 53s","accounts_in_message":[],"_revision_number":2},{"id":"98971bdd39f8afae261b27702788d1ccd4f05d06","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 20:45:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"24a0c00b394c71a2149b6c6c963a0cadfc0297a6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-14 20:46:49.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e02ba82837e4749a6ebc58322603820\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6dd85c4d5e664913bf8be86a5dd4438a : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f4d89a8c888e48ba8c8444d191faaa04 : SUCCESS in 50s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2d2ebd77d41a498daf430e9291a74f33 : SUCCESS in 34s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9ea6276071ae9802fc9e5f83ca0bf3e84588d0a4","revisions":{"49ce32beea06767ac4eed20aa9903ce17bef72e8":{"kind":"REWORK","_number":1,"created":"2025-05-14 14:50:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/949766/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/949766/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/949766/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/949766/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/949766/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/949766/1"}}},"commit":{"parents":[{"commit":"2fd857a8e1ef297ed52efca5ad4b987f80495229","subject":"merge master to feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fd857a8e1ef297ed52efca5ad4b987f80495229"}]},{"commit":"c12aa812c9f434d4584e65d48c8f77b6c8c37862","subject":"Merge \"Drop support for py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12aa812c9f434d4584e65d48c8f77b6c8c37862"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 13:47:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 14:48:12.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: If7ba922fb6f432ae970d12cb484fabeb2bc9015b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ce32beea06767ac4eed20aa9903ce17bef72e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ce32beea06767ac4eed20aa9903ce17bef72e8"}]},"branch":"refs/heads/feature/mpu"},"9ea6276071ae9802fc9e5f83ca0bf3e84588d0a4":{"kind":"REWORK","_number":2,"created":"2025-05-14 16:18:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/66/949766/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/66/949766/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/949766/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/949766/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/66/949766/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/66/949766/2"}}},"commit":{"parents":[{"commit":"2fd857a8e1ef297ed52efca5ad4b987f80495229","subject":"merge master to feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2fd857a8e1ef297ed52efca5ad4b987f80495229"}]},{"commit":"c12aa812c9f434d4584e65d48c8f77b6c8c37862","subject":"Merge \"Drop support for py36\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c12aa812c9f434d4584e65d48c8f77b6c8c37862"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 13:47:45.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-14 16:18:15.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: If7ba922fb6f432ae970d12cb484fabeb2bc9015b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ea6276071ae9802fc9e5f83ca0bf3e84588d0a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ea6276071ae9802fc9e5f83ca0bf3e84588d0a4"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
