)]}'
{"id":"openstack%2Fswift~949951","triplet_id":"openstack%2Fswift~master~I3dcca36744bbdbce7e5e22e50b3a3fa263071717","project":"openstack/swift","branch":"master","hashtags":[],"change_id":"I3dcca36744bbdbce7e5e22e50b3a3fa263071717","subject":"CI: Move lower-constraints job to py37","status":"MERGED","created":"2025-05-15 21:46:16.000000000","updated":"2025-05-16 04:51:08.000000000","submitted":"2025-05-16 04:49:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"949951","meta_rev_id":"9307286a72c187f2c88757a614ac1febbb5c90c9","_number":949951,"virtual_id_number":949951,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-16 04:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-16 03:02:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-16 03:02:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-15 22:31:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 01:34:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8887030c2919149b2eeb0dbff6ab688f9e244fcd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-15 21:46:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74e8f00617594a0900f6bd49798f3917987a22f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-15 22:31:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28a3114a402640f9ab6d9ffab1b8021b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/924fba49a22e4833b726c6b044852232 : SUCCESS in 43m 25s","accounts_in_message":[],"_revision_number":1},{"id":"0e1332a06ee1a3679a29a631ad088eb8499437b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-16 00:23:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2e9f5f68ef94e6628a69e695c66cc5f1b05cef7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 01:05:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0106edff1c7642dc9d65e003b08c0c06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6fbab071386a427e8f9fb0ec34a62d32 : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":2},{"id":"fc3b0729a01c128a38c8f5120219a8f915882b72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 01:34:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36d6bcd1a5fc401ea088a14cffe22100\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5ab4e658fdc54b31b11161e8eb5f4e12 : SUCCESS in 2m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4910eec0f1f84814b7da8914d5918157 : SUCCESS in 10m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a7f39244a53a4f0b8a8b8583a00b302d : SUCCESS in 2m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5661cccf23a5428d96b8e5b3b51d4eb2 : SUCCESS in 6m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/4b4717fc07bb4364bb75e830ddabd963 : SUCCESS in 1h 01m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b4443ad079de40029c4f3242a550f36c : SUCCESS in 44m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb86e0fb43ca49849d5027c56e81c590 : SUCCESS in 42m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47ff4aa2775d4eeda7a6e0cd9296e5f7 : SUCCESS in 56m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4c2476bde14e4ea0b621bd582fcda72d : SUCCESS in 19m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dbe4d50594442eba36db39a2ed0be93 : SUCCESS in 9m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/033331ac4d6d4b0089c65b777a77c1ec : SUCCESS in 18m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/484d7621333c425e8a9847f8dc98ed69 : SUCCESS in 4m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2710cc9db1ba417d9e88e9c8a9fb5698 : SUCCESS in 13m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0a073604000d4d929c11b6cdfa1e40d2 : SUCCESS in 23m 04s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7aaa1475f7a448dbb0d7980a1cc22f9a : SUCCESS in 27m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/afc5a5c856df43179ad6632c3f68a455 : SUCCESS in 10m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/77d8de0016ac4d0aaa3eb825de73cc35 : SUCCESS in 13m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c033eb908ba04aa285a07e2f99ff6ef7 : SUCCESS in 15m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8a9be33dd7d049d391fda7cff75eb788 : SUCCESS in 15m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/47bd947845cf446692623c4a6764a351 : SUCCESS in 5m 28s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/59827b396cb24439aee9e81a7163bdf5 : SUCCESS in 18m 48s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/02cba3226a8c4eb5a7e85d25910b008a : SUCCESS in 6m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6a1ae19d05c49f8834772754397c19d : SUCCESS in 54m 33s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3d4f094d90cb436b8421227e7b025098 : SUCCESS in 30m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a3f560d8601405698647c02689c43fc : SUCCESS in 51m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8d77a117d63840918fe813f657bd61e7 : SUCCESS in 13m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2aa86c8e9f94402833a8e1e433807d6 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47fecae484ca4a9b879835dc0e65b021 : SUCCESS in 26m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e31b1c7dddfd46f7b3073384f587a937 : SUCCESS in 35m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/48467ea461c6433b922b712a0349253f : SUCCESS in 45m 55s","accounts_in_message":[],"_revision_number":2},{"id":"a43635c2dd1a5e4be1af4dc0bce8a9155346d313","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-16 03:02:06.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7da69e07b971049dbe315d8a9c79367b004e5629","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 03:02:41.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48691c362eafe528f6ac2d7c228aca043d0eec9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 04:49:33.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/72dedc614b6c4bd5af6519aafdc143f9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/cf7ce67f693d4a719af5e438a72a456d : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e21ea5148d504ef1b2b44926cce63844 : SUCCESS in 10m 45s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ca4d6a8f98094a76801a2e1164c20d46 : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e25007ff9798418b9ad0d12bec8bca4b : SUCCESS in 5m 50s\n- grenade https://zuul.opendev.org/t/openstack/build/650bcc8166d34a209790b65467e19d8a : SUCCESS in 1h 01m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2a2ec1ef5aab44eb9f5939c44912298e : SUCCESS in 58m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/53b7723721e44a45ae1395e20397020c : SUCCESS in 52m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2470e0f07c584a59af3f16000c1c1a89 : SUCCESS in 59m 29s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/d22ab80072e64e8bb175ec3cec5b7811 : SUCCESS in 5m 33s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7865aa378669492087243dbb38e6f95e : SUCCESS in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/07065bc047bd4027bf51413a14679ec6 : SUCCESS in 21m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c11ac4b173194abc8755db2f5f5b470d : SUCCESS in 13m 48s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d56524c252564ff98dc3d0b9900c5ebb : SUCCESS in 9m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e68c2c0a1e8940ac8aa7295b80514666 : SUCCESS in 13m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e2c624f6e7b44cfea902f16274408eab : SUCCESS in 9m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e30ee642203d447e9e24288bc0df8349 : SUCCESS in 6m 18s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/370552cc31144f27a932ecfa5ad14eba : SUCCESS in 4m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e4697d4f3c5480b9c5d6a37f3cbda0e : SUCCESS in 48m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/237a73c80470455a9138e11b22b5963f : SUCCESS in 34m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34d627eaa9684bfcb638112c308e77f5 : SUCCESS in 39m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7b381f81050547c0b7cc5baa988970a0 : SUCCESS in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49b9b82d4f6d46e69f55c86130b1994c : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5c882ceda3524ecebf07f7b05d02e4d4 : SUCCESS in 23m 13s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15fab42fbeb14e168d8789137924f29c : SUCCESS in 30m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7fef00d0646a41dab804057e9e3b3f16 : SUCCESS in 29m 04s","accounts_in_message":[],"_revision_number":2},{"id":"91c8f9fa4a766e37679bdc92549921cf08d1631e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 04:49:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9307286a72c187f2c88757a614ac1febbb5c90c9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 04:51:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54022f2a92e545289ccc39824d54b586\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/4a62ff867a8140eb91adacdd1f32df1f : SUCCESS in 44s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee1d76c5677d402e8931c7cb978a2d08 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/552dcb50046c49e0877e76573f1408ac : SUCCESS in 37s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/f0ef815dcc2f44318bf95c079e055507 : SUCCESS in 30s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a2725ea99fecf11506293f80fdf02f72254cbcd6","revisions":{"e9b6d2314f4cc7e57eff5616e1a0e335cd36cec7":{"kind":"REWORK","_number":1,"created":"2025-05-15 21:46:16.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/949951/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/949951/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/949951/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/949951/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/949951/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/949951/1"}}},"commit":{"parents":[{"commit":"4e2d08041ade8005d2552783f0b34161f797a718","subject":"Merge \"tests: Use subTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e2d08041ade8005d2552783f0b34161f797a718"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-15 21:46:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-15 21:46:09.000000000","tz":-420},"subject":"CI: Move lower-constraints job to py37","message":"CI: Move lower-constraints job to py37\n\nThis got missed when we dropped support for py36.\n\nRelated-Change: I6f0cbde23f0a0db75bdb3e598ee862af0bad0dba\nChange-Id: I3dcca36744bbdbce7e5e22e50b3a3fa263071717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b6d2314f4cc7e57eff5616e1a0e335cd36cec7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e9b6d2314f4cc7e57eff5616e1a0e335cd36cec7"}]},"branch":"refs/heads/master"},"a2725ea99fecf11506293f80fdf02f72254cbcd6":{"kind":"REWORK","_number":2,"created":"2025-05-16 00:23:02.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/51/949951/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/51/949951/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/949951/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/949951/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/51/949951/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/51/949951/2"}}},"commit":{"parents":[{"commit":"4e2d08041ade8005d2552783f0b34161f797a718","subject":"Merge \"tests: Use subTest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e2d08041ade8005d2552783f0b34161f797a718"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-15 21:46:09.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-16 00:22:53.000000000","tz":-420},"subject":"CI: Move lower-constraints job to py37","message":"CI: Move lower-constraints job to py37\n\nThis got missed when we dropped support for py36.\n\nRelated-Change: I6f0cbde23f0a0db75bdb3e598ee862af0bad0dba\nChange-Id: I3dcca36744bbdbce7e5e22e50b3a3fa263071717\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2725ea99fecf11506293f80fdf02f72254cbcd6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a2725ea99fecf11506293f80fdf02f72254cbcd6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
