)]}'
{"id":"openstack%2Fswift~950162","triplet_id":"openstack%2Fswift~master~Ia23aff3c59c3106868c4a75c2ae96a81eca1204d","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_object","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-19 15:42:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia23aff3c59c3106868c4a75c2ae96a81eca1204d","subject":"sq: add tests to test_base_storage_server","status":"ABANDONED","created":"2025-05-16 13:14:19.000000000","updated":"2025-05-19 15:42:45.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b7b732b2e4d447bfb12ff885af33e4ef73ee720d","_number":950162,"virtual_id_number":950162,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-16 16:25:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-16 13:54:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-16 16:25:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eade4a4ce906baa8eddcf0561de033b7bebee9cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-16 13:14:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fbbb6a2a9c0699c8a4ed63bf3e340f19cae165c6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 13:54:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dedda533926b47258d23a60803c61de3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/84a7bf7554714cfcad2a010edba2cc9d : SUCCESS in 37m 25s","accounts_in_message":[],"_revision_number":1},{"id":"6cc57f6223376d4b286502a85e05a56c1407a93f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-16 16:25:02.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b244f8c24b446388258b49d4de152b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33fdf1e5c18c493abd436da8ccb0754e : SUCCESS in 8m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/deb34423574d4834bdb45928e86ba50a : TIMED_OUT in 3h 02m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4f48788bd20d46e7bd2083cbd83a6fc1 : SUCCESS in 19m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b42814216014dbaa91b74883ca6e6f2 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f3f7927099f493cb5e9f09d0961debb : SUCCESS in 10m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d06f51a890804bc2bc84a9d6f979e2bf : SUCCESS in 4m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eef7d2cdb11241dc9bc2eebf185e6695 : SUCCESS in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d63b534308094888857059601a93fb30 : SUCCESS in 20m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3f1dd5015e43449bab4c03e20d5d7a2d : FAILURE in 4m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7e659b11dcc74e48bd50e3c01a99f680 : SUCCESS in 43m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d03ae1477be34185b7cff22c537ede14 : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dc40675378864bdb9fce61aff412493e : SUCCESS in 17m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/511f3e919a634d5eb72c79da421acfe9 : SUCCESS in 16m 54s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7dea313c66bd45f79f6aac08a6aeee9f : SUCCESS in 42m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ea89f61df7384ef88febc78d090f4157 : SUCCESS in 45m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3c195eb83f434b4bbe735bcbec25b125 : SUCCESS in 23m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/903a23aa971a4e409e9ef29b439b75de : FAILURE in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7141747b2e104d4a9c510183236344ed : SUCCESS in 18m 52s","accounts_in_message":[],"_revision_number":1},{"id":"b7b732b2e4d447bfb12ff885af33e4ef73ee720d","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-19 15:42:45.000000000","message":"Abandoned\n\nsquashed into parent","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"33fe7cad12f61c954ff9844f2649756376e5b1ec","revisions":{"33fe7cad12f61c954ff9844f2649756376e5b1ec":{"kind":"REWORK","_number":1,"created":"2025-05-16 13:14:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/950162/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/950162/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/950162/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/950162/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/950162/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/950162/1"}}},"commit":{"parents":[{"commit":"69bb2360e1c45c9a84b81c6531b36ca905dff1ee","subject":"common: add http exception handling for base storage server timing stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/69bb2360e1c45c9a84b81c6531b36ca905dff1ee"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-16 13:13:42.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-16 13:13:42.000000000","tz":60},"subject":"sq: add tests to test_base_storage_server","message":"sq: add tests to test_base_storage_server\n\nChange-Id: Ia23aff3c59c3106868c4a75c2ae96a81eca1204d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33fe7cad12f61c954ff9844f2649756376e5b1ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/33fe7cad12f61c954ff9844f2649756376e5b1ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
