)]}'
{"id":"openstack%2Fswift~951042","triplet_id":"openstack%2Fswift~master~Ia6e856ffaf8fd1b4a905e6976ebdc62ed5ddf32f","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_object","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-05-29 10:14:07.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-05-29 11:59:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia6e856ffaf8fd1b4a905e6976ebdc62ed5ddf32f","subject":"Object-server: change labeled timing metrics sample rate for debugging requests","status":"MERGED","created":"2025-05-27 21:18:58.000000000","updated":"2025-05-29 12:00:44.000000000","submitted":"2025-05-29 11:59:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"951042-tmp_labeled-metrics_object","meta_rev_id":"67f9a9f26256ce43e2a4fb6a38ca8e53a2344b21","_number":951042,"virtual_id_number":951042,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-29 11:59:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-05-29 10:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-05-29 10:14:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 22:01:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-28 00:22:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-28 08:48:02.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"72f441552294e45cd646c30a036548a63fa0e972","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-27 21:18:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8de4a53ed80084d54217a0e363117f113bc8a8da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 22:01:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbbaf07897724403b86e1e25a73a66e2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d2730367f5a4e37a24655f00ecd241d : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"24bf98e85bda31dee9de5cd1a29451e24166cb0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 00:22:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d1f63eec077c468dac3bc6e9e039856e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bafdda6ee0ca4a9b8b9c213d45acff68 : SUCCESS in 8m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/dca38ad33df543be9e9074707e10ab86 : SUCCESS in 53m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c1e605dbcbaf454c992773a9231abc8b : TIMED_OUT in 3h 02m 15s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/75daf845504a46d3a8be892fef72281a : SUCCESS in 49m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4b6aa72fb1cf4fe6b241629ae8491b30 : SUCCESS in 48m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a51052d3eedd464994a9013b17335dd9 : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/548175bd2a5948fb85a6d6127b2b5c4a : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e24c347d71cf4d38bae7d055b4d035a0 : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/60c7c87616ed406b9a59a3d502d4b0fd : SUCCESS in 5m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d46c332851d94aa69f3434c17628227c : SUCCESS in 20m 09s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/788f8bba93bb412f8430fa28ea6cccc1 : SUCCESS in 16m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0e1b03f7961144c7a6eacfe812076c3a : SUCCESS in 21m 50s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0cafe5b30863402493d4a90fe745a5e2 : SUCCESS in 28m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3cc702c868b446c794e835c1605a05eb : SUCCESS in 14m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/caeb998ca0a94447b086010bff33f6d5 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/65e0a1c2c53948138a78ca20f250424f : SUCCESS in 16m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/12c86185cdf94594b3254e59ae6c555e : SUCCESS in 5m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1f11051b24494ce3a83877a4cbd279a9 : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/648c78a124d44375870b76ec9f15ff98 : SUCCESS in 8m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b0e6502a3a664c639494cb3498f845d5 : SUCCESS in 57m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1243c76816e4da2969fe9b4973dca7d : SUCCESS in 34m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/cbc84cbd79044fb39b06fb702a8a666a : SUCCESS in 48m 53s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/20c5df226322415e90aaca56e18d106a : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52a49cdd13784942bc440dee590594f6 : SUCCESS in 3m 55s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/471346a1dbb8474ba34f8c677846f93b : SUCCESS in 24m 57s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/def51dd42550402fb3472d09328d8b32 : SUCCESS in 38m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2dc44973f2e74101b1c7f9c3a8d9a268 : SUCCESS in 47m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cd4f293d44fa8f8df8446406498cb5c42be1688e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-28 08:48:02.000000000","message":"Patch Set 1: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d0e9542e3b0118cb7165a5ac18341ed11633d942","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-28 21:22:04.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"7e91dc6fa1118b80c431127d35bfad502cc368ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-05-28 21:23:26.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"90bd462a51c58a873f70cecb100330769cb5607a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 23:05:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bef38a45de2b44259ba8b9b9dca8e2d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c89088470e8467590729ef5d24821d1 : SUCCESS in 10m 22s\n- grenade https://zuul.opendev.org/t/openstack/build/9dfa0dcdc6324157b456666461e732fa : SUCCESS in 1h 07m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0946d045a0884a18aad65e81f588c196 : SUCCESS in 59m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5dda08e394e845f7920e9eca899fdbb5 : SUCCESS in 46m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a19a0c6de4104e5791d65e6242aebbfe : SUCCESS in 53m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/86b5a9739de94a2f97f4d9fccddf9fcc : SUCCESS in 17m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9637a16748e841fea52f62348c40c24c : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56d91f1f81784b16bd8d71d02654acd9 : SUCCESS in 20m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8adf6497baff4249b8bf60db8db6c6b9 : SUCCESS in 8m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ad3743ad66584cc08d641d532e8f17a7 : SUCCESS in 26m 12s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/94fb807e50d14ef18510618778aa6b0f : SUCCESS in 26m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0ed1f321db45478dbde05c9ddbf75e16 : SUCCESS in 28m 43s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e1cab57a71d74a02b2239c6c326dd8f4 : SUCCESS in 28m 08s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c95a392ca0704a81a0adf28284bd7bfa : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/16c2adafc95c440a9dddbd86e4cbe51a : SUCCESS in 15m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f57fce5f87234afd9429d49e8c45a177 : SUCCESS in 16m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/80e663ff8fdf4ba684dd9301cfc0fc7b : SUCCESS in 6m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0b91caac6274dbcb71b42e9d4f67617 : SUCCESS in 19m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/85b81f517b8841988d0ee3f03b032c6c : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9e66822ce9b14f8283d12614ffc7e34e : SUCCESS in 52m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f13b2b1458f04f81932afdd80171bd67 : SUCCESS in 47m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b60085df4a334c9b95bc6ca133d395c1 : SUCCESS in 56m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cd98495b2f5b4f45a0be3d8f0eee81ab : SUCCESS in 17m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84f07d8dd20747f79df056f2c32d66bd : SUCCESS in 5m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9c1a5db399884b6196d07a954256155a : SUCCESS in 17m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b4a5d7659aec47ee89806a028e5a7e47 : SUCCESS in 49m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d13ebc0536b9481a9d975a822f25716b : SUCCESS in 52m 46s","accounts_in_message":[],"_revision_number":2},{"id":"157ebbd537baabc272e2edfcba2dcbb91e6ae322","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-28 23:42:11.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9130379b3d3401fbcae8d12aed8e5f8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2d64573848454a489f123bc01beb9221 : SUCCESS in 1h 11m 39s","accounts_in_message":[],"_revision_number":2},{"id":"a872a6dd1be41544d512d6f5e14211523e157805","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-29 10:14:07.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f092b686e8bbcc9fe6b9d659cb06d80600cfb508","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 10:14:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9b88c44e6e68629dcf7e760fadc40e8daadfe814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 11:59:50.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4afc0b9fbcc24affba24fca2902d69ac\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58e48b24f3244e49b31522bffa5f8b18 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/dae8a293a7f1437bb52d81d20b22ab23 : SUCCESS in 1h 06m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/597b08ea42d34c60835185c065c41924 : SUCCESS in 47m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1ff509a480714bfbb27d94b398536386 : SUCCESS in 1h 01m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f1a3198bcf384b5c80a5ad3318a70227 : SUCCESS in 54m 59s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/b9e57e51550541c990fb76a1da3727e6 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b21880d818b2460f8c3eb1038d03a2de : SUCCESS in 18m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f04174b3438848fa9de9fe3eef0233b8 : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5865d094fed54b1891a178d1d1c122e5 : SUCCESS in 19m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f25450ef97d743eabd3636c48e48a1eb : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7ca822345f0f46208bf717523873794c : SUCCESS in 12m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/23ce93565f1d48028eb76483f490edf6 : SUCCESS in 16m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6a970da8a28a4bcdb70e2700977d7b4e : SUCCESS in 7m 07s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/566eacd520da4995bb6026acb012e907 : SUCCESS in 6m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ede84647ba1d43fca78e605e6d635670 : SUCCESS in 1h 09m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fd2d83f5c47a485d93dfc85e18281a90 : SUCCESS in 45m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d369c934129d4bce93450187578052f5 : SUCCESS in 51m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82bb7f423fd146dab85ebb2e018523da : SUCCESS in 20m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13b4705fd5e247c1a9ac3c30093906e7 : SUCCESS in 3m 45s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/19778184ba604ef1a86d46abfc982606 : SUCCESS in 19m 17s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/cac6de2939e74bf98ee493b3d5b425b0 : SUCCESS in 50m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/333a3c4e02a240a4b9d56234372e60a4 : SUCCESS in 55m 34s","accounts_in_message":[],"_revision_number":2},{"id":"3bbf899b50ee203d7b3d321febe24023b656a0f0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 11:59:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"67f9a9f26256ce43e2a4fb6a38ca8e53a2344b21","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 12:00:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/299744ba4f8640b392b11aad96147304\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0d6f5bb9aae43b1a24d4c14612fac5f : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/cd1fbdde22754ad79095f0e38485347e : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"313959ae9253681760a5639c1256aa036692a603","revisions":{"1905922765fb712eeaffdda8d3cdb5ddcab66d99":{"kind":"REWORK","_number":1,"created":"2025-05-27 21:18:58.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/42/951042/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/951042/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/951042/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/951042/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/951042/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/951042/1"}}},"commit":{"parents":[{"commit":"f64269b981753c21c87c3ec7c2b4bbcd5978292c","subject":"Merge \"Object-server: add labeled timing metrics for other object server requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f64269b981753c21c87c3ec7c2b4bbcd5978292c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-05-27 19:16:51.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-05-27 21:15:34.000000000","tz":-240},"subject":"Object-server: change labeled timing metrics sample rate for debugging requests","message":"Object-server: change labeled timing metrics sample rate for debugging requests\n\nChange-Id: Ia6e856ffaf8fd1b4a905e6976ebdc62ed5ddf32f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1905922765fb712eeaffdda8d3cdb5ddcab66d99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1905922765fb712eeaffdda8d3cdb5ddcab66d99"}]},"branch":"refs/heads/master"},"313959ae9253681760a5639c1256aa036692a603":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-28 21:23:26.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/42/951042/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/42/951042/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/951042/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/951042/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/42/951042/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/42/951042/2"}}},"commit":{"parents":[{"commit":"f64269b981753c21c87c3ec7c2b4bbcd5978292c","subject":"Merge \"Object-server: add labeled timing metrics for other object server requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f64269b981753c21c87c3ec7c2b4bbcd5978292c"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-05-27 19:16:51.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-05-28 19:36:51.000000000","tz":-240},"subject":"Object-server: change labeled timing metrics sample rate for debugging requests","message":"Object-server: change labeled timing metrics sample rate for debugging requests\n\nThe labeled timing metrics previously have sample_rate similar to\nthose added for non-labeled metrics. However the down sampling has\nnot been helpful when using labeled metrics to investigate customer\nissues, for example those related to object server REPLICATE requests.\nThis patch changes labeled timing metrics to not have sample_rate. The\nnon-labeled metrics are unrelated to this effort thus not changed.\n\nRelated-Change: I05336b700120ab5fcf922590d6a12f73112edb50\n\nChange-Id: Ia6e856ffaf8fd1b4a905e6976ebdc62ed5ddf32f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/313959ae9253681760a5639c1256aa036692a603"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/313959ae9253681760a5639c1256aa036692a603"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
