)]}'
{"id":"openstack%2Fswift~951219","triplet_id":"openstack%2Fswift~master~If590482a376aeed02391627931bd5b1f24569c6a","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision-patch","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-06-25 11:58:19.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-06-25 11:58:19.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-06-25 11:58:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If590482a376aeed02391627931bd5b1f24569c6a","subject":"Object: log breakdown timings for slow DELETE requests.","status":"MERGED","created":"2025-05-29 01:09:42.000000000","updated":"2026-06-25 11:58:19.000000000","submitted":"2026-06-25 11:58:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":56,"unresolved_comment_count":10,"has_review_started":true,"submission_id":"951219-p-timestamp-collision-patch","meta_rev_id":"31fc64c92f3b882fb43b4416fd410f44b6576a14","_number":951219,"virtual_id_number":951219,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 11:58:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2026-06-25 10:15:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":1,"date":"2026-05-06 10:50:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2026-06-25 10:15:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 01:48:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-29 02:32:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-04 04:03:27.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-07-04 17:52:30.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-08-22 10:49:48.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-03-10 20:23:06.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-04-15 17:00:31.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2026-06-25 10:15:30.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"667f558fda97e84c85ae67f67c734151d1502267","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-29 01:09:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a4203b918c6cb0ebb98e9dc7aa5c4377a8aca91","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-29 01:12:04.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"3435bf193d553c0da8615204cd29f7789eb2bec5","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-29 01:12:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"7d28b7edb6160969ca0be4989b171aee87ba590a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-29 01:14:29.000000000","message":"Patch Set 2:\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"d16562cbc307db3a10c02ab1789534cc15a2df83","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-05-29 01:14:33.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":2},{"id":"52a41c41b9116f923ad540ccb98889d2eec69282","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 01:48:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dedebb6abc644cfaa050eb96b192ff88\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c1a8d24ed9024c8285cd27546f5af6cb : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":2},{"id":"b78faf7f6a42fa0febd75c2420fb408d7cc65cc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 02:32:58.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c9df85f7b4f4bf9a781909196c750e3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0740f6189ed64077b510d12b0b270c6e : SUCCESS in 7m 40s\n- grenade https://zuul.opendev.org/t/openstack/build/1a3085ce5b56465cb50f819c95d56b0d : SUCCESS in 59m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bd63b31a02e2436ca9c1a2630a324de1 : SUCCESS in 1h 03m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5ea244d8943f48e09c11616908721564 : SUCCESS in 48m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b78a233776af4a80ad6892e7cd652b23 : SUCCESS in 59m 39s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5201453f6ab942c295b870a326a14d43 : SUCCESS in 19m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71fafd86208c454a95204b7ce43ea243 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee5ff216a79044ab961b97473e41bd8c : SUCCESS in 17m 23s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fb8ce074e73c49c09f948e6ac7f64923 : SUCCESS in 4m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/49142be12d37438e83a56912d579ea99 : SUCCESS in 14m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/060468d4929a458da3e0ca0b30827c75 : SUCCESS in 14m 02s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f9480f34366c4c1d87aa1bc5dbfeb341 : SUCCESS in 26m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/32494690ace841eea4707a075c51237b : SUCCESS in 17m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a8f32526f699492c84c657aaa253a115 : SUCCESS in 12m 03s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b3feae03690f4f84af974032ed7ad0f8 : SUCCESS in 11m 23s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/291a2a1328ce4b1cb8686169d0564055 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1ee017716be24b55a09bdb14b8dc490a : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/13018adb1f8e4c8c8eeb05ef53673b08 : SUCCESS in 14m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d48f3c083cfa481eae9a82d3a82117f8 : SUCCESS in 6m 35s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2193c6f36bb5454eb5f64dff91753b78 : FAILURE in 1h 02m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/43dfc34031ed43ed8ba17fb85802d1bd : SUCCESS in 45m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6799bf6dbe6b447486942113422cac36 : SUCCESS in 30m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d50cc2f8a1b45089e41519530298106 : SUCCESS in 20m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/080e6b7108874f16a705c662b6cd7e5d : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b30dc24534f5457d8e35dbb770325df6 : SUCCESS in 24m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e8de5bf5da804aec9afdbe3486ab44b2 : SUCCESS in 55m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/81d98efb8861461ebae3efe3bb6fb70b : SUCCESS in 41m 44s","accounts_in_message":[],"_revision_number":2},{"id":"e6f02033d91258f0881ea31255d77ffccd409533","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-10 15:08:36.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"73b05d68a4d5a007b38185de927b78ab4a03e910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:10:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3a3d348c143466584088492bb13b569\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c1e5f768f4f428e962b7d8c08f6516f : SUCCESS in 11m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/ba8aa2757d6d4fb3b2e9da0f4ab54c96 : SUCCESS in 59m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1360df48608b4ee59235cf592b23ae2c : SUCCESS in 1h 01m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cd10f2146c3c4d8280c2159bae53399c : SUCCESS in 33m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02b9502f727e4a8490f61f8376befb9c : SUCCESS in 33m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0d11127d6c2c4d21a4163d2f75a8330f : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/18d76f2c05f04c61813fd57b9edd49b7 : SUCCESS in 11m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5db5b23cdaa3400fa107ebd97b2b4878 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9dd8f7f2889948ee806c7bd7e89ee5ab : SUCCESS in 3m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7b14727485a445b4a60ac7feeaa96b75 : FAILURE in 20m 37s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff83c345af5d4bc09a26994946fe9ee0 : FAILURE in 8m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5d652cbc8b7c48b58ed7294bb12ebb11 : FAILURE in 27m 12s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/db7b903ca22f4f379f4fff862d59fe08 : FAILURE in 24m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e3a699457a6444208a7d1d4e28c1e4a2 : SUCCESS in 7m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87c16215dfc04c0c8d38fb6af2fb3638 : SUCCESS in 14m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d98eca23ea5340919a91da46ec1a3c11 : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/658e1b3e4d68423fb13721e5005aea89 : SUCCESS in 3m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9a44af6d10384f4888a90ef2b33a599c : SUCCESS in 15m 08s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/72410bfb093a4692b5dff2210c84c62e : SUCCESS in 4m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1745ec290164d9c802b86da9f01d92f : SUCCESS in 56m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/874cca2d14974ac98ad9d8e0067cca82 : SUCCESS in 51m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b22e8eab955f40c4b2f07de579a35231 : SUCCESS in 50m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/651be34aa46e4754afdc4b22f9e011b7 : FAILURE in 21m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e1bdd0a8074e37bf31a0fc405ccb47 : FAILURE in 4m 05s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/13a7b17cf1724cbabb632bf6e69e1ca2 : SUCCESS in 19m 37s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/06b992ac7e2b47f4ac5425c3bd97a1e9 : SUCCESS in 46m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/29a7f0bfdb454187aaf01c474d3f0461 : SUCCESS in 50m 48s\n\nWarning:\n  Comments left for invalid file test/unit/common/test_base_storage_server.py","accounts_in_message":[],"_revision_number":3},{"id":"3c8a5e44ccd4db8882030730519cfbd7d2fd0d30","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:45:26.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d2b1baf694b4264badfe5c44833585d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d06e6746d423417caf3456eae9e155c2 : FAILURE in 1h 27m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1bd836eb2e5c9ed2a809b469d140ca873339b298","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-10 17:40:47.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3e250bb205837772b6b5092188994fdc2cc54c4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 19:32:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9afa3af858704984a8a2923e634fc6a6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c8a35fe2e9d46bd9c772b9bcfcd1b8b : SUCCESS in 11m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/a4d80b97aa6e473db80b8e25ef470e03 : SUCCESS in 56m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b8b46d4778e4452eb4d784c9bb1b9c09 : SUCCESS in 1h 00m 26s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1d85565462e44f8b8f884227bb76d180 : SUCCESS in 42m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cc7707bac7124569b2e2437d47aff719 : SUCCESS in 34m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2a1dae478ebf4d7aa9ba4f9f2e0cd594 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5034fff770d04c42b4a1af3e36197140 : SUCCESS in 19m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/653cc054dc1c4d1db971c2d8cd57e6c3 : SUCCESS in 20m 37s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0cb8b216fdff43edbccc0ffc63300e46 : SUCCESS in 6m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8db3a6fc84374cd9b847f7d468c4a5ce : SUCCESS in 21m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ccc6b6d786cd417e8b09507754e9398b : SUCCESS in 9m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c9581212da9549aa8a4a657b19496b81 : SUCCESS in 28m 56s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d11f30b801554d0bbb1b7700ab18343e : SUCCESS in 10m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b331fb3a11814102bf0d48942c2bf1f7 : SUCCESS in 19m 00s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/556824c501874761a008a9d978f3e3cf : SUCCESS in 8m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b816451fa21d4a2381f43ca831ad17ad : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bbcf4f40417f4500a5e6a541b453d9be : SUCCESS in 5m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ef63f482d9c247f88f0595cfcfb89d51 : SUCCESS in 8m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2236ae940b234d87b2c5398a0ecad438 : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d3f52761fb0843f6a4ff51054fd44fd0 : SUCCESS in 53m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aea91921b941403aa0ff31fc264af474 : SUCCESS in 44m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d01383cb7a444f26a309e64baccf18a7 : SUCCESS in 20m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/681e07fb922844dabe06b0db9a213637 : SUCCESS in 21m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1671aa0040874befa3f6b3f3151c5a22 : SUCCESS in 3m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c1c5c1ed779442a49f5239da8b054ef6 : SUCCESS in 18m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d4ecf9028afc4a1dbe897addbec356c2 : SUCCESS in 32m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e1571a2dba8d4a6583272983f03f75ec : SUCCESS in 1h 00m 19s","accounts_in_message":[],"_revision_number":4},{"id":"77fa3d2baf4ce34dfe65ed3ea49f74596cd7b021","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 23:31:56.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cd72c6ff5bf46c29ca55aec8cecf18a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d1209a6a27a4c19ae362d981ac9b634 : SUCCESS in 51m 52s","accounts_in_message":[],"_revision_number":4},{"id":"e6f73028b1c6d7517d3fcb423dde46173f9fced0","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-12 20:07:25.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d5e0d3f127394fdfe043723fd4079139c2afdb10","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-12 20:08:37.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":5},{"id":"92b56811c88ab559cf125bdf76931db837436cc7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 20:45:06.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe3d4b7c5cfb4307bec100539941ad6f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b53ebc3d0262426fab877654672ec5a6 : SUCCESS in 36m 15s","accounts_in_message":[],"_revision_number":5},{"id":"5a745631137fc07bcec879e9f5340cf57dfbda9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 21:15:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33bf40e67853429ba629f1015033e7a2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/24abaf4c76de461e9e4c2da61ef1273e : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/2a1a4bf9b9a74688af50adcc0188d527 : SUCCESS in 1h 02m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2d53ffdb97594146a0782167fdbc0374 : SUCCESS in 58m 42s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fca3ce19c2a24ca99ff9df6584c79fdb : SUCCESS in 48m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/431fc249542949a89d0075f6c8ab8c9b : SUCCESS in 46m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8f0dd23182034d72a1ba79735d6ffe86 : SUCCESS in 19m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5835d66e17874018b6a3a313878b300b : SUCCESS in 19m 15s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/127f63cadb3a4df6a9fdd30c32f2b9a5 : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e83d8cc55b4f486090f81743278167f1 : SUCCESS in 4m 59s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a17c001384a0490e82dd11603ba6946f : SUCCESS in 18m 55s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/30b5c97218d74606bc2180e1b6870233 : SUCCESS in 17m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5bf931cc2e3549ba86b48c0b29f53406 : SUCCESS in 21m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ce0b905769454515b2136bf2102a8ad3 : SUCCESS in 22m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e4deee4ce6af43508a21f550be085904 : SUCCESS in 14m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e9c457feb1894a89afcaef778ec22a81 : SUCCESS in 7m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/916654fb5e424de2a450dd83a712a644 : SUCCESS in 15m 03s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/50b882b857104e40ac18556cb996eb91 : SUCCESS in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a301227bf96543af9826d4fb2b970e74 : SUCCESS in 14m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/417c87ca4b744c28be2b0c851e695b45 : SUCCESS in 3m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4d37371f1a964ff4a69df6dc5ffa7b4b : SUCCESS in 49m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/0dec1f5f7c874d129400e9ef4aee3ca1 : SUCCESS in 33m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3db16feb53a04a71bef453a28693f366 : SUCCESS in 26m 20s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4662cbdaae9f41dc87a14d4bd538ced5 : SUCCESS in 20m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d7ed90b913e47f38d99a746e29d9055 : SUCCESS in 3m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ae4d29a8f5024f83b6df473860428dd4 : SUCCESS in 17m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2f6c16a793894a0581accc0b4b05e8ab : SUCCESS in 51m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2f221284f78f4ae9b08ee7393bc844f2 : SUCCESS in 27m 01s","accounts_in_message":[],"_revision_number":5},{"id":"203582c7746a244ee046328b5a8395129cb2c087","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-02 21:03:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"d39b98216bcfb0eaf92e5e1e4ca05403fd5e97b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 22:12:16.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8976273b5a744cc29550e22e865b6ade\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33a244b5399548148d0d8b140d640b0b : SUCCESS in 9m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/95ec785c852f4825a81faaa73e9fb3e2 : SUCCESS in 56m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/29cd9463dc8c4072985eacbe79d9a734 : SUCCESS in 50m 13s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/092c0963367144e59413de220f3f91ef : SUCCESS in 39m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/131023fab96c458db3dd82012e82e6b8 : SUCCESS in 1h 03m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/804901cd44244b3ebce75ec69300a2c0 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bafef826efb4448aa56331aae02c6606 : SUCCESS in 10m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ab6680b2e09c40708c15f6675bda6307 : SUCCESS in 20m 21s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d458461451824c1eb8c71aefc718084c : SUCCESS in 5m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f4dadee5b17e41758efad708d4f2a849 : SUCCESS in 14m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a5fbd91dff82470cb7d84a5121fcf732 : SUCCESS in 8m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/470dc52d93ad46aaafccccbbcf378a41 : SUCCESS in 19m 49s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3beeb69630184a1cb89d22de9775d4ff : SUCCESS in 21m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d93c2d7b07514e3da1189a2d79f1fae1 : SUCCESS in 13m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/277a146945704baf80b603a21b68bf9b : SUCCESS in 15m 27s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/da5adb168c6b40f7a254a0849919eb2d : SUCCESS in 13m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5bf175283dd54a1f9b8c76f156589e95 : SUCCESS in 6m 41s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/16eb7e6d6f1a4b859b66c0cffa1b5b0f : SUCCESS in 21m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e2c052367be6494f9f137e78cd84ea44 : SUCCESS in 7m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/283c10df009742cbabe77399c8872df1 : SUCCESS in 56m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c500362be13043c1ab8984cef42e26be : SUCCESS in 19m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d098685ef71749249ecb68291b8f4b98 : SUCCESS in 36m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2d03db61c514449a9ea98a4e13147d2d : SUCCESS in 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a97f56fbae345b08936064351d44073 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d67dac0989594986b0ba9e63979b180a : SUCCESS in 17m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e56239113c32425d821ac12c35f1de53 : SUCCESS in 38m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5f9b8973080243b48a4b9da23b4e2e61 : SUCCESS in 45m 15s","accounts_in_message":[],"_revision_number":6},{"id":"86dda1586b94982a02c5efbaff232b7398e80de2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 23:17:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/757af09e98144d57a40006c56a75de9a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f62faa7cdc424e0f91979ff5af4a1c07 : SUCCESS in 43m 37s","accounts_in_message":[],"_revision_number":6},{"id":"506cb6b1b3f5adb15c7f0dae6ee027908c943a8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-04 01:31:37.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7b6709afd27c88214af82efae995de8f6793505a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 02:42:54.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8f1f553e4fe4e78908acea74f39641d\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b27c96d0fe248cc91d86cbb94e7b008 : SUCCESS in 1h 09m 38s","accounts_in_message":[],"_revision_number":7},{"id":"6424593efe83351ad4173b69b4c39bae015bf450","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 02:44:58.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc3e29b1bd9b40e6a7dd0dd37dc7c07c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3d4575bb7c142db9660932366277f83 : SUCCESS in 8m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/3470f445044446188bc4fa10b943e041 : SUCCESS in 32m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bbbb76a5d50b4539ac60c0269651b229 : SUCCESS in 1h 06m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a0218bfa514040ff96d515d1685fbb7d : SUCCESS in 39m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/024ff164e9d84a63a3b79a3ce0ea6ea8 : SUCCESS in 36m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee29de9fd6c746a8bf7cb7909c371071 : SUCCESS in 11m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d46272ac3bb942daaa4c45d91ea845c0 : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/13297ce4a46c49569da3b48d2472f7ef : SUCCESS in 22m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/087ba031074341ef82db254a33be5b8e : SUCCESS in 5m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4de4445c32f94667a07bbb022b5cf0ec : SUCCESS in 15m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1cc767fd76c54a7aa70d1e3b38570085 : SUCCESS in 15m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/771fee646c0042d78fa81367bbbd6369 : SUCCESS in 22m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d0430819528744489ce9e0d42514c3f5 : SUCCESS in 28m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2568f1a955fb4aa2b11577e14552f956 : SUCCESS in 16m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b57bdf815f96473faec54eb6485b80b6 : SUCCESS in 11m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/32f290dffc6f4050b3afdd422f28fc2b : SUCCESS in 15m 50s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1aca97384ba94e499ee7fa59af1ac429 : SUCCESS in 5m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/02e493e99d1741ae93ad4f10a3b30123 : SUCCESS in 17m 54s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7cc358f520af43b5ac86674a5f6610aa : SUCCESS in 7m 02s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ba712353cb549abaae1771da9d4a485 : SUCCESS in 1h 01m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cc8e61f28f3e4c44a8ab4224ed47d8e0 : SUCCESS in 50m 58s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/693a715341204a448f6a99a5b541dcf3 : SUCCESS in 23m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3084c77bbad640da8328008238570c62 : SUCCESS in 7m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d7b317c9f4f4f7d97d8359e6630caef : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d62772603fd44234aaa5a9e23a72c01b : SUCCESS in 21m 12s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5ce4839fe1b94027b4ba3e0e9742c6b5 : SUCCESS in 49m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1953caaca1944f54bb20ba7e501da5ac : SUCCESS in 55m 39s","accounts_in_message":[],"_revision_number":7},{"id":"57e63c9a7abe87d19119f45508b97c75197ac09c","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-04 04:03:27.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"951adafb9c1ee5692018ce2e49b468ffffc2046f","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-04 04:05:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"6dd3550ae7576c0e5fa8a4ee663580b6e03a5346","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-04 17:52:30.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"31716740b9e2727fcce63c8db6fa0ea27a735a54","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-07 12:04:28.000000000","message":"Patch Set 7:\n\n(6 comments)","accounts_in_message":[],"_revision_number":7},{"id":"b0d77adda82794debebef3c38a1ded7ee1d3ee5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"date":"2025-07-14 06:22:47.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2a7b72ddbb25e30e105e6e72955a46e287f09a2d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 07:33:34.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/092c52596ad44b24a70ecd3db6ce5d56\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c316c75b771c4a7c82c5c11c14e92b37 : SUCCESS in 11m 02s\n- grenade https://zuul.opendev.org/t/openstack/build/29885cea53a347df9f8a1faf840e1cb5 : SUCCESS in 30m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d3d91409a74f484699d51b5f167e53b4 : SUCCESS in 1h 05m 40s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8a3cec312fb24a5d863e53d0ac5412b4 : SUCCESS in 37m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/73d7690343694988985424fd1ec8cc3b : SUCCESS in 32m 22s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e49b687c4cff460d8868d26379b6fbe4 : SUCCESS in 10m 02s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6d02ce4d22684c31ab9f51b3b054ad62 : SUCCESS in 18m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/519ec72e0c524f6a92d5c6abafd8b37e : SUCCESS in 19m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/11dff5241881460ab40caf9294bb48a6 : SUCCESS in 5m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/021acdf3a7a54d59a5836e68c9dbf7ee : SUCCESS in 16m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f7eae72deaae423fb7b2bb89131b40fa : SUCCESS in 7m 41s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/32cdae09abc54793a6dd1a4b83fff0a5 : SUCCESS in 26m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c2011f6edb8a4a54991871493f29bd8c : SUCCESS in 22m 17s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d163c66b8b3d43ab969389102058a858 : SUCCESS in 15m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1ddfffe8bd9546efaede2ae54979aefd : SUCCESS in 13m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/484f5b4da1244040b795b5fa5341dcd7 : SUCCESS in 13m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6bd658988a1242eda8e4eb5c4f97f8d1 : SUCCESS in 6m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/84a7d0e533504813a776ccb0882dd89a : SUCCESS in 14m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ef7e33a2a35641dcb261fe676eb92425 : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a67c7ff1853c40c7a65df6dc84fe20b1 : SUCCESS in 54m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bbfaff27dd74fe8a13b0cb90bddb4af : SUCCESS in 20m 03s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/01e36b413dba411abc5849da048beeca : SUCCESS in 47m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/37a473df95254387b276a1d7c65d96a7 : SUCCESS in 16m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/be8681889dbf4066956c219c197e2c8c : SUCCESS in 3m 19s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/611c4207d39d4641b8242bd4de7257cc : SUCCESS in 8m 22s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7882eb9416524e9ab137890cf0ba5a0f : SUCCESS in 51m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a55cd1252e3c4504874d9af01a592b49 : SUCCESS in 26m 48s","accounts_in_message":[],"_revision_number":8},{"id":"18f2ea8c8745888501380b8d6fc63c42c1d769c2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 07:50:18.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a74131163b0b4db596b71e9777e3b83a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cbae141e2f1847549a1b8edc66f34962 : SUCCESS in 1h 25m 24s","accounts_in_message":[],"_revision_number":8},{"id":"65516389b2466a5556576f6fbe01d886c716cb0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-14 18:58:34.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"77ec1c1fa5cafb80ec47e632e4edb1ee83fea414","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 19:51:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/170b6595f53747a8bec8d1ffeaa3e39a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40f955193cd34d71be489ce13fd2b3e1 : SUCCESS in 49m 47s","accounts_in_message":[],"_revision_number":9},{"id":"0a533b9fcf11a353be62800ff413df6eb0ae32d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 20:07:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b51a0d6b138f4b50bea94f71965dee20\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82f114733d2d432d9486a6f5733f74cb : SUCCESS in 7m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/b4138e4a9e1143aba47b32884e93ca07 : SUCCESS in 41m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ffe32686a04c4ffdb2bfaa8dd48f5768 : SUCCESS in 58m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3e604b7740b74dcc986021e0aa95a7ff : SUCCESS in 38m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e601e4e8e954997a9a61b8682f99a95 : SUCCESS in 1h 04m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bd302adaf79a49f189d86ef7d169d635 : SUCCESS in 19m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8b7333a102484002a9b790a30471fdaa : SUCCESS in 16m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/291ec874b0f348659a1c4052c150e0ad : SUCCESS in 21m 40s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f9b1ffdc518410f88127ebfc7251768 : SUCCESS in 4m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/25705a31d0d540b4a7f30e8a23cabb07 : SUCCESS in 20m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d88066c776da45a9906c1cfdb3789d01 : SUCCESS in 17m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9f9623a871d04dfd806889f2774680f2 : SUCCESS in 36m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/30e59fd37f274ee2b470dc6392ed4889 : SUCCESS in 18m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/828a7139d96c4fd5a616e337a1aa6506 : SUCCESS in 9m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c565355995d34b5ea5fd932a0e436eb3 : SUCCESS in 13m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8a7d516b53a4434fab73a60e85f66d76 : SUCCESS in 9m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3b8bb0bddfaf4e8e9fdd23205a2704fb : SUCCESS in 5m 45s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d642fc4aca82465f9f8d6c9a8265e11a : SUCCESS in 14m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee7ab766242f4f27ad26e21bcdddaef1 : SUCCESS in 7m 39s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3c3dd89904dc4c68908293f459d16213 : SUCCESS in 1h 01m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e98a305a41ef49009367800b380a90dd : SUCCESS in 20m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/885af42cdf1f43bba943cdd95949f768 : SUCCESS in 32m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af022c08c3e547e594b7d70e11994589 : SUCCESS in 20m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f6bbbc361464954a8606d735b5ddb09 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3be095c53a464295ba9dda44800e4d21 : SUCCESS in 15m 28s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d2ea96d0a8a74cab91159256db6291fd : SUCCESS in 47m 52s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9f151f0e4ae442a59def3db5cd996dc5 : SUCCESS in 29m 43s","accounts_in_message":[],"_revision_number":9},{"id":"b792eb1e0d2fd9133c761f77e1081cef850d4086","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-21 23:58:36.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"a0ffe498bc37bca6505f2acde11c05216403a2e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 01:10:44.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10a68c0d471c422da30ce1fa75e18ae8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6b37b60d561e4d9ab4ceeee513bf948e : SUCCESS in 1h 09m 21s","accounts_in_message":[],"_revision_number":10},{"id":"c91a48f562b9726a4c64cb1220815de1ec2523ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 01:12:38.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0b968d6edefa4b9bae2e00f621a33222\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/772b5969cb7e406d86b0dbab0fc85809 : SUCCESS in 4m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/3abaf46780dd4d3baa1501f10c0ef74c : SUCCESS in 58m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/22a5e0319e37400e855a6db70f21f639 : SUCCESS in 35m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9e6d4c5805744619827915bb701440fb : SUCCESS in 29m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9f4b03d4d0234ba48c17381442fa3e8e : SUCCESS in 1h 03m 54s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c09444a65eb1451ba62d150e1626c959 : SUCCESS in 11m 47s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/063eda18b259494f86e0b1415d6c9401 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1a7b2c34de0b46358e4d0e4060ec729d : SUCCESS in 12m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a144100943de497198539d6a7caf6547 : SUCCESS in 4m 10s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3ba2352a48874600bc3d24168abd4a03 : SUCCESS in 10m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/ff3def64e79644f7898266e1271e285a : SUCCESS in 13m 25s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/fefdd5e7ea6e41449e6ada18ccc425f2 : SUCCESS in 23m 30s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/261baa45063944bebaba2867982ea1a9 : SUCCESS in 23m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/dd54a5d8e7c1483ea9036689f3e06698 : SUCCESS in 15m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87dfa52c85e74bef97db5dd96fdcfcab : SUCCESS in 14m 15s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/aa67c349c62b46c78b25d8ea483e56ea : SUCCESS in 14m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2c4fcdd70fae42c1920c99c19ce0499e : SUCCESS in 8m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/809c3e03137e4d37931b802cfd7abd48 : SUCCESS in 18m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b24bc35eb8ce46ad9acd307818de6076 : SUCCESS in 8m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4cb412ecc8fa4d00ada0a2ea0bf258d9 : SUCCESS in 58m 03s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f149af52394a472e9d916f71a1de5a96 : SUCCESS in 33m 50s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2277ff2433ff485e860cb90ffcbe3d51 : SUCCESS in 52m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16651ac76dce4bd48ba05a47364d40a9 : SUCCESS in 17m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/294422c07c5c4b2aad45e83dceffae94 : SUCCESS in 5m 20s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5a83aedcc1c648049df7af76a4392387 : SUCCESS in 11m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/870e8b80574d40a78ff8bbe8416de2a1 : SUCCESS in 27m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4083f9a363fe4a9b9e4e7f1979de9dbb : SUCCESS in 29m 01s","accounts_in_message":[],"_revision_number":10},{"id":"335f4a6f335d03726cfd9f8958dc48678e40670f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:28:06.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"cd7eab79005ed422c3b7fb4771cd013e0a80def6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:39:09.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d97286ae290d26465c11bda7133be6ef6d5990dc","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:39:38.000000000","message":"Patch Set 12:\n\n(9 comments)","accounts_in_message":[],"_revision_number":12},{"id":"b572ebcb0707cc9cdfde4c0130d98560978c0238","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 06:55:56.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bf1237cee8cc48e6a2f9b517ddf74c7f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e844296afce40e984df52bbae5a2f1e : SUCCESS in 12m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/b9eee9dbf87841899a6952680bdc0116 : SUCCESS in 29m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbbb0aa42bf244d4ab036d90015163c0 : SUCCESS in 1h 07m 58s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a592abda37af473a9175b038b3c890ba : SUCCESS in 49m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/188b358aa3574e899af7b6eff5ec90cf : SUCCESS in 32m 14s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/66e5767a2d5346338617c47efec207ff : SUCCESS in 10m 32s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2ecddd13c7cf4af7b65087c0b702bf9c : SUCCESS in 18m 59s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49baef9e50bf42a3919f16d141aa79e3 : SUCCESS in 21m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c7dac1c940b64cdabe18f3912b4e18c7 : SUCCESS in 5m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d6b76af67beb4c5180be777bc24d438b : SUCCESS in 27m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/129b22b087f44717b49df407f0af8318 : SUCCESS in 21m 05s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0ecc7c5ad7f1475e8325f076056a27cc : SUCCESS in 24m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/95fab4a1b6bb44e09f6fa72869fe9cdc : SUCCESS in 28m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5797a94acd134b26b2d4149f9cdbe1bb : SUCCESS in 14m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f5ce5113d27c49f5b764bb4b25665b26 : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b0d29da64f374513a39584f7ba42b79c : SUCCESS in 16m 42s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2518386a785a494cae53eee977f4ae2f : SUCCESS in 3m 48s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/caeb75a6b39d43d48d101c6923533c3b : SUCCESS in 8m 47s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e080ff4ee9a0480f8f78f33a6cf02265 : SUCCESS in 10m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b820bdce4e3646269514c85ec317021a : SUCCESS in 47m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cec41f6e3a7448b186cad5f95ef5010a : SUCCESS in 37m 07s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/10f780a1ee904b289e48ae1958693875 : SUCCESS in 34m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c78b3ac8082f48fd991fcce3e4f0b828 : SUCCESS in 21m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293218cc64cf42019311a45dfc3154d3 : FAILURE in 4m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/446647abc9994939a787a0840fd7d1cf : SUCCESS in 16m 58s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5bd6262137a84a46bea00283dd6f5b02 : SUCCESS in 37m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/453f77bf224b4428acedd48708d24c0a : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":12},{"id":"fcee30b03dc0f0e125ce9fdac3699d83d9b81efb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 07:33:58.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/532eac6b7acb4278a439b019deb87e5e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cc2c2d3cda2d47d8b1911bc9a03c6bbf : SUCCESS in 1h 49m 39s","accounts_in_message":[],"_revision_number":12},{"id":"db1eb31bcb09a08fbece947acc27e4c33137ca5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 15:54:02.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"fef0ec60701b980f867030edc22239ed6db27780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:01:43.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9cee95e084e43f983aa0542414b2101\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac856fe5ff849f1b81fd49e52079404 : SUCCESS in 8m 17s\n- grenade https://zuul.opendev.org/t/openstack/build/4204b6a92cc54623b0d62485ccfa518b : SUCCESS in 48m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c68d0f4effe24526ae7006065d3f27a6 : SUCCESS in 57m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/466776b71aad4880a5065a48c7974230 : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb1fa8ba803d4f9fa8c6c82d4b3c1787 : SUCCESS in 1h 03m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cef2109f17c94126b35f7b03c4fd9be4 : SUCCESS in 10m 39s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0052c2d62f9b4482b6bcef1577b7ac1b : SUCCESS in 20m 29s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2451d8f50c3d4a3683a9f117f7110b8b : SUCCESS in 19m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/81a2ca4c49654f45947895e69404c385 : SUCCESS in 5m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dca68699622447dea243172c88caaab9 : SUCCESS in 11m 35s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5d045892840145f99ed40f669aa0549b : SUCCESS in 14m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/192e7b03098c4f30b8cb4dcd06ebc796 : SUCCESS in 18m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4bbea9b894c542ff99903b7857cae1b4 : SUCCESS in 24m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e120394691f342e2ad5f1c90fcb028e5 : SUCCESS in 14m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9e03862aea6842099aa004e2b0606262 : SUCCESS in 8m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/60559af04d2448e79ad987d3202cfbb8 : SUCCESS in 8m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4d4d3c7131924fb7a1c848f9cf3941aa : SUCCESS in 7m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c0e3bfc10ac54c49aba8583176406734 : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/feceea879e09423d872aeb1db08f75e5 : SUCCESS in 8m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8eca065921124979955b40fd2a95d4c2 : SUCCESS in 43m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6189c4bfbe214e62b032b9f5a7e8b2bd : SUCCESS in 57m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5f36980e52504f14b137ccae32f7867f : SUCCESS in 22m 13s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/362bcb944f00405287b556050bf51d96 : SUCCESS in 31m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87fe2d8297ea4807a2eb941ecebc291e : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/de7f640a51ed4e79ae41f7c77666ead0 : SUCCESS in 9m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b56242cfdc742d1a046c15920936b8d : SUCCESS in 52m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5877b61d34a749198e0e262ccb858ab8 : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":13},{"id":"a3ca32328ac1d5fe37a45f14319d484b8827e46b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 18:00:14.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8cfbff390ad4e06824937ffd258fa07\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3e96b9d2f5f74b66a6aa413efcf83551 : SUCCESS in 1h 40m 22s","accounts_in_message":[],"_revision_number":13},{"id":"215ffe17de7f73b8b751b07eef4ab83602556ac3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-22 05:20:44.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"a56193d9e4ab7ed2cbb3afd8551b4461ed002a21","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 06:29:16.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/732534b3e831415c94df5fa93734eed3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cfa66237755b43a79714aab528ef1ea0 : SUCCESS in 13m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/f9c3a43c1d1c463c838cec53baff92ea : SUCCESS in 1h 02m 19s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/50c56726bda6450f875865e1c293a354 : SUCCESS in 37m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ea415b7a355d4f1882c926f082dc9217 : SUCCESS in 50m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e0cd1ecaa5ec4ccdba1ff9fb0a3e9e69 : SUCCESS in 59m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b408485bfe014978886ad0728b5708a8 : SUCCESS in 15m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/31de98f0bea648a798405251e4e3137d : SUCCESS in 20m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b40f0c97202b4a1a815caee42018290e : SUCCESS in 18m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/28d9da4e53a14e5baaf34799b4971934 : SUCCESS in 9m 00s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/247d91cef3eb45cdbfb0df99f7bcb521 : SUCCESS in 21m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8fb6a1ea231a4d57a6c880212cb25d3f : SUCCESS in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d3d0af7f9bc545b3992a663262a4feb6 : SUCCESS in 18m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/319b13b430b64aa0b77b9a36de1ad286 : SUCCESS in 32m 56s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/861978ab543442828b3bc7f29f2043da : SUCCESS in 14m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6b3b51d5f3e24a0391007291c9726319 : SUCCESS in 16m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d1d64edfee954b829b3c77eb304439cc : SUCCESS in 19m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/83abcecfb513474c8c4660fd1d2f0581 : SUCCESS in 4m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1a44fbb169a54e038332a678eca0a7ba : SUCCESS in 15m 19s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f188d262e2c497594ba8cddd6f73ac8 : SUCCESS in 8m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/659d71cf84034e5290087c3f3373f3c3 : SUCCESS in 58m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/eaf62fa91ac94f158a95e0cdda615641 : SUCCESS in 51m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/bb230e2f754848338b37e6ae7f56807d : SUCCESS in 49m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1ecae846d2b44b8f8e66b7eb46b8136f : SUCCESS in 11m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/360cce7372e746f29ccf6ef4c5ab13a7 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/62247d2c04fc4ad99bdf3252db982f57 : SUCCESS in 16m 06s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/57986e8cbaa54152841671db66524a90 : SUCCESS in 49m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e0c1bc43ee9d4c4d88f8582f13695ff2 : SUCCESS in 56m 11s","accounts_in_message":[],"_revision_number":14},{"id":"a9e85f70dbe6c170b1d527c64521ee4d4fe2637f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 06:57:14.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4941f221c1524bcaa9843ab62e0e8f65\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e29a594af9cb40c09afcca8027e010b0 : SUCCESS in 1h 33m 34s","accounts_in_message":[],"_revision_number":14},{"id":"b7b5fe80814e2e780aa7a7c1c2dc103f8a5d9224","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-22 10:49:48.000000000","message":"Patch Set 14: Code-Review+1\n\n(7 comments)","accounts_in_message":[],"_revision_number":14},{"id":"e7c48bcdf1aeeed0c5dbf91b5772fc2c1f593215","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-23 01:17:19.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"95556eb3fa51a200043d8b197d8085d443501c10","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-23 01:17:32.000000000","message":"Patch Set 14:\n\n(6 comments)","accounts_in_message":[],"_revision_number":14},{"id":"2f296477cbad2ba0e73b511d6a90e7283dcf45a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 02:21:40.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a2cd73d0bf4540adae268de57567a3a2\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/872d83927e6a4dd6a83fc4a91279fb2f : SUCCESS in 1h 02m 19s","accounts_in_message":[],"_revision_number":15},{"id":"ea6b63059b3488dcf7b7965162e586569528f19c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 02:32:30.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08c13a45f8f94ae88950d48690c6693b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3549edb79bb049c8acaf1a8748e0c75d : SUCCESS in 5m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/37d14af2ef764931a51f0aa013e7adec : SUCCESS in 1h 03m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/baabd09f6bca4532870ac5b19500a428 : SUCCESS in 43m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2bb703822ef24997af0611d6cc30feb1 : SUCCESS in 30m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ecb25e244aff4d2da67b50001d596d87 : SUCCESS in 38m 09s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2d438369e80435386c43762f939d137 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b2b6d16a40174c1aaf4f9e2381b172eb : SUCCESS in 21m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee32b8c42fc54ff3ad32ef300458ab7f : SUCCESS in 10m 38s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f3bf6739cfbc48369e39f26b4e0b3fb0 : SUCCESS in 5m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4614c70cfe3f42849efe4571dc6f8998 : SUCCESS in 22m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0122941489a94cb2a53b7e00b93a7043 : SUCCESS in 24m 46s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a71fde6c5e464129b15bb18a977b0dbb : SUCCESS in 30m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/2023e1b138a14f66b84e9e72c4cb6d4f : SUCCESS in 11m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2052384c3b794d49ab0ffa86df5840ba : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/766cb813854b492ba7ac23dd1aba39f6 : SUCCESS in 15m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c11344c5e39d441aa9fe001c523cd0dd : SUCCESS in 14m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/44c44802da9e41899d83a62a52c8c886 : SUCCESS in 7m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e2203d5fb0294050924e2a067c65643b : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d7fbe23443a543d8a0ec7791ea739aff : SUCCESS in 8m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c60097e00b384aef9488aa7a03312fd5 : SUCCESS in 59m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/067e12d03e88436eab08602ace407999 : SUCCESS in 53m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c904d7899c794d65983fe1ae84f81783 : SUCCESS in 52m 15s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0fa5d7e7dc6c4bf785922d5c58934c52 : SUCCESS in 26m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34af7ea0e1e542108263de9318bae3c9 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bc0762c9fcf04461a6c772394268b5f7 : SUCCESS in 17m 36s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/abd4440954914f34a5f3e4d99f50267a : SUCCESS in 56m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6bdb24351ea47aab42fdcb7f5414439 : SUCCESS in 57m 15s","accounts_in_message":[],"_revision_number":15},{"id":"bb1424b6fb2434a70a1cd452b61a9dbbff6ec77b","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-01 14:44:39.000000000","message":"Patch Set 15: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":15},{"id":"72bd880d3883ca1a4947161014c593435ee4cbdb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-02-10 04:43:10.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"6f8d108c30dd57c4ffe7a64f8e9b3bef4c8d89bc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:16:53.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/521f9a79da6c4e56a46fe54ab6aa15bc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cccb78960de44419aa334e8618dc4231 : SUCCESS in 30m 49s","accounts_in_message":[],"_revision_number":16},{"id":"760dd1e31ee2cf71ec1194d7ae324a1ea19a1757","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:40:37.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0f6bbb3eee3841e9af67281f59da1833\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/704fce09d9ca49929b20b4b79e2bfb27 : SUCCESS in 10m 05s\n- grenade https://zuul.opendev.org/t/openstack/build/afc32ed95e2846f59fc27648096d0507 : FAILURE in 7m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8cc428d86743480bb824db140a180943 : FAILURE in 12m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8d6df743cccc45a395cd3eb3884d3ec2 : SUCCESS in 40m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ed2a186af0b4062be8a93b1f9b58ca0 : FAILURE in 50m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/cc2f894d3335435f8b5b5126fd4c5177 : SUCCESS in 11m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0aba589cd2854ed680567ce3bcd59f72 : SUCCESS in 18m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/79551adf1e90418fbc3bcbfc7539b66d : SUCCESS in 16m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a06673619770460bbadeadd10a9d5665 : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/770dd8c61e8946f2854f01f758e9b829 : SUCCESS in 16m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/69c71641185b45408dc458e993b5cdf3 : SUCCESS in 10m 29s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2bb9ab970ce44fdb9509e57384377ae6 : SUCCESS in 12m 58s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6a3d69352ba04ed3884195a4993faabe : SUCCESS in 30m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/07d332fa03c94bee80d28b67d7b69f6f : SUCCESS in 13m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d19fe1cea0c54e338b8f36273f5b108e : SUCCESS in 12m 25s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7a2fe2750f964ee19123ada45fbbcb7f : SUCCESS in 15m 38s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b9cca85b814e4c4088fe965072402b1e : SUCCESS in 4m 27s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/accdc922a42548ea9e6e57494f8b8c3a : SUCCESS in 17m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/91a527beaa604087bfbbb49dc983590e : SUCCESS in 8m 34s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6c715c380216461894db5eb799918805 : SUCCESS in 43m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3db4cc5dd63b4bf48f30702426c1429c : SUCCESS in 33m 17s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34e8d4b1c51e4f8a8ca981023ed35973 : SUCCESS in 33m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7c6b195cb96e43a3872ab67290d00fef : SUCCESS in 31m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a774a8fc3584579adf63da9a1330539 : SUCCESS in 2m 28s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/39e88eceb4cc473a9f032890bf923d58 : SUCCESS in 19m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/1b8abc2140a741138c7e7efb02fd6d6f : FAILURE in 34m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7c4187def0ab4f10922eb5269dddcdad : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":16},{"id":"12b1b02fd2c1560ade510d5fe29aef49c20d5b48","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-02-18 17:11:48.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"8162908e2985fd7ad651d28de1031b7b2f835027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 18:23:33.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ddec33ea03f40fb8df15012ecc84343\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a41915d95869477aa34cbb9fc03972e4 : SUCCESS in 8m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/5b5c9b629dbf4a298842570e6e39671d : SUCCESS in 1h 05m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c4dd7dd4a6514338a26be8c2505af04c : SUCCESS in 59m 44s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/673a03b9adc5488c867dc4c3e7b5606c : SUCCESS in 1h 00m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/afebae55642646aabe7b2f06c657d144 : SUCCESS in 1h 07m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa67356402c44c0d9f638bb02f742118 : SUCCESS in 19m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89efca31c3de4870868b2511ea51af8d : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/baffeb1a4f934cd0a799e6d3abfe3300 : SUCCESS in 20m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/12610418c4f449348101ba33adc2f632 : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/35b9a436b1774c11b4b80bad8e70c1bb : SUCCESS in 27m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0741a66141c741f1807e58f770627c87 : SUCCESS in 20m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5b15c3773712484885d6f512f783f69b : SUCCESS in 43m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/20300757aab0489e91cd9675d4af8751 : SUCCESS in 32m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdf5ddf4354142c889aeb057d2a6999b : SUCCESS in 15m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/64d77093f5fb4a2b886279730719e0d9 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9c3f3b49783441ddb252d81fc256b34b : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d445e118d9745278e92273d65a58f5d : SUCCESS in 6m 44s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c53bb224f3f549ec9cac19cffd0c4417 : SUCCESS in 14m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/02e1886d355c440780aa9c4804297ec7 : SUCCESS in 12m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab93ba89e4094a07b74bf87e8aba4631 : SUCCESS in 1h 01m 14s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cb41a2750be547778a50ea37803d0344 : SUCCESS in 50m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f41f1a6ca04244b89aa725d1929f97ce : SUCCESS in 53m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0c91f8cee1d1487a9266438b65e9e4e2 : SUCCESS in 27m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73bfbd6652bf4b358a8e94860b1c9317 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/622e8f1eab6743978ff9a1959ddaa4e4 : SUCCESS in 8m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef91900360554f7096689eeb2d18201f : SUCCESS in 58m 17s","accounts_in_message":[],"_revision_number":16},{"id":"0a44ce8f6137d18dc4199b1551efe4d883ee533b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-18 22:47:35.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20287b49aacb435998fa81c8958b9709\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/14052a59bdd24ee685cb0f4aa538a1e4 : SUCCESS in 49m 21s","accounts_in_message":[],"_revision_number":16},{"id":"91bed236bc3fea0cadcbc4db591322ded6867a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 05:39:12.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"d3d5f636b1780af37513d912717f9200c894f9e4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:07:26.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ba3cb74e11a466a99247325a1d07c6c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1db94527837a4f4791fa7ad046a0111a : SUCCESS in 26m 09s","accounts_in_message":[],"_revision_number":17},{"id":"aa112a1f35a2d4c92472c1000faef86e464d1786","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:54:23.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a9712ac881a432c850d803284b53751\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aac0fd1ff6104821b790614294d8afe1 : SUCCESS in 11m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/c5040f806f7c499dad52bf19d01967cc : SUCCESS in 58m 10s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/803e746088004ebbae592bc8926b46f4 : SUCCESS in 34m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2d93729268d54ac69814b010f4478bfa : FAILURE in 27m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aad8d3f50c41439780ed67107139ff66 : SUCCESS in 1h 03m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1119e2a290a74e8e9bcbb13ca3c3ab80 : SUCCESS in 10m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/101d064943ea4911b1964f6440a976bf : SUCCESS in 20m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2c48b174150e48679f90ff630f239d58 : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d2e99d0f06f1401f865574e80540c9b0 : SUCCESS in 7m 06s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ca163ed2f5ff43a48aede05a05c7bf5a : SUCCESS in 24m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a41d6100866e4a1bb92d6a036b19104b : SUCCESS in 14m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/75bbb55bbaef41309490279da4f849ff : SUCCESS in 27m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb76e32ec9284bac9b575c686dc8d1dd : SUCCESS in 11m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f9752d16daf6453da919add2b3505a8f : SUCCESS in 14m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f8dc3c9d99649aab084ce95356c1618 : SUCCESS in 14m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/741f0a76044e453d99827f739a83201c : SUCCESS in 16m 27s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/afbd336ebb7247939513016455061d83 : SUCCESS in 9m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/98391f4b5fe647c58e7efcc67c4c45a5 : SUCCESS in 14m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5db1774347e1426983753add316ec4d8 : SUCCESS in 8m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3ff30a65cbc04777b0f90faa99e2d760 : SUCCESS in 36m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/15453abf382b4d24acf5c98ca68a5053 : SUCCESS in 32m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/64420763de874bfaa94f0bd8266df098 : SUCCESS in 50m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/396172a49f424cb3adc5ae2a7bbcda27 : SUCCESS in 24m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/233015d84a6948eb9cd956366812f165 : SUCCESS in 3m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2b0a0a9d31d64536b5dc68964ba81848 : SUCCESS in 18m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c53ad8fb0a9445dc9f04240ad62cf09c : SUCCESS in 1h 04m 45s","accounts_in_message":[],"_revision_number":17},{"id":"5f253890ce915d4dd0c88523de9a6f3f878ed560","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-10 14:34:29.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"2fd1be80d7fd2c5b5b81a2c41b4399fc746a87fd","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-10 15:06:20.000000000","message":"Patch Set 17:\n\n(4 comments)","accounts_in_message":[],"_revision_number":17},{"id":"c5bfb9692f6f00f48754f4650c9640a6c473358e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:07:15.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76b54589ebf647e38e312d6a06a8fe92\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fea2d941aca04f8393bc4a31a2de002d : SUCCESS in 29m 19s","accounts_in_message":[],"_revision_number":17},{"id":"713a62d38a4039547cef59deea4f83ce9734e1c1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-10 15:51:19.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2773d4b56454cc1a2932d2e02a37b98\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15281cc822a14c42a7f19897d34e60d8 : SUCCESS in 8m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/41e679cbfc784b0292a54e1ed50dbb96 : SUCCESS in 37m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6e3331e4de2e462f9cf503176b75feb9 : SUCCESS in 1h 06m 07s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/96ffb01e4ef44b4cbcb3ae25fbf292f6 : SUCCESS in 42m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/38bcad30813748b39bb8ddfc7f0a51b1 : SUCCESS in 40m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2fa61ee0d5484ed9af02c0fec6829c9c : SUCCESS in 18m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2989756d9ba34d5592f4ee717554154a : SUCCESS in 11m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a03870e29b23484791d1c2e1028f7f1c : SUCCESS in 13m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/94850c31be3441bdb14e54ddf34c9d29 : SUCCESS in 7m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/571f1d5150f84faaac19706ed38ffccd : SUCCESS in 17m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6071c58327a444d2a0a35ea620c82be4 : SUCCESS in 9m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12b8108691d7443c986904a24fe23640 : SUCCESS in 20m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e983d2bd359e4ea0befb4b26a105ca1c : SUCCESS in 30m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6a398dd8c6d547e081eff8f26662a8db : SUCCESS in 14m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6068e6f637c448e68a8af6254efb5f6f : SUCCESS in 15m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/33ce417564c341f2a13d2a013f79f063 : SUCCESS in 14m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/be1fbd67548c476db08ac429eba2baf0 : SUCCESS in 6m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/32b1ef13d62c423588447bfbe97a8be3 : SUCCESS in 15m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5b2119bf1b214b978c9903f7e1cba88f : SUCCESS in 7m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4a4793bf90d3428aaa1fa1861bbe20d3 : SUCCESS in 55m 18s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a58b582183034169833d5ca8fcefaab5 : SUCCESS in 55m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c742d846223d49a88ff30cc3a834cbdb : SUCCESS in 51m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1b53495bcab24595b89a0ead89ff6acf : SUCCESS in 28m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da529cacc87342cca64c14ef829d3191 : SUCCESS in 4m 23s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7524450b090e4ec69850e1b44758b42e : SUCCESS in 18m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2481ae5ad15e48b7a18abe0374e831df : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":17},{"id":"28a2f9377e1497f42efe1cae034f00530df68a3f","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-10 20:23:06.000000000","message":"Patch Set 17:\n\n(8 comments)","accounts_in_message":[],"_revision_number":17},{"id":"09cab97e43fe983a703a642d303b5ef9d9b900f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 13:35:59.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"f509b26c494a3d1a4dee223369aba9658ff4ce24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 15:06:04.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e860fd719724d22ba0b16efbb6c9e35\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c27ba959e9ba4b18b807d58a74b1ddca : SUCCESS in 6m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/23ceb10aa7e24f09bf61d63fe70a99ba : SUCCESS in 1h 14m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/578b1ee0cea848979f6dbc85bb752b87 : SUCCESS in 1h 05m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/32c18366d17f49d9a73e0728dce782c7 : SUCCESS in 1h 06m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2deb3314b8f447b4ac878a5fdbd019f8 : SUCCESS in 1h 07m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/64b1ab3261c1447a954b49e889a61588 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7a325b617c44ae898349be25846f790 : SUCCESS in 25m 05s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f46d9f253724d3b8c168439d074ddfe : SUCCESS in 13m 11s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/d862f768191d4f8a8ce794148178fb75 : FAILURE in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d34c43d51f734f4eb706fcf747af8db6 : SUCCESS in 25m 42s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/850223f2a8a84cdc88a908bff007a396 : SUCCESS in 20m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a4d1a1d0420f4c4e8913a178beb80eae : SUCCESS in 29m 23s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c17a12a09bfe4c8dbcda5e9456e3995f : SUCCESS in 21m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7d50a38d7bbc407288ecb8a00ea1ee33 : SUCCESS in 13m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0961064c8c3b45258f1f6276c7fa5d62 : SUCCESS in 16m 57s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b748416af5a24c39a24e277770e516f8 : SUCCESS in 16m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3ba4df2feebd4eb393b59b91b796aa9b : SUCCESS in 7m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/0938cc3b1d1a4f799addb98bcc92d364 : SUCCESS in 18m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1e794e1058924702bdeb8f2d390524f3 : SUCCESS in 8m 42s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7321737f0f334550a2f6acf4fb6c0c99 : SUCCESS in 1h 01m 49s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3bcd2eb361944c8cbafdbbf60892fd7e : SUCCESS in 53m 48s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b37615eccfd24cbc98b0abd345d8bd02 : SUCCESS in 48m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5fca0bb6aafb4f1aaf2c8e7c77b57fd7 : SUCCESS in 27m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465d3c6587bd4b1ba614a00cd257adff : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c772d372784844a7b62322cd828fcc4d : SUCCESS in 19m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74319a8f48b843458b02faa553a6d1a6 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":18},{"id":"c25abd79c6ecad415bf0a8b99777b226ffb37c59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 15:18:47.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"b38cbfd81977806da8f1bc5bf686283bf21c8f9e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:29:10.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8cdee5a6ae7451c99b2fc59434faf53\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5e58334fe48402887cfe8e4bf96c139 : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/8cc8de4aa49544548ab1f28f1e0a2066 : SUCCESS in 40m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/df810185f0424af9b94973c68618227f : SUCCESS in 59m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e19a68f9027b42a98ae9d6a28830a8b1 : SUCCESS in 35m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bad7001e0b184ee398893ab53569a43d : SUCCESS in 1h 00m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/786c1d36cb54442794a0e1d0a6c94ecb : SUCCESS in 14m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6f406af450c7461397bb00884fe89840 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3b1d7fde93d94fba925ff426f00e8e7a : SUCCESS in 20m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f157a47f5668471793711dcfb597e91a : SUCCESS in 10m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/466d571208b54515b4ce95ff96b932f0 : SUCCESS in 28m 55s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2c8227319adb4c1fb8add8d061eb3e07 : SUCCESS in 25m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/47672edaee3e4e1a9d8c335591bddfab : SUCCESS in 18m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/85cf9c45c7284a208fbccb4251c508f9 : SUCCESS in 14m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e60d0c1fc47a4893aa89c52c273b5baf : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cd8d22856d0b450db69094ce12505233 : SUCCESS in 17m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f3b2e9362f074ca597742405f7d33467 : SUCCESS in 9m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3dd81a0bab7046648566313bb8c9d2a2 : SUCCESS in 6m 34s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7421e288da23448ba0b502f2f743c295 : SUCCESS in 16m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c885937a98854744ba5433d06d5d4448 : SUCCESS in 6m 22s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c48b7abef994f2d8119c473ffede776 : SUCCESS in 35m 47s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d2a1807e886d4a9ca4722a2e6ae80f50 : SUCCESS in 24m 22s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b0242b86a8a24065affbb0e81002edaa : SUCCESS in 51m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/31e2482e1e734abfbc2c40cb9bc0ca7b : TIMED_OUT in 41m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdf0fd36de874eb7a86fcbe79e0fa8bc : SUCCESS in 3m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d2bdd7b5b3cd4896b37b67abe2e38613 : SUCCESS in 19m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/14c5e5a491664dc58e9207d2ee8807b1 : SUCCESS in 53m 23s","accounts_in_message":[],"_revision_number":19},{"id":"467a46f13060e336d49cbe3070fd4f5ad2b42ab2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:26:52.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60005357e75f491a86f1bf0bbeb7863f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/037b91fbff644a4ea35797df24b28b14 : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":19},{"id":"44f3b3aca993723d9c7cbc031b1f274af3fa5864","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 17:58:55.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"1f4e65bca4fcc1dfe70884b8699e765bdce7c4b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:13:03.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/283ff17419e44a08913520b1921a5602\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0c0b864709141ab83e78c0592d11f9d : SUCCESS in 9m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/7b3e8b230e8c422f87cf53e1a71b98c5 : SUCCESS in 34m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7e3055c8e7ff46adaca77c895106350c : SUCCESS in 45m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b276cc85831f472d875293660ab22fb0 : SUCCESS in 37m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0ab464148fc24503abe6976e953a1eb0 : SUCCESS in 1h 00m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6b38d915f66e433abe699a316e7590ed : SUCCESS in 15m 18s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acb284e62f494495a70d9c11c7d4ae4b : SUCCESS in 12m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c34315f6055342f4874e1524eba68422 : SUCCESS in 11m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2e68a332c7344bcd95a496018a7b4f89 : SUCCESS in 7m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c4bd81a15f0d40e4b33239be78b35b41 : SUCCESS in 11m 40s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/38b5858ff358449589ff47dcb3191fbe : SUCCESS in 10m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f41232962e2f4f16aa38c95891909156 : SUCCESS in 21m 10s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/44eefa2d321746aeac6c4938af063d97 : SUCCESS in 15m 40s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5a1c1181c974494ea255b57c5967cd2f : SUCCESS in 13m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c3054231e4b34f86aa07a0fa655b95ab : SUCCESS in 16m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2759512434fb44ef9803f4fac4552414 : SUCCESS in 16m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ccafdd1532144cd892271e37cc1e0f8d : SUCCESS in 7m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5bd8aa92730a4b36958dd823e2e722ee : SUCCESS in 19m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/dcb053f5f6584e7a8411ba09777f2392 : SUCCESS in 9m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9aff6dd063694c3dbf0ca23c8e3f762c : SUCCESS in 1h 05m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/df928852db9049ba9b3bf6a78b024e52 : SUCCESS in 25m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0d515d2f08f6455fa552eead0d36ba10 : SUCCESS in 49m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b5a04ca03784eccbd97205223ba0535 : SUCCESS in 22m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/35ab33061a564a1da1492088a11aa848 : SUCCESS in 4m 27s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0cdb551db4774427b466c714186cb849 : SUCCESS in 17m 11s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7f36356228864f77beef5e2f2fb09d52 : SUCCESS in 50m 30s","accounts_in_message":[],"_revision_number":20},{"id":"f3f5db6fe8fda99213866de476d1ab9a4adb40b0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:55:12.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4edc443cbb884bbe9bd6e27219b97ef7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2122cc1abb714235944fa382c777e0b3 : SUCCESS in 34m 34s","accounts_in_message":[],"_revision_number":20},{"id":"a62a87a4f70d7eea198fe3fe6e944e8ccd49579b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 16:30:13.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"be8c56604b0fdca07400f01ecb067d4a96f2145a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:42:34.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f187e31a310c4f13bf193c63f7d99030\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1beef3019ff47e9accf6c49b8ff05bb : SUCCESS in 11m 15s\n- grenade https://zuul.opendev.org/t/openstack/build/7afe22c3e2224d5282292f5471e9b8ac : SUCCESS in 1h 02m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0d1dcd8e03c14211bc69cf551f28dde1 : SUCCESS in 33m 18s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/b433e1703701455c884684cba710ecab : SUCCESS in 34m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/74d69cb1519b42aba81881b8f04b4f75 : SUCCESS in 36m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/203386525c1d45ec9c64ddc7c547a63c : SUCCESS in 20m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/06ebc3ecc63b416ea250c270d7e4ae22 : SUCCESS in 21m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7c60ad664dc4732bd74f8734a019a5a : SUCCESS in 21m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/280bbdbe7b6b415ab27a631137b63375 : SUCCESS in 9m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5ae40f97db32420b8d65d06c6edee435 : SUCCESS in 12m 26s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/a862d71ad5dd4d369da34840492b9f74 : SUCCESS in 19m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b9c0a4790fc346ac87749da73d9e6ca4 : SUCCESS in 25m 05s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/168b4e59920d42db94463d776802e151 : SUCCESS in 10m 14s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ffe80b3cbc904723b48cf6b4614d1eaa : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dfdbb73f7833475f9e0578cc18b3ba47 : SUCCESS in 8m 59s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b044dfa306db439b901fc1461f5e8b0a : SUCCESS in 17m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f660d8adf176462dbe1337acf4a4209e : SUCCESS in 7m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c264ee6c95f84e29b04c0e85195cf9af : SUCCESS in 16m 05s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c411e50cb2ed4db6a25271547e0ebfa9 : SUCCESS in 9m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6ab71152c0c5497ba1eba9ec63cb8b0e : SUCCESS in 53m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e490d80621f9468e9562accd4bbd592e : SUCCESS in 38m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f38d8b7f25114ba2babfa41946e47526 : SUCCESS in 48m 19s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/da95f224c78a41d1855486749c504773 : SUCCESS in 30m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13308524ae694f5ca91d0437748a126c : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3b0eb473b8c041c29a1816d1945d9724 : SUCCESS in 10m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/908670bca9c24267baec5d76e74e981b : SUCCESS in 1h 04m 14s","accounts_in_message":[],"_revision_number":21},{"id":"c482a8719d28781e6b3cc3d9b46e71b4b3f05eb6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 19:31:47.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a55ef51f2ea8487dafcdf50f8172f201\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2786fb8858ed4503a39489bfd46e4149 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":21},{"id":"fb67a26f468051854247ac578eb1b92c32a157c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 14:54:16.000000000","message":"Uploaded patch set 22: Patch Set 21 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"3dd556e38175380469376e63b626aac1bd08be09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:19:57.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc34f7ee229c47f7be463d7afff3f0b8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c50a8c7ab3954bc4ac97d771e7715f62 : SUCCESS in 6m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/c80ea7a30714456b8ee6e5bd40ee98a8 : SUCCESS in 1h 12m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2931dc779b554722aa0cafc649e59ffe : SUCCESS in 1h 06m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1e7cef378bd44c59bbdb6d310db9f8d6 : SUCCESS in 49m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd0b38f257334d6fa0832612af71a900 : SUCCESS in 1h 02m 01s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9cda91e924ac4a6d8ace829dd8f28649 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e2162097d00241b0902b62b7faf95e64 : SUCCESS in 17m 41s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/326ad9e54245413a85e58c8d8fb9db0f : SUCCESS in 18m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38f0f2d36ef84ddbb559dc627c0b0531 : SUCCESS in 7m 03s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e059325bf8cb4e88bde2ba1173e676a7 : SUCCESS in 24m 50s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/17b772fce7ef4f01bc1bca90d588f9d2 : SUCCESS in 18m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/af183fb6e7bb49c3af882aedde24961a : SUCCESS in 18m 07s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/65c92730bbba486f83ec357606051ce9 : SUCCESS in 19m 51s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fea66493c6ec44e48d681ab0cb2272a9 : SUCCESS in 13m 42s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/200640c617f8404989f47970c0b92082 : SUCCESS in 12m 44s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a0c6c89fa83c40848929471c424de190 : SUCCESS in 17m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b34b8e342c0342a98858b394823d317c : SUCCESS in 6m 16s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/153bb18e1b514b1dafd54ea2b02461d4 : SUCCESS in 17m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7782fd876dd04c60bc93f5a40f5594be : SUCCESS in 9m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/135541dda41c41a7a851434ff4cd450d : SUCCESS in 36m 46s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3206063428b445f69181cf3382dc5932 : SUCCESS in 58m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/536f849880c140aabfeb27855cdad83e : SUCCESS in 41m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bc640c1ed354f7db8b87406f234f8ab : SUCCESS in 30m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c680e8110e0e4d63b886f97de35a056c : SUCCESS in 6m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ec49f656af9540468b3b52a0454d71d4 : SUCCESS in 25m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d133c3f1e2dd452a8bb2e1629d37f7dd : SUCCESS in 55m 30s","accounts_in_message":[],"_revision_number":22},{"id":"9a9ac590ef6401ab9f4f1a48c2452cfe277e811e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:17:37.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3aeff392c7641c69345d9b3ea721e5e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2fdcfaee09304db29846f57d33189838 : SUCCESS in 35m 20s","accounts_in_message":[],"_revision_number":22},{"id":"f02f84b3313fa13c5ea8c7062e89bc7ae54013d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-09 09:21:56.000000000","message":"Uploaded patch set 23: Patch Set 22 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":23},{"id":"678b5d1e1a69591a4c9ed86782187ce5b4bd05a4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:48:23.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07796d47cb3c44ac8b4791c8adc51d95\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0149b395150e4ce9a3c23b0f4a5162ff : SUCCESS in 5m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/3d4cc4d51dbf4a9ba7f32818f2193809 : SUCCESS in 1h 07m 35s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6ddd0ce3c49d4b5a9d1e05d2cb1616ab : SUCCESS in 1h 10m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ad9689ed788d42cbbc5d2c52c3967038 : SUCCESS in 53m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d027e1cd7b2b4bf18f0f883c06936968 : SUCCESS in 1h 15m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/20323c6f83bb42a8aa333972c35abe34 : SUCCESS in 20m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/75582fc83fe241afa9062cb07eeb4b86 : SUCCESS in 20m 39s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3c38788f73494061aa2c67fdca311ccd : SUCCESS in 23m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/981843cdbbcf4d60b821e26c9e878de5 : SUCCESS in 7m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f04fb5df07aa4a7295fb95768457237e : SUCCESS in 14m 38s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/18db107fcbfa4c6ca6689d586b63f208 : SUCCESS in 9m 42s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/de80ab126747465abd0fb4b2f0f92190 : SUCCESS in 32m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c89b5719025f40adb131b5ebc420bb91 : SUCCESS in 20m 33s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fdbba275bc0544e68884592983c25e2d : SUCCESS in 10m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0be81e50a2324b7187d68f8896d9265a : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f85b41411ac6459286cbe5fe40e736be : SUCCESS in 15m 45s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e559c10849e4d81afe669bdf52b6532 : SUCCESS in 4m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bc7efca0ae39408aaecc3808badf9efd : SUCCESS in 10m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e5536626fbf948c5b1c42fd85eb9fd91 : SUCCESS in 6m 46s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2cf3cb1891ed4590b980d3be00a69cf8 : SUCCESS in 1h 04m 39s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/350c9555dc8341198e428911f5ba3cbc : SUCCESS in 34m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d722962736a44b02a9ddee6aea63d7ab : SUCCESS in 41m 50s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/520dc7a05ade46fbbd5e9f5b9bbffa07 : SUCCESS in 21m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e5745bc6127c40ecacf49f2b4e799544 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fcb757f555644cedbf59bd2459c125fa : SUCCESS in 9m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1f078598033a48379791b7eeef06ef68 : SUCCESS in 57m 32s","accounts_in_message":[],"_revision_number":23},{"id":"e6ce54b7351167f1edb5d8e9a51a81526c8691c9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 11:41:59.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a20dd0ddf9f341a981811f54c8b92646\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e97ae0c6cf1d4bb0914fdef244dae492 : SUCCESS in 34m 33s","accounts_in_message":[],"_revision_number":23},{"id":"289a101350a18a8d04841ec5ee66b27f8316e73b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 09:47:13.000000000","message":"Uploaded patch set 24: Patch Set 23 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"ed8c2b729259c799fcece2d6bafcda66fbc1afb3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:34:02.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c923ca02ba474e9e94dc2d466666e4f8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4610601a639849938a426563cfbdad86 : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":24},{"id":"0ddf7625c020f241c1e79a5c360cd34f55c15ecb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 11:49:23.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d4719cdafed4c8fbb82299eeb5be98a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a75bba62f942477aad67f10112e5fbba : SUCCESS in 10m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/3e68840845664d96a2da23eb48939eaa : SUCCESS in 1h 01m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c563a265d38404a9f31b2e78bac8e6a : SUCCESS in 1h 11m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c7f7c1b3c9404298b7897f28daa7a375 : SUCCESS in 47m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/466595d3d0c04939996980b3a110557f : SUCCESS in 1h 06m 44s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/806f762eb3684762994f4a860314a498 : SUCCESS in 11m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a65287606ec447db9d41ef8248573b7f : SUCCESS in 26m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39d6b13a56b64ef98e9d32290d43a34b : SUCCESS in 19m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/757e235f84de4b1ba9893461bac7995f : SUCCESS in 7m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/54d41f4c06674fc4b0585aec8f661fbf : SUCCESS in 19m 24s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/38e850da993c41dbb19218b01e5a1257 : SUCCESS in 22m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/a594933f4836401e8fcda5372fd52c11 : SUCCESS in 29m 27s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/9e80edcfcac44655bc0c6a8780a29159 : SUCCESS in 21m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/673c37dd01b0473498582053b5452150 : SUCCESS in 16m 01s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9811a98627eb432aa4a195f95ef028f1 : SUCCESS in 17m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/60197e3ebe294a9b9dc7dd2aa7f0c56b : SUCCESS in 17m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/56d42beddfbe41b3ab54edb18574ac27 : SUCCESS in 6m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/42c37a156c5a4630a6d18bb5b2c7dc53 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/61224682f38148b4bf5570a69c3ad04a : SUCCESS in 8m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/90b2728f25dc471f9c1b9587a140b115 : SUCCESS in 49m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b29f489941d64b8d8f43fd2dcd2b3da6 : SUCCESS in 38m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f0805f0ed6944542b34ff4d59a743f45 : SUCCESS in 52m 32s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b18e3a9ce6ad415594027e99458c2d06 : SUCCESS in 33m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eecc1b050e714fe58d216f51e00e59c3 : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/3cb1a890b52f459687e43ba88957ace2 : SUCCESS in 17m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0bdfe81a66f249c8a4c3ea02b9c96407 : SUCCESS in 1h 06m 21s","accounts_in_message":[],"_revision_number":24},{"id":"868f8692396977134eb885384d5068931da2b053","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-14 15:09:20.000000000","message":"Uploaded patch set 25: Patch Set 24 was rebased.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"08cf60f3b117daf67d3ca58ff049ca9a76dd5a4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:27:54.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/334502013edb4ab4802c02755f000a4c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5e11ae2fa6a436eacb762ee9f5cb0af : SUCCESS in 11m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/c30d5ab280d049b4982031e2ac2e6f95 : SUCCESS in 1h 01m 12s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/869eb4c5bc714e8c9e7dc06578db62c5 : SUCCESS in 1h 04m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/888e3b67d28d4c8988a9964a496192fb : SUCCESS in 1h 05m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ade3ca099d1041149a7a8ca1add1331b : SUCCESS in 1h 08m 08s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94c60defbdc040749e6daf6d8fd4a5f3 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/59e621b6633749308936f177b14c08a7 : SUCCESS in 20m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2d115668a54a488a8c1aedfd4a75bcda : SUCCESS in 11m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aa84252214274f3fa3e4629baefcb512 : SUCCESS in 9m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/337d73f050cd45abac26c43849030ca1 : SUCCESS in 26m 33s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3a4d08e306564dc38d08a824775b0531 : SUCCESS in 27m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/05bec038af124c8c9050f66659d7e9f0 : SUCCESS in 29m 32s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/ccfb0f9166d848b68fb62b8a72fc1b65 : SUCCESS in 22m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/53257602a5e24240bdb8389ebd585000 : SUCCESS in 9m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/69e1a38c0ce64045994699d69b934c2c : SUCCESS in 15m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ec9a2be7f78344b1a3606d5cc0dcc5b0 : SUCCESS in 17m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8f72f2d7d8a242c28d7dbef5e031f59a : SUCCESS in 7m 02s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8b8b3af1f69d4413b925296d7e8e59ca : SUCCESS in 19m 10s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/678afe47643b4fb2ba427587ad44d8d3 : SUCCESS in 11m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/eabfbe56bb7b498a9e19ccbd7fbf7190 : SUCCESS in 1h 01m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/afb65667952a47d0be47b6c37d2d736a : SUCCESS in 1h 01m 40s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a09f8846f1274a89916e4e433f0699d4 : SUCCESS in 57m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e28deed0319d4d62941be1e34edd2a2a : SUCCESS in 33m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7cf45a4c157458a89d7c8f23853f94b : SUCCESS in 4m 42s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e025f74611d34c0d8b881e43446e0f1f : SUCCESS in 20m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0848006b028b4b2fb8424117bd100deb : SUCCESS in 50m 08s","accounts_in_message":[],"_revision_number":25},{"id":"70ab906d5d0a29c77e6a41697e8a5db7e11ad196","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 20:43:23.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17b9efc21f8040d3ab9f80afecd4d5a8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/23e62e58486f4d03ad17b6480c221b98 : SUCCESS in 33m 11s","accounts_in_message":[],"_revision_number":25},{"id":"65ef969682056fab34bdfbd529aba81035ae22d4","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-04-15 17:00:31.000000000","message":"Patch Set 25: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":25},{"id":"a54d2fdad81cdeb78e853c82fd6968eaf536e6b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-06 10:50:40.000000000","message":"Uploaded patch set 26: Patch Set 25 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"c349aee7deb3f6aaa240f23b404d134103787693","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:47:08.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f67a5a8b21f249d1a710a8c7f889950b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1093088a386e477b9b6c86d87468bf52 : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":26},{"id":"642fa2519e5e92c01873817e0652f041ce739759","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:04:18.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d272764833842739a616879d16e663a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df69776bb3b34d878a1c5e5f55673493 : SUCCESS in 5m 45s\n- grenade https://zuul.opendev.org/t/openstack/build/c1a44a1490c642288b08ef876f7cb6db : SUCCESS in 30m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b524921af921459497e5e37ace5ab08e : SUCCESS in 1h 05m 03s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afffb8f2906f4c7d9f0f5d01534f61ab : SUCCESS in 59m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/61983472973c440a8a8f519efbd9a73c : SUCCESS in 34m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c78cead7382d47dc90521a4504270dfc : SUCCESS in 21m 14s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8355939138b47f9ae2c1a97583212dc : SUCCESS in 11m 28s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5f6fe492b7034cc9b316ed80311572fd : SUCCESS in 21m 15s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c5dae5afe0f94d62b198bb6ac4488d2c : SUCCESS in 6m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8b5a4b1b5bb4cbdafce3c1c9b70cc66 : SUCCESS in 26m 06s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/2a03ace447e34055bb62d979e04201db : SUCCESS in 9m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/dd063c3990d345909e368ca9ddbbff16 : SUCCESS in 20m 40s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/c121a1aaa7ba470eb52739f257ecfe0f : SUCCESS in 20m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4ad05112d4ad421397889902c5b98d97 : SUCCESS in 8m 28s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1238415e27d749218002286a49ef6475 : SUCCESS in 15m 08s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c1bcebb9d78b48559a253659f5e03f4f : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b580a319becd41dba966af7bfe2d6ab9 : SUCCESS in 7m 18s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ed5594e72ae4889861ec20ff87abe4e : SUCCESS in 21m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/d6ffb8eb23664469b9d21d85222f7539 : SUCCESS in 12m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/de625e9a050c43a6ab6c33fbef74b086 : SUCCESS in 52m 44s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3e9a97f546244bc6af00d0fd51acdab2 : SUCCESS in 50m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0272f61a79c248b18cf9f9a0af1952f4 : SUCCESS in 56m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fc4a3e9aab604cca80e9ac8974dccab1 : SUCCESS in 28m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b812bc42e36402a93bd0f841785e8e5 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cbb0dc3174524d0e9687b1fc3fa6d348 : SUCCESS in 13m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eae0eda3ce284ddba1e91f2ae8ae541a : SUCCESS in 1h 03m 27s","accounts_in_message":[],"_revision_number":26},{"id":"74225bda0fde89c6dc3fb1d8d97b73e0de3fe214","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-25 10:15:30.000000000","message":"Patch Set 26: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":26},{"id":"340265217e39851a2300ec0e5615f8cede4f8c35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 10:16:50.000000000","message":"Patch Set 26: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":26},{"id":"e43328481daf098ff440ef74b77ea132cad24fcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:58:19.000000000","message":"Patch Set 26: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2175c403fb48c3a135cf53d6b3b3a3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cca043e6f174c7c802dcd649988fd5d : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2b56b7c6b19d4f6b918e9400ed298c7d : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/22f67187a48b43b08564b9992279de70 : SUCCESS in 55m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4c240f6f8ed04091aec7b25f6cada9c1 : SUCCESS in 1h 12m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d7e9be0a54df44d1b3c75595a7b43dcb : SUCCESS in 49m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/17080f2f042b4b819ddd07387ccbba3d : SUCCESS in 1h 11m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/588f0c28d6334e06af76570dc7b35e93 : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/88fb3ee2912348cdadb3181498f95a20 : SUCCESS in 24m 59s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/f7baecba222346dcb9a7aa5a0512b2da : SUCCESS in 21m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ddfe54d1c6994feb85f45466f25d47e7 : SUCCESS in 22m 24s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7efd3d49576347f39398fb7357aea965 : SUCCESS in 24m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6681212778c9459a8f7aa7e1ffbbb454 : SUCCESS in 15m 56s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ec7c8fa38afb488a80458af78297cbde : SUCCESS in 17m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/abafeecea03d4b2ba09e94786085af2b : SUCCESS in 14m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/641bd0f46c5143de891a0d9849733b3c : SUCCESS in 8m 24s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/8f47b2856b31428bab55ed5f26120e58 : SUCCESS in 9m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/df55b1edd6094a5aa296c2db832be34d : SUCCESS in 1h 02m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d8a0c1bf2ed440869fda659ea3c60c89 : SUCCESS in 58m 16s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/364737247fc548118cbf9327513ad95c : SUCCESS in 52m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/007f5a9de00e49e6a088495404d5b788 : SUCCESS in 28m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9e2f7f51e741480f9377c5292db75575 : SUCCESS in 4m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c953333fd85f4d30b7efb81d6b3e9d27 : SUCCESS in 19m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/efc15f2402b041f299d417df47f16591 : SUCCESS in 1h 01m 55s","accounts_in_message":[],"_revision_number":26},{"id":"31fc64c92f3b882fb43b4416fd410f44b6576a14","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:58:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":26}],"current_revision_number":26,"current_revision":"81fb83459a1ad9fb8788fa3069774b95feb95f2e","revisions":{"feac67b684294d9c347512e52a0c0105304190ce":{"kind":"REWORK","_number":1,"created":"2025-05-29 01:09:42.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/1"}}},"commit":{"parents":[{"commit":"c04061894f81ad52e6c210ba2f21bf773c23cec3","subject":"replicator: Add sync_batches_per_revert option","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c04061894f81ad52e6c210ba2f21bf773c23cec3"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-05-29 00:58:39.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-05-29 01:06:35.000000000","tz":-420},"subject":"Object: log breakdown timings for DELETE and REPLICATE","message":"Object: log breakdown timings for DELETE and REPLICATE\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/feac67b684294d9c347512e52a0c0105304190ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/feac67b684294d9c347512e52a0c0105304190ce"}]},"branch":"refs/heads/master"},"6d46fafa5ded9edcbe106ebf6a166ff4ad8db0ea":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-29 01:12:04.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/2"}}},"commit":{"parents":[{"commit":"d0e97650f3d1e832135c3b220d0928f5cd986501","subject":"Merge \"CI: run S3 cross-compat tests as part of DSVM jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d0e97650f3d1e832135c3b220d0928f5cd986501"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-05-29 00:58:39.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-05-29 01:11:55.000000000","tz":-420},"subject":"Object: log breakdown timings for DELETE and REPLICATE","message":"Object: log breakdown timings for DELETE and REPLICATE\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d46fafa5ded9edcbe106ebf6a166ff4ad8db0ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6d46fafa5ded9edcbe106ebf6a166ff4ad8db0ea"}]},"branch":"refs/heads/master"},"5f7f686aed04fd162ac2c3e704b554003844365c":{"kind":"REWORK","_number":3,"created":"2025-06-10 15:08:36.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/3"}}},"commit":{"parents":[{"commit":"366916b21f209096238e3b77e0bb9f511d426026","subject":"Common: add sub-operation timing breakdown in labeled_timing_stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/366916b21f209096238e3b77e0bb9f511d426026"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:51.000000000","tz":-420},"subject":"Object: log breakdown timings for DELETE and REPLICATE","message":"Object: log breakdown timings for DELETE and REPLICATE\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f7f686aed04fd162ac2c3e704b554003844365c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f7f686aed04fd162ac2c3e704b554003844365c"}]},"branch":"refs/heads/master"},"2eb4360a18a4316560db51a5df2812a088111821":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-06-10 17:40:47.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/4"}}},"commit":{"parents":[{"commit":"3da1dd9523965d934f1f1f32455624de31e2a58a","subject":"Common: add sub-operation timing breakdown in labeled_timing_stats","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3da1dd9523965d934f1f1f32455624de31e2a58a"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 17:40:37.000000000","tz":-420},"subject":"Object: log breakdown timings for DELETE and REPLICATE","message":"Object: log breakdown timings for DELETE and REPLICATE\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2eb4360a18a4316560db51a5df2812a088111821"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2eb4360a18a4316560db51a5df2812a088111821"}]},"branch":"refs/heads/master"},"70d1c087ee9fb6bcb6e78af00c9a95a2002ea90f":{"kind":"REWORK","_number":5,"created":"2025-06-12 20:07:25.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/5"}}},"commit":{"parents":[{"commit":"155bc4c58848538ec6c4458204f915e072ce2f67","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155bc4c58848538ec6c4458204f915e072ce2f67"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-12 20:07:13.000000000","tz":-420},"subject":"Object: log breakdown timings for DELETE and REPLICATE","message":"Object: log breakdown timings for DELETE and REPLICATE\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70d1c087ee9fb6bcb6e78af00c9a95a2002ea90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/70d1c087ee9fb6bcb6e78af00c9a95a2002ea90f"}]},"branch":"refs/heads/master"},"82696cfc17d4896ab6f28ddb1530791e186b07e9":{"kind":"REWORK","_number":6,"created":"2025-07-02 21:03:41.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/6"}}},"commit":{"parents":[{"commit":"80f8b5418bbdda3cd00d0c7161edeb02581d0e28","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80f8b5418bbdda3cd00d0c7161edeb02581d0e28"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-02 21:02:36.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82696cfc17d4896ab6f28ddb1530791e186b07e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82696cfc17d4896ab6f28ddb1530791e186b07e9"}]},"branch":"refs/heads/master"},"45dcc34f99c9c1d650cc554465defe9fb959af87":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-04 01:31:37.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/7"}}},"commit":{"parents":[{"commit":"00fd734b25bb5eec1653b86df1c606a20221efc4","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00fd734b25bb5eec1653b86df1c606a20221efc4"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-03 19:21:07.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45dcc34f99c9c1d650cc554465defe9fb959af87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/45dcc34f99c9c1d650cc554465defe9fb959af87"}]},"branch":"refs/heads/master"},"a0d4c193a5275d3925895795ffc2769a52b91b5f":{"kind":"REWORK","_number":8,"created":"2025-07-14 06:22:47.000000000","uploader":{"_account_id":22042,"name":"Chris Smart","email":"distroguy@gmail.com","username":"csmart"},"ref":"refs/changes/19/951219/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/8"}}},"commit":{"parents":[{"commit":"00fd734b25bb5eec1653b86df1c606a20221efc4","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00fd734b25bb5eec1653b86df1c606a20221efc4"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Chris Smart","email":"csmart@nvidia.com","date":"2025-07-14 06:22:19.000000000","tz":600},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0d4c193a5275d3925895795ffc2769a52b91b5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a0d4c193a5275d3925895795ffc2769a52b91b5f"}]},"branch":"refs/heads/master"},"c81f47c5d4c810bd9a08ed27b1cf98b807782546":{"kind":"REWORK","_number":9,"created":"2025-07-14 18:58:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/9"}}},"commit":{"parents":[{"commit":"f368f48bb1d4b9084e72f67ce5c9a02d05589981","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f368f48bb1d4b9084e72f67ce5c9a02d05589981"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-14 18:58:17.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c81f47c5d4c810bd9a08ed27b1cf98b807782546"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c81f47c5d4c810bd9a08ed27b1cf98b807782546"}]},"branch":"refs/heads/master"},"848aed8668a3bae0e080e0997c541a73bd485cc7":{"kind":"REWORK","_number":10,"created":"2025-07-21 23:58:36.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/10"}}},"commit":{"parents":[{"commit":"f368f48bb1d4b9084e72f67ce5c9a02d05589981","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f368f48bb1d4b9084e72f67ce5c9a02d05589981"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-21 23:58:21.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/848aed8668a3bae0e080e0997c541a73bd485cc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/848aed8668a3bae0e080e0997c541a73bd485cc7"}]},"branch":"refs/heads/master"},"4198a2a73f18c33be7bf6c5124ae1e6e106c46be":{"kind":"REWORK","_number":11,"created":"2025-08-20 05:28:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/11"}}},"commit":{"parents":[{"commit":"92542e7c31003021aa49faed40ebe29c32a33ed7","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92542e7c31003021aa49faed40ebe29c32a33ed7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-20 05:18:31.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4198a2a73f18c33be7bf6c5124ae1e6e106c46be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4198a2a73f18c33be7bf6c5124ae1e6e106c46be"}]},"branch":"refs/heads/master"},"1bd8db907d3985849e2de06e54dfb0d9d33c6652":{"kind":"REWORK","_number":12,"created":"2025-08-20 05:39:09.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/12"}}},"commit":{"parents":[{"commit":"92542e7c31003021aa49faed40ebe29c32a33ed7","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92542e7c31003021aa49faed40ebe29c32a33ed7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-20 05:36:07.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bd8db907d3985849e2de06e54dfb0d9d33c6652"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1bd8db907d3985849e2de06e54dfb0d9d33c6652"}]},"branch":"refs/heads/master"},"e732e0c46836e49b0b4959cc488a901227fa823a":{"kind":"REWORK","_number":13,"created":"2025-08-20 15:54:02.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/13"}}},"commit":{"parents":[{"commit":"92542e7c31003021aa49faed40ebe29c32a33ed7","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92542e7c31003021aa49faed40ebe29c32a33ed7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-20 15:53:44.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e732e0c46836e49b0b4959cc488a901227fa823a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e732e0c46836e49b0b4959cc488a901227fa823a"}]},"branch":"refs/heads/master"},"4e96a00fee3d6aa3434249f112ba52cd78e9c285":{"kind":"REWORK","_number":14,"created":"2025-08-22 05:20:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/14"}}},"commit":{"parents":[{"commit":"1066e05a5dfd611f2f763c2ecea27cd6d60e9c1b","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1066e05a5dfd611f2f763c2ecea27cd6d60e9c1b"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-22 05:18:32.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e96a00fee3d6aa3434249f112ba52cd78e9c285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e96a00fee3d6aa3434249f112ba52cd78e9c285"}]},"branch":"refs/heads/master"},"b404f5634eabc8e2d9c89f838e0838ce1321c589":{"kind":"REWORK","_number":15,"created":"2025-08-23 01:17:19.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/15"}}},"commit":{"parents":[{"commit":"21a72996eef209baad0a333531f2518bda12667d","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a72996eef209baad0a333531f2518bda12667d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-23 01:17:06.000000000","tz":-420},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b404f5634eabc8e2d9c89f838e0838ce1321c589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b404f5634eabc8e2d9c89f838e0838ce1321c589"}]},"branch":"refs/heads/master"},"ac67a7601ace5b02d4104e49f9b989fa861805c1":{"kind":"REWORK","_number":16,"created":"2026-02-10 04:43:10.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/16"}}},"commit":{"parents":[{"commit":"4c3e95ef100f8b7c2f3c60beb7ab78c2c25972be","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3e95ef100f8b7c2f3c60beb7ab78c2c25972be"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-02-10 04:33:17.000000000","tz":-480},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac67a7601ace5b02d4104e49f9b989fa861805c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ac67a7601ace5b02d4104e49f9b989fa861805c1"}]},"branch":"refs/heads/master"},"549761e841a264ef13f1d65876bb65e84e4da66a":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-03-03 05:39:12.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/19/951219/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/17"}}},"commit":{"parents":[{"commit":"9a1adb5c67fa0022ed4880c0202aca5f3d0e2b21","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a1adb5c67fa0022ed4880c0202aca5f3d0e2b21"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 05:34:32.000000000","tz":-480},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/549761e841a264ef13f1d65876bb65e84e4da66a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/549761e841a264ef13f1d65876bb65e84e4da66a"}]},"branch":"refs/heads/master"},"7ccb0d0925a8d5406bd6947e31c0e9e6743b9534":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-03-18 13:35:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/18"}}},"commit":{"parents":[{"commit":"1d4c87ca9121c856804ab328becbf81cc3b75474","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d4c87ca9121c856804ab328becbf81cc3b75474"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 13:19:02.000000000","tz":0},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ccb0d0925a8d5406bd6947e31c0e9e6743b9534"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7ccb0d0925a8d5406bd6947e31c0e9e6743b9534"}]},"branch":"refs/heads/master"},"f39b629370feeca877285061ff86ef0dcb6bcad2":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-03-18 15:18:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/19"}}},"commit":{"parents":[{"commit":"92a8f1bcd8d97e2291ef8366a4d473ef75c148a9","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92a8f1bcd8d97e2291ef8366a4d473ef75c148a9"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 14:57:12.000000000","tz":0},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f39b629370feeca877285061ff86ef0dcb6bcad2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f39b629370feeca877285061ff86ef0dcb6bcad2"}]},"branch":"refs/heads/master"},"7e3a8373f268aa5866cf0dd9e1c61ae0b63ac06a":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-03-24 17:58:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/20"}}},"commit":{"parents":[{"commit":"0bd26531142205b06418786dec82c08ec95f3b12","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd26531142205b06418786dec82c08ec95f3b12"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 17:40:00.000000000","tz":0},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3a8373f268aa5866cf0dd9e1c61ae0b63ac06a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7e3a8373f268aa5866cf0dd9e1c61ae0b63ac06a"}]},"branch":"refs/heads/master"},"c432ab5899a66edd33c71e41d1e9e94719bf87b6":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-03-25 16:30:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/21"}}},"commit":{"parents":[{"commit":"3941f498fd456c193a07488e9088ab876db96da3","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3941f498fd456c193a07488e9088ab876db96da3"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-25 16:24:45.000000000","tz":0},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c432ab5899a66edd33c71e41d1e9e94719bf87b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c432ab5899a66edd33c71e41d1e9e94719bf87b6"}]},"branch":"refs/heads/master"},"99a31d46811a899b3080103257b4584a5a393cf0":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2026-03-31 14:54:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/22"}}},"commit":{"parents":[{"commit":"9786fb07f0377a4a19cd6564d19a271fc11faa58","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9786fb07f0377a4a19cd6564d19a271fc11faa58"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 14:53:54.000000000","tz":60},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99a31d46811a899b3080103257b4584a5a393cf0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/99a31d46811a899b3080103257b4584a5a393cf0"}]},"branch":"refs/heads/master"},"c0f3f81784cd6b994da24eac11d819a8de4b040a":{"kind":"TRIVIAL_REBASE","_number":23,"created":"2026-04-09 09:21:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/23"}}},"commit":{"parents":[{"commit":"de984474c9f7142b4a4106c57d978c6f8cc6be40","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de984474c9f7142b4a4106c57d978c6f8cc6be40"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-09 09:16:31.000000000","tz":60},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0f3f81784cd6b994da24eac11d819a8de4b040a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c0f3f81784cd6b994da24eac11d819a8de4b040a"}]},"branch":"refs/heads/master"},"89400f2f75deb3e402de47c02cc8fb61049ca934":{"kind":"TRIVIAL_REBASE","_number":24,"created":"2026-04-13 09:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/24"}}},"commit":{"parents":[{"commit":"8ca043f23388852b9e2f828bd4d3575a41d7411b","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ca043f23388852b9e2f828bd4d3575a41d7411b"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-13 09:46:53.000000000","tz":60},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89400f2f75deb3e402de47c02cc8fb61049ca934"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/89400f2f75deb3e402de47c02cc8fb61049ca934"}]},"branch":"refs/heads/master"},"f1e1ab317506b2feb5d8574c1f5cba9e861b13ee":{"kind":"TRIVIAL_REBASE","_number":25,"created":"2026-04-14 15:09:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/25"}}},"commit":{"parents":[{"commit":"93f0abc8e3c6494bccb79c5842ada2950f8f2cad","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93f0abc8e3c6494bccb79c5842ada2950f8f2cad"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-14 15:08:58.000000000","tz":60},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1e1ab317506b2feb5d8574c1f5cba9e861b13ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f1e1ab317506b2feb5d8574c1f5cba9e861b13ee"}]},"branch":"refs/heads/master"},"81fb83459a1ad9fb8788fa3069774b95feb95f2e":{"kind":"TRIVIAL_REBASE","_number":26,"created":"2026-05-06 10:50:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/19/951219/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/19/951219/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/19/951219/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/19/951219/26"}}},"commit":{"parents":[{"commit":"5a16b75b6bd8bcf40c0ba3f2502d3506d03a2e06","subject":"Common: a new decorator to log timing breakdown for slow requests.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a16b75b6bd8bcf40c0ba3f2502d3506d03a2e06"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 15:07:38.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-06 10:50:06.000000000","tz":60},"subject":"Object: log breakdown timings for slow DELETE requests.","message":"Object: log breakdown timings for slow DELETE requests.\n\nChange-Id: If590482a376aeed02391627931bd5b1f24569c6a\nRelated-Change-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nCo-Authored-By: Alistair Coles \u003calistairncoles@gmail.com\u003e\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81fb83459a1ad9fb8788fa3069774b95feb95f2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/81fb83459a1ad9fb8788fa3069774b95feb95f2e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
