)]}'
{"id":"openstack%2Fswift~951352","triplet_id":"openstack%2Fswift~master~Ief0a29c4300edf2e0d52c041960d756ecc8a2677","project":"openstack/swift","branch":"master","attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-06 10:15:10.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2025-11-06 04:04:58.000000000","reason":"\u003cGERRIT_ACCOUNT_7233\u003e replied on the change","reason_account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}},"hashtags":[],"change_id":"Ief0a29c4300edf2e0d52c041960d756ecc8a2677","subject":"s3token: Enable secret caching by default","status":"MERGED","created":"2025-05-29 23:48:41.000000000","updated":"2025-11-06 10:15:10.000000000","submitted":"2025-11-06 05:09:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"951352","meta_rev_id":"c8bb1d7dea78485545ec4f2a105fd40e5c43c624","_number":951352,"virtual_id_number":951352,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-06 05:09:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":2,"date":"2025-11-06 04:04:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"all":[{"value":1,"date":"2025-11-06 04:04:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}],"REVIEWER":[{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 00:40:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-30 01:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-05 23:27:06.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-11-06 10:15:10.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"}],"messages":[{"id":"78d2d4dd6e894c410f97e5b70a57f213d3a34332","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-29 23:48:41.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76d49a2731f49dfd18dbbca53e7091e686e4857a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 00:40:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51007703fa7d4270ba157f1ea4635375\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9f5914279174163bf3fee23d1952650 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"f738657b52f1bd0a7ada5559da9dfd6c54af212b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 01:35:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd7f7b2343c443999beba53331549ce9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2bed184ab0d2412bbcc357ea51b2af00 : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/e1669f1981734fa699a06b60e1ecb23e : SUCCESS in 56m 55s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bf19ad35be1343bd9dbd9df68e77daa9 : SUCCESS in 57m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/913952e50d284b18b7ec16769c429487 : SUCCESS in 51m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/da81417163a24e4d85f54522246c1399 : SUCCESS in 1h 07m 12s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91200e2a72c042949bc00a047b16a51f : SUCCESS in 14m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9beb98052b8347cf8956360bc1507887 : SUCCESS in 20m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0a20eb7a7e9749839473b9797b81f906 : SUCCESS in 17m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7857bad626af4b189417b2f23349a9bd : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/14d8c23b05a44bfbbe2e14c4600b24ed : SUCCESS in 16m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e207deae27594bd5a4219755bdac68ea : SUCCESS in 20m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/814c66ec29654d95bc05464e461ea3ba : SUCCESS in 19m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bdab1a183c6048e39b61364e8365a2ea : SUCCESS in 21m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/70abec46668c44d5a356003bd8ec4936 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/53909d0fe3c3449498c782a0cb8c4e5d : SUCCESS in 14m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0c437a66b7614010a3371b818e484957 : SUCCESS in 15m 26s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/573efdbb44bb4f6284ae51e39c3c6340 : SUCCESS in 5m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6bb534cf9b8245c0a9a782daa044b88e : SUCCESS in 14m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/54193e8a55074142b57eb15459d58ced : SUCCESS in 7m 49s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/608dc07fa26546ef92a357055a01412c : SUCCESS in 50m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/cec4481bba3d4d8fb1b0a32357c54a1f : SUCCESS in 52m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/be075775d44f47acb79f641ad2889eed : SUCCESS in 49m 22s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2a4a3ea654e44c2496b71988ba858837 : SUCCESS in 18m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/226845dc5a224374834bf119d1a1289e : SUCCESS in 3m 43s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e72df57b90f24f379ab187959f01d1ee : SUCCESS in 16m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/046dfe9faf604818b41ffaa28a73f48b : SUCCESS in 40m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b544565f3ef94e059f8ae1bc56761be2 : SUCCESS in 53m 27s","accounts_in_message":[],"_revision_number":1},{"id":"6172d3cb336e65289157996584b87925ed5357e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-02 20:20:29.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7d2f0ce0407ea737d87b9befeeafd1e92e60ebd7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:56:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/414d13ff163349cd869fe4f623e461df\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/469521a2b3204090beb875553ebe1802 : SUCCESS in 34m 37s","accounts_in_message":[],"_revision_number":2},{"id":"2aeb7d3d6be532dd015b17919a87e237aebafb24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:57:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7299d842d90947468227897a729bb895\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba952802842c44d2a2982ef2138d2088 : SUCCESS in 10m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/0fb11af4124c4f79905ff730c98dbfc7 : SUCCESS in 1h 00m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/01b09d77d92d4dd0ba5ca3bf6dec532b : SUCCESS in 1h 10m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/afb1afca53b44db885524fde53006375 : SUCCESS in 52m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7801b43b3f5a452f821c286381d67157 : SUCCESS in 1h 01m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/80d8b70748454e8893c75ff8a1fc6db9 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/39ef2b9b66f44df0ad4af1b070ab012b : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0b93bb7ca6e649fdba75eac709d28d4a : SUCCESS in 20m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/74623ee7c5d74a70ab6c6f6094784f3b : SUCCESS in 4m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ed3bd3c6831348f2920572b09d27bb82 : SUCCESS in 19m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dd65a567d529427b8b3c93edeff07bcf : SUCCESS in 16m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b2dfdb15274149b295f74be991e43b8d : SUCCESS in 23m 14s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/22739e54f24549a18ba143ef2a6105de : SUCCESS in 30m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cef4a2e980de489ca2f3217f5ef81832 : SUCCESS in 15m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3f72c9f0034b4041abac4ed5b8919174 : SUCCESS in 15m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6396d4d5b46f471a8ac27d499a5923f8 : SUCCESS in 16m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e30ad9259c3f47e5a5499625343a966f : SUCCESS in 6m 11s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd0b7bdf44194bb4a7a65f39da8c9c88 : SUCCESS in 16m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3dbb082dabcf43e3bde90d95e2815bd4 : SUCCESS in 7m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/03a74d551b5d4a719f9847ef6bf89fb7 : SUCCESS in 49m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c82787fa1ae843d29dff9b3d2b266b40 : SUCCESS in 45m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3e86f9f81b644f86aea19798c0ebcf29 : SUCCESS in 49m 47s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ebbc6924c76c4f6b9127bf60efb9b2fe : SUCCESS in 24m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a2a6030d02498a8a3ccb0fa78c8edd : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/421bc6ca96ce4831919ab8994cfc749f : SUCCESS in 22m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/a3e109a16bd14f9aa0f2540125ef63b1 : SUCCESS in 46m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f90ea19d43f4402a5f5514033659ff1 : SUCCESS in 51m 08s","accounts_in_message":[],"_revision_number":2},{"id":"3f1353127ecafaeb459b2becdce2108d0ab40713","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-20 04:46:35.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f6565d4606cfeeb7404ca52b5ad5cd091f6b4e9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-05 23:27:06.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9ab6920f8a3c1a949c6277ae1a41c1a0549de33c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 00:27:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f65f953892074fd59b00369a2e2ff87f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4cce6ff4c6a04c9fafbf39e7bf1454e6 : SUCCESS in 57m 37s","accounts_in_message":[],"_revision_number":3},{"id":"f091ec99ac9e2537fb7a3f870f51299b530cd872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 00:38:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f4b8fac9079645698084f67174fbda2d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1070f842d3444f9d8189760aedb96bf1 : SUCCESS in 9m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/2d934515a8d540b18de2acaa69b95720 : SUCCESS in 59m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c305b5ea3080463aab470a6c0c7a6367 : SUCCESS in 56m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a22e1bb5519041599370e164d38353ae : SUCCESS in 49m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/23ac2cbcb7454a3bb42823e2e36b5776 : SUCCESS in 58m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f8f0a1d19564565af2adfc01f76256f : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b071a4fca85c4b478f524683dbf94dc6 : SUCCESS in 18m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bd7486c02bd4e629b31d052df563277 : SUCCESS in 17m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/838b5b25887748bebd16daa24feeb9bc : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/deb2c4cff94043eabcd838c17f481582 : SUCCESS in 25m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/426d9946cc984af6a5d25876805cb716 : SUCCESS in 22m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c1f4b09f3ad047c4b743030fbf54448b : SUCCESS in 34m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b0ff2c869f85483d828b0824705a7c25 : SUCCESS in 31m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4e789a605cef4441822bebd5358c855f : SUCCESS in 15m 55s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b5a9200e007a43bd99bbeb9dc1e6d091 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/571a0053ca1442698c7a8f64c012f5be : SUCCESS in 15m 18s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/9b8a297264dc4261863839e606cc27d4 : SUCCESS in 6m 35s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e00424f719434cfbbf4ed45fd33354c7 : SUCCESS in 15m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee027aff59fc48d8b071929562a6eb47 : SUCCESS in 8m 06s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd9c36e952164b808349a3caf9a243e2 : FAILURE in 55m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aae7d2230df0419bb1a9484f33d40371 : SUCCESS in 32m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/04181a21ca6140708317f1d11f6d1291 : SUCCESS in 27m 02s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f0255d76f1de4ad2ae12c790b8793340 : SUCCESS in 23m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9bf7e6bc3e524d3887d757885b938e0b : SUCCESS in 4m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ace932b693374d79b555d37e2e4c1151 : SUCCESS in 20m 05s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/787b765f6d9b4c908a3d1f330cfd3b23 : SUCCESS in 46m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44d06a8a06824ab8b5febc0cbe5c948f : SUCCESS in 45m 02s","accounts_in_message":[],"_revision_number":3},{"id":"1b2f0fd2aac8648406a9bafff750cf5b6ceae666","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-06 02:26:09.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"92d967d92911777357a1a94816153a053ec259d3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 03:41:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/915fddb4c097414fba815ca327f3455f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f20366c652444ef0acfcf47816a5bedb : SUCCESS in 34m 27s","accounts_in_message":[],"_revision_number":4},{"id":"31a11d848ea65b2340102603e51bb7586fed2812","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 03:47:39.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3bc024eb7824d29ba3df5348f79b1f7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f170a83f678479f8e9c0a855b7c405e : SUCCESS in 11m 08s\n- grenade https://zuul.opendev.org/t/openstack/build/4918859be3764ac38fa2ebcd4b17946a : SUCCESS in 1h 00m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/78266e838f3845f48b50fe7f6866484e : SUCCESS in 1h 02m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f67017792a8c4c2ebf8dd2e2a5b52015 : SUCCESS in 54m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c47a1f5758c4307b51ddfc0a1cab169 : SUCCESS in 1h 05m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41fa5163b12d466495fbea00df4e5d6e : SUCCESS in 24m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/497ac5f60c434a0fbb2e4040fc6e7c33 : SUCCESS in 19m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56990df4a0704a789251ac795c3382ca : SUCCESS in 27m 03s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/34fa1348bc7342bbaef97f8c43fa3d83 : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e7af92abadf34605ba86ebd8569fd53f : SUCCESS in 26m 32s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2e75855ac84749148b30764595ea930a : SUCCESS in 23m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e325189403d44a379a9ab0deb7081b52 : SUCCESS in 32m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cd4297a67ba44097bc5b20cf6c332eb8 : SUCCESS in 14m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e8f68cd52d6847249a886bf95ffccdfd : SUCCESS in 15m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d9c888dc185a47628ecf20e674e0e951 : SUCCESS in 18m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7b9406f015f74b39adccbed039d958bd : SUCCESS in 14m 34s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/663bf419228c404cb7425a1a4d519129 : SUCCESS in 8m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b28290d6d5ff45b29a9e30b2c528add1 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/c335825c99454c0b867ae935f779c2f9 : SUCCESS in 9m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca0a765293d74d899590036f1b118a46 : SUCCESS in 39m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b0d630072a5f453b9a129c236f6b023b : SUCCESS in 35m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f516e58f7aff42c6b05649e430f033e9 : SUCCESS in 35m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bf15d9e7c48e4b3c8c3c7334da0ac57f : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe3b350526448d695983f3bde6ee9d2 : SUCCESS in 4m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e63a8cef6e904647bad44b1b66f23abb : SUCCESS in 24m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ff8194463abc4671afbc5759d0721e09 : SUCCESS in 53m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7240d13fac1f45d5b9ea98a34893da38 : SUCCESS in 51m 38s","accounts_in_message":[],"_revision_number":4},{"id":"0d874c34865019f5e32d1b502cc728ee1e873868","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-11-06 04:04:58.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"78fd85c1ace14d852f56ac5d2f7fb699ba5495ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 04:05:25.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"7e6cd219ec0ce1bed65094de1b071b9e11d90607","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 05:09:00.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0dad9a8ed1594beaa913d630e8268f04\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c681fc74d0c46cb9099a035c18c738f : SUCCESS in 9m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/323672611752406dab7cc02940784511 : SUCCESS in 56m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/743f180a6df1441c8d86cb5f7c5ed572 : SUCCESS in 33m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f51e148682f54dd296f066d81a9156a9 : SUCCESS in 34m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e5535fd296847398ce47571a76a4b14 : SUCCESS in 59m 00s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/0546f12b96fb43ef849f1dcdb4cd7eed : SUCCESS in 7m 08s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b8da2372a2344788bdbefa21ea748c20 : SUCCESS in 21m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8b6c43cd72fb4a9ca756bb10d6014a27 : SUCCESS in 21m 10s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/cd6a24a30d174d9f89290e03c434729e : SUCCESS in 34m 32s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/065d26b8b1784229be7a9cfb81aa8ee4 : SUCCESS in 24m 02s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/6c8e426f7e2f4f88b83c05de3d99d8f3 : SUCCESS in 10m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0f00fbe337f14925aee1e98f3587e205 : SUCCESS in 14m 01s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5e5f8fe7c02147d39f4c8321970cce5d : SUCCESS in 9m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/eff61dc8122849e1b42c23bc89c5bcc8 : SUCCESS in 6m 25s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3f210376311346799c60d6c29fa3f30a : SUCCESS in 7m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/79c81f5c93c94c1bb0fa65c0e7948988 : SUCCESS in 57m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9a61f71980384df391e8f5b905cde7fc : SUCCESS in 46m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de902a19a3fa4b84857c7a7e2e8abfb5 : SUCCESS in 55m 26s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/fee0f2bbb2cd4a18ade2b17deb1fa85b : SUCCESS in 13m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73c9285943c84c679f79f02cc7f48e51 : SUCCESS in 4m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/91cc39d89f684e46bd40753a1e7a0d3a : SUCCESS in 19m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90e58577925a4fd98f27334fdc7e32e4 : SUCCESS in 49m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77be05326b9b4c58bb21f947ce2c2aee : SUCCESS in 56m 52s","accounts_in_message":[],"_revision_number":4},{"id":"7a4c75fde3742475c617bdd99393abcb14dcac64","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 05:09:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4b53c14057a4efb744f79b62c6d14413e947299a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 05:09:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fccdff2c02f64bb090005f777a73ef33\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2760b786e0a34f3f90f14466b675f907 : SUCCESS in 42s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4c1a2516dd0449c3886e8b30e3d8f274 : SUCCESS in 42s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c8bb1d7dea78485545ec4f2a105fd40e5c43c624","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-06 10:15:10.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951","revisions":{"c111e1119ce6d6268d144421341cad8806c31342":{"kind":"REWORK","_number":1,"created":"2025-05-29 23:48:41.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/951352/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/951352/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/951352/1"}}},"commit":{"parents":[{"commit":"afacfb6cea88d1edd71c7a70a5532e3e12bd8304","subject":"Merge \"Object-server: change labeled timing metrics sample rate for debugging requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afacfb6cea88d1edd71c7a70a5532e3e12bd8304"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-29 23:48:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-29 23:48:26.000000000","tz":-420},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c111e1119ce6d6268d144421341cad8806c31342"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c111e1119ce6d6268d144421341cad8806c31342"}]},"branch":"refs/heads/master"},"97dc0ae89c211de8bd991276310bcd5a05664163":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-02 20:20:29.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/52/951352/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/951352/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/951352/2"}}},"commit":{"parents":[{"commit":"ad41dbeffe0ed7d27fd4eb3ae8537c11068f374e","subject":"Merge \"s3 compat tests: sanitize object listings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad41dbeffe0ed7d27fd4eb3ae8537c11068f374e"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-29 23:48:10.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-02 20:20:04.000000000","tz":-420},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97dc0ae89c211de8bd991276310bcd5a05664163"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/97dc0ae89c211de8bd991276310bcd5a05664163"}]},"branch":"refs/heads/master"},"fc0a602ce8a8860bf6f128ebe0e629198f97841d":{"kind":"REWORK","_number":3,"created":"2025-11-05 23:27:06.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/951352/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/951352/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/951352/3"}}},"commit":{"parents":[{"commit":"4c83b5b88075aaacddad7b9a50739acb2017d456","subject":"CI: config Keystone users and s3token with creds","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c83b5b88075aaacddad7b9a50739acb2017d456"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-29 23:48:10.000000000","tz":-420},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-05 23:26:40.000000000","tz":660},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\nSigned-off-by: Matthew Oliver \u003cmatt@oliver.net.au\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc0a602ce8a8860bf6f128ebe0e629198f97841d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fc0a602ce8a8860bf6f128ebe0e629198f97841d"}]},"branch":"refs/heads/master"},"5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951":{"kind":"REWORK","_number":4,"created":"2025-11-06 02:26:09.000000000","uploader":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"ref":"refs/changes/52/951352/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/52/951352/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/52/951352/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/52/951352/4"}}},"commit":{"parents":[{"commit":"b147eae95d0c53e0fb4a7b672914f9a818da3e31","subject":"Merge \"trivial: tests: clarify exception handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b147eae95d0c53e0fb4a7b672914f9a818da3e31"}]}],"author":{"name":"Tim Burke","email":"tburke@nvidia.com","date":"2025-11-05 23:30:21.000000000","tz":660},"committer":{"name":"Matthew Oliver","email":"matt@oliver.net.au","date":"2025-11-06 02:23:00.000000000","tz":660},"subject":"s3token: Enable secret caching by default","message":"s3token: Enable secret caching by default\n\nNow that we need to pass the service creds to keystone, we might as well\ndefault secret caching by default now that they need to be provided.\n\nThis patch also adds the required s3token configuration to CI so we can use the\nswift service creds to fetch s3api secrets.\nAs well as also configuring keystone users for cross-compatibility tests.\n\nChange-Id: Ief0a29c4300edf2e0d52c041960d756ecc8a2677\nSigned-off-by: Tim Burke \u003ctburke@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5f89d14ebb6e6fed3b2ef507f8084fe7f05fc951"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
