)]}'
{"id":"openstack%2Fswift~951412","triplet_id":"openstack%2Fswift~feature%2Fmpu~Ieb7a9783731c11f1c08db398eae07ffafa127460","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-compat-test","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-05 13:35:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ieb7a9783731c11f1c08db398eae07ffafa127460","subject":"s3 compat tests: sanitize object listings","status":"ABANDONED","created":"2025-05-30 14:13:44.000000000","updated":"2025-08-05 13:35:13.000000000","total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4306f024695e665a726fc84bb22570e1093883f9","_number":951412,"virtual_id_number":951412,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-30 15:08:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 14:50:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-05-30 15:08:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 16:15:21.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"}],"messages":[{"id":"303cdd199165b3c499dbeb6d95dab842b95aaa1f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-30 14:13:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e63bbcabc63ee361b244e7f2a8464852c8d4f082","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:50:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/019bdb2258be47b9ac1827170a4eae06\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ba114621ad4a4b208080e1f52ad588dc : SUCCESS in 35m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0d523e3f0ab6297c7f5f0119b2482a25ed1f25b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 15:08:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4934a1ff32fb42bbb1c96a5226f6dc5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec38dc1cad741d3a1ef4f39fd30bd88 : SUCCESS in 9m 34s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/45382c01135a4f9399fb17db13eb91e9 : SUCCESS in 5m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/223ebdb809584b73839fa6a296d11f0e : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8fb8670bf0144a6d9c9dec885076295f : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"13824222340f9c68ed10b347571ed42460d3e9fe","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-30 16:15:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d89f4be7c0b84ca7c2964dbe8aa176675897ff04","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-05-30 16:49:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"659a10aea654930fa9ca8122edf0b1e5b42d742c","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-30 20:43:37.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8f7b6457b23899efb20182872764711a5efe252f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-30 21:07:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4306f024695e665a726fc84bb22570e1093883f9","tag":"autogenerated:gerrit:abandon","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-05 13:35:13.000000000","message":"Abandoned\n\nmerged on master","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e00c11ef054ec53fc3d1ca8c81a8bc8b14c27d78","revisions":{"e00c11ef054ec53fc3d1ca8c81a8bc8b14c27d78":{"kind":"REWORK","_number":1,"created":"2025-05-30 14:13:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/12/951412/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/12/951412/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/951412/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/951412/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/12/951412/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/12/951412/1"}}},"commit":{"parents":[{"commit":"889ae251d3ed7cefc1891ef9ae366492dd6dc326","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/889ae251d3ed7cefc1891ef9ae366492dd6dc326"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-30 13:32:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-30 13:32:02.000000000","tz":60},"subject":"s3 compat tests: sanitize object listings","message":"s3 compat tests: sanitize object listings\n\nSwift does not return all the parameters of objects in a listing\n(e.g. ChecksumType and ChecksumAlgorithm) so pop these from listings\nbefore making assertions.\n\nChange-Id: Ieb7a9783731c11f1c08db398eae07ffafa127460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e00c11ef054ec53fc3d1ca8c81a8bc8b14c27d78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e00c11ef054ec53fc3d1ca8c81a8bc8b14c27d78"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
