)]}'
{"id":"openstack%2Fswift~951443","triplet_id":"openstack%2Fswift~master~Ieb7a9783731c11f1c08db398eae07ffafa127460","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-06-02 17:11:16.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Ieb7a9783731c11f1c08db398eae07ffafa127460","subject":"s3 compat tests: sanitize object listings","status":"MERGED","created":"2025-05-30 20:46:39.000000000","updated":"2025-06-02 20:13:42.000000000","submitted":"2025-06-02 20:11:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951443","meta_rev_id":"b287a55ef4cccdffac3d9b49bcda6c87d79c13ae","_number":951443,"virtual_id_number":951443,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 20:11:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-06-02 08:24:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-30 20:47:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-06-02 08:24:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 20:46:39.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-05-30 21:49:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 08:24:11.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"}],"messages":[{"id":"9d3c68670075b399a33af223a3c5846e38c7c56e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-30 20:46:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a711c518dff94d14652ea5e04a9a6892ad237ec9","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-05-30 20:47:27.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53a9c1c998f2a6370db7ee9aa023cc81e9f963e5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 21:49:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/55e09c2117ba4467bca07f62b5f99180\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb8564459b6845699a844f0a0dadb00b : SUCCESS in 10m 38s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/603ad9cded7e4cc5b8b7af4c71c58328 : SUCCESS in 55m 13s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3f13cff1c07a410088fff3396c789462 : SUCCESS in 7m 04s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/49b66c33f2a445008bebced552f7dad1 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/897943769e4b40a0949c6e9ca463c6e5 : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":1},{"id":"7ad2d4fcc747d9e60c87436fb979f93552c27856","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 22:08:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/408e8ff114f645a9b50639a79213b84a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1b5bd08b44d145729989b04be6539980 : SUCCESS in 1h 19m 03s","accounts_in_message":[],"_revision_number":1},{"id":"40f9ec719fa43bba49375c696f6a20358fc12db0","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-02 08:24:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"fc03cdbcb79d9f43b0b540ec2a5a0fb46fba6671","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 08:24:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb69c031ff26a95aa792de2f05f7bbc8393aafda","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:32:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc91ff92707c415789ff8f830a49c8b5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aa042a8d62994a4094b98cbe7879c28a : SUCCESS in 9m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1bba8e14f5324b7785599ed61216a069 : TIMED_OUT in 3h 03m 42s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/326f1672c813464bb42a30ba63e29274 : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4255ad74d561400c9a35f1c9f4486d2e : SUCCESS in 18m 57s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/cbc90efeaabf4156890670800c8437c5 : SUCCESS in 26m 13s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3678adce08354dc9ab9a2f29db146fe9 : SUCCESS in 34m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1e22319f60244002a38685fb0a95737e : SUCCESS in 15m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d725879ff86422e8786108c89343a9a : SUCCESS in 15m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/312f38b0fbef4ed8920aa8f9aa29ae26 : SUCCESS in 21m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ba542042338c401ca8b75431749eb6a0 : SUCCESS in 9m 06s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ee3f21021b464764aff05afbcf5866fa : SUCCESS in 7m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d4ba215820f4be2b5c9176e7031af35 : SUCCESS in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"655c4cacf7559801483f1b5b2b1a60c96ae47666","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-02 12:16:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d15862362f6f4f8d60c27628e3e8eeb182420d22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:15:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c171cb8c96744db8b0a4a4cc486c1d58\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e16db8c417a4245815c9b551ae515b2 : SUCCESS in 10m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/53378a5dce1f404bbe926ee37b5ed014 : SUCCESS in 51m 27s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2304def6f2db4a8997617751724ca4b3 : SUCCESS in 4m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9df6c33f664c412e9c47f2d2aeb7df79 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/967ea782e27145c0bec0519a814d352e : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":1},{"id":"234bb0ec38d1da00a5cb6ad54fe40a631928da03","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:15:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e417e6918f9d0f29932240c675d985b7a55d4c0e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:36:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c0c5557ea3f4e7b8c7d9f7704053ed8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/affe8fcce7c34de7bff3e75466aae2ec : SUCCESS in 1h 18m 16s","accounts_in_message":[],"_revision_number":1},{"id":"96893f57971379b5035e6e776d2dc930a26d635d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 16:28:30.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/48e37783cce5451ba58c2b39826976c8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32eb4e37d443447d9b4786c20a19a483 : SUCCESS in 8m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ef5acfafa434422901a4a66fac13272 : TIMED_OUT in 3h 02m 33s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/6d7157e384e641818403ae6a3bba20f9 : SUCCESS in 5m 18s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/79e29fff72f5496cb3e649c07e59a142 : SUCCESS in 14m 52s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/36ef754853c042e88293ecd3d5771d26 : SUCCESS in 16m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b14f4704fdd547e1a0894f165a548275 : SUCCESS in 18m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a59ecad87cc24301b07eb020b1611321 : SUCCESS in 12m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/676b9d13ac634ddfb5d8bb72938ae73e : SUCCESS in 12m 46s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/19342426687b49a5a04fa4d8ab96564a : SUCCESS in 14m 12s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/58b905009f5b44b5aa51dcfc7cb92c17 : SUCCESS in 7m 42s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2d6ef13bf74043d0971164ce2c87b62d : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/225e357e42d3448c9c8ea89f2ab57fe2 : SUCCESS in 4m 13s","accounts_in_message":[],"_revision_number":1},{"id":"486410aa18b9041e4099ba45206a884a103dc833","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-02 17:11:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b7e7ce101d3e82941ec75a0a66121478a34edf15","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 17:49:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4df34f7f08a64796b15f8c96f62983bd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5db486e1e36d4f5b8329a4c41070e18d : SUCCESS in 36m 37s","accounts_in_message":[],"_revision_number":1},{"id":"a18fce732fa168da9b27725d380c72bc31449a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:24:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c81e8f65fc342d2a0505cc5351eaf39\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/962a94c6b7fe477bbc4ec05e43bfe791 : SUCCESS in 10m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2ec0c9953cfd4587b1fc5b7369de50b9 : SUCCESS in 1h 02m 00s\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7b3e37ceca5d4b12a914ff47e3e59d1b : SUCCESS in 5m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/abd93951a6024d99989d769e7d5cc22d : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15e5a9cc07f54895ac578fb8f74ed648 : SUCCESS in 6m 56s","accounts_in_message":[],"_revision_number":1},{"id":"1878cc9a4ac88cd306e8dda29ae19efd4266b1f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:24:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"da6fc8edc0dbf415e17354e91e750d69bf0ddebc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:11:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f045c0eb70d4147807b8c78ff509626\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acedf9195f9849388b2fbb8d9d5ff232 : SUCCESS in 9m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c06562a0d9514a22ab13005982cfd32a : SUCCESS in 1h 00m 35s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/360210ac4f4a4ae4ad9865131e700236 : SUCCESS in 7m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0081879618cb45dcafce5f50e74aeeca : SUCCESS in 23m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3db6ef6c755f4c8485e91b64aeb2b61e : SUCCESS in 15m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5431e4303ace4198921d96683ad35bcf : SUCCESS in 18m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/db8cde3fbc13403ea80334837e05ea74 : SUCCESS in 15m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/fc191402a74e48c8b21e20124dec4518 : SUCCESS in 15m 19s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ab938ba342274e6d9468ede07166bad7 : SUCCESS in 15m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4974210b90794bbc9f65afc54db5fd91 : SUCCESS in 9m 09s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b23ee103f91b4146aad852d1e005ebc7 : SUCCESS in 7m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8518748d397d475b88bf9a0b7434b29f : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":1},{"id":"a22acc39cda4ab1c0f05b61468a65dcabfcb9108","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:11:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b287a55ef4cccdffac3d9b49bcda6c87d79c13ae","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:13:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32f8470b514b422f9a6dde60879ca3d6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fa2b96e357b48468c870b33514b43e8 : SUCCESS in 1m 13s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4085906bb03441bbb9b7e21fc784cc7b : SUCCESS in 1m 18s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"962084ded0af2a1eb5875111cb239def423bbf88","revisions":{"962084ded0af2a1eb5875111cb239def423bbf88":{"kind":"REWORK","_number":1,"created":"2025-05-30 20:46:39.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/43/951443/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/43/951443/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/951443/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/951443/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/43/951443/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/43/951443/1"}}},"commit":{"parents":[{"commit":"afacfb6cea88d1edd71c7a70a5532e3e12bd8304","subject":"Merge \"Object-server: change labeled timing metrics sample rate for debugging requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/afacfb6cea88d1edd71c7a70a5532e3e12bd8304"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-05-30 13:32:02.000000000","tz":60},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-05-30 20:44:23.000000000","tz":-420},"subject":"s3 compat tests: sanitize object listings","message":"s3 compat tests: sanitize object listings\n\nSwift does not return all the parameters of objects in a listing\n(e.g. ChecksumType and ChecksumAlgorithm) so pop these from listings\nbefore making assertions.\n\nChange-Id: Ieb7a9783731c11f1c08db398eae07ffafa127460\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/962084ded0af2a1eb5875111cb239def423bbf88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/962084ded0af2a1eb5875111cb239def423bbf88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
