)]}'
{"id":"openstack%2Fswift~951578","triplet_id":"openstack%2Fswift~master~Iea1adfb93891e4bde62a11bfcef478d9b9696fd4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-06-07 01:07:45.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"Iea1adfb93891e4bde62a11bfcef478d9b9696fd4","subject":"CI: Skip grenade-skip-level-always similar to tempest","status":"MERGED","created":"2025-06-02 17:31:13.000000000","updated":"2025-06-07 03:26:09.000000000","submitted":"2025-06-07 03:24:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951578","meta_rev_id":"9e0e9214dbcab6d2a8013abbd4f05ed77e915e56","_number":951578,"virtual_id_number":951578,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-07 03:24:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-06 20:46:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-06 20:46:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-02 18:08:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-02 20:35:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4a1b32cbd5a7b9ec9b7e6a8ab973e762d8e0b090","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-02 17:31:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a211efaa38537d46cb83b964ee5278457eb1e949","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:08:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/badd4cf05cd94127a0b5ef0bf7fda0e0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fe880e8579dd4b709f7e6ec64f5cbc1d : SUCCESS in 35m 16s","accounts_in_message":[],"_revision_number":1},{"id":"14b09cf1aff398ae7899a93c470ab71db11a7a8b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 20:35:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2422ae9b754544b7b33358aed3b941b3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e601d833086a4c3eb42550648bc272b6 : SUCCESS in 10m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/e09db1019220493ba28749476306ade5 : SUCCESS in 54m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b2aa6fe160614797a1aa0312d5fc1700 : TIMED_OUT in 3h 03m 12s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eefa762d0a8b43baa1a1432a9a5ad70a : SUCCESS in 52m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7749cbbcf67e4f9eb68bd4bb4c43f915 : SUCCESS in 59m 31s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/471fa7684cfb448eaed41874aab984ad : SUCCESS in 17m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b47551637d22410198c60d09505e14ec : SUCCESS in 18m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/594182e08c654a2d869aded5eee2bf5e : SUCCESS in 19m 48s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/4284ee758668487387c628f2916c5da6 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c0aff6adf85c4b44b147df781d35379b : SUCCESS in 16m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b2521fbcc378446fa6c971a47e28471b : SUCCESS in 21m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/064074ca46d64407b31de071f13c3199 : SUCCESS in 26m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4677c95ae671480c821cfbfa01f8cc46 : SUCCESS in 29m 01s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e36a29b65e1d4efab128cb801c8122cf : SUCCESS in 15m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8d9d1faa269c4862873a958c1cdd917b : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1585f61d31f340df8c15d948519261e1 : SUCCESS in 15m 53s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2d6a8885450a4ce590d5fbf4d9187401 : SUCCESS in 8m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4506c43e04f74f52b15582774087c6ae : SUCCESS in 17m 25s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/18e1017f0c53418aa5ff9e4cfaf39d35 : SUCCESS in 6m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/87ec3199eddf4bafb33ce56bb15e2499 : SUCCESS in 53m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e59d9dace3234ad985aefabf756b3e0d : SUCCESS in 39m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a44b365aa2744ad5a4744db6c42879ec : SUCCESS in 47m 35s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7abf0648f043487f8fa24f67651bb3c5 : SUCCESS in 20m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7550225a80f849be8d1d33df7ef1ba0a : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a7b28faa9a964ca1a4cbb4e5b81fbe60 : SUCCESS in 18m 21s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e4b50870c72a450e82ea835894fffa81 : SUCCESS in 44m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ef510dae1f39426ebfd3dd9705787e32 : SUCCESS in 1h 00m 07s","accounts_in_message":[],"_revision_number":1},{"id":"1d1be20da46b465cfcc9572d566234cc822c7fe0","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-06 20:46:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"083f1c9e241b708dfc754a22272021ad3a1b56aa","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 21:21:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e1abbadf69e4c9ca32c9efdab8c71aa\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/194f2f91895a44f4b694be28e82329df : SUCCESS in 33m 09s","accounts_in_message":[],"_revision_number":1},{"id":"0c037397eca3fb24864290b50d284ae1500bcc72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 21:59:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c32a70d76e4768b1463b680d7980df\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d38ee2d313d43559748bf5f69854b6a : SUCCESS in 11m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/abb2d6a150594b0f98c3964e496c09dd : SUCCESS in 55m 09s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ad490c7436d401a8ccfd02979ba5cd3 : SUCCESS in 57m 34s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/90b9db746abc40309b7bfc4e4ac7d6dc : SUCCESS in 49m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f1e74c86eb146aa8d4bef28626e1b30 : SUCCESS in 38m 00s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/028b2ef9e1a84372b2f2f15f47e4f254 : SUCCESS in 21m 44s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa9181fd554441928a48b5f2ab770941 : SUCCESS in 12m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5442203c1fe24581a1114bcca3ddf08b : SUCCESS in 23m 36s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c8c0540c8e4b4448bca0ea1e4c9c9781 : SUCCESS in 4m 46s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8e55f3ed6ce3427abbf4e6d3c65d6d71 : SUCCESS in 14m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/f0d94221baa04e01a1522d6bc9eb1469 : SUCCESS in 10m 36s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ad9fd139cf784402a7a67818eb6b3146 : SUCCESS in 18m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5377de34e9d0424d9896e4c8c22e3a75 : SUCCESS in 22m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e5a04ae8f61342779f4567ba89612c1d : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/593c54c4dc5547e982ced037410417b8 : SUCCESS in 12m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/9f1f3c59a39b4465ae828b6c91ac0897 : SUCCESS in 15m 06s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ccefe2839d00417f9f5d057b0883fd27 : SUCCESS in 5m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/504926512fe24ee2b55c5d03efa93a5d : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/db5c786bbab44332bb27e2d58992068b : SUCCESS in 7m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/41a48293976a469c889f9b0192a7ae2b : SUCCESS in 52m 36s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6de4ad27cdd94efa8a294c6f4e51300e : SUCCESS in 43m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fe94dda6a8964f46b9e46f5169888984 : SUCCESS in 20m 37s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2e8eb4ecc9af47119df68a076af5c02c : SUCCESS in 8m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da38aa65f6514c75a861185e17fbe840 : SUCCESS in 4m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01713f94257e4f10a1069856bed91b40 : SUCCESS in 13m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/be231bbf883d4302bcd192ee9875d246 : SUCCESS in 50m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4f302a8c01b74c5ea94c4ea4125e9a1c : SUCCESS in 59m 58s","accounts_in_message":[],"_revision_number":1},{"id":"dc683c27d91809f06e73da3d7e6b935d08f78bfe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-06 21:59:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56f82e2ee945b14c41e915824c212dfa873bd378","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 01:02:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/689cafdbf2b14955afc1ca19ba2e85d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c60cf5ded8341f3b87895cfac806ba4 : SUCCESS in 8m 25s\n- grenade https://zuul.opendev.org/t/openstack/build/54418be84790412da3620739a8366fdc : SUCCESS in 47m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f734525df2f04e8caf5086a46f01896c : TIMED_OUT in 3h 02m 28s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/17ab3666df9b4c58a5a2429df475b08b : SUCCESS in 28m 57s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/47184b56d17740468fc7c2817acdfd96 : SUCCESS in 52m 06s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8ab4f26fb322451ba6bef1c992bb0c01 : SUCCESS in 3m 57s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1cfaa568e7984d1f98857c65c3bbb54e : SUCCESS in 22m 40s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c426275f46f4479bb049d28fd374a928 : SUCCESS in 7m 45s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4e46e343a0cf42a5946fd6875a839cfe : SUCCESS in 18m 53s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e4209f43077c45e2943cfcedda6880dc : SUCCESS in 17m 38s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5eadf9c76ee24a4bb2c485f5e6b8af9e : SUCCESS in 11m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/2796e7640b144433b032fc2ce27c2cb8 : SUCCESS in 14m 51s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c2b2180a8288409ca9507ae9829b6812 : SUCCESS in 3m 26s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bc9db7191062463f97c3281f3a751519 : SUCCESS in 6m 48s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/322412cf64f14c238de3181fa168d604 : SUCCESS in 32m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ef424f63ca274342982278b96668b1bc : SUCCESS in 54m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/99fa5dc4ff6b47069ee16bf1dedbf79a : SUCCESS in 31m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/277691a887ac4f9daa1818088ce3553a : SUCCESS in 22m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e05f0dd20e246b2b7e8171180df13f9 : SUCCESS in 3m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/883dc5d1f33140d780b9505f3e7531fa : SUCCESS in 11m 15s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b574d1b8b264e078c2c0c0411e8366e : SUCCESS in 49m 54s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18f765481d6444998ae26ef0a4e6c9d0 : SUCCESS in 54m 31s","accounts_in_message":[],"_revision_number":1},{"id":"5ec2e4a2876ee9197a1c28419d25395d8fe8d7b7","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-07 01:07:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8c29907d52aea37ae8b3f74c9b5cabc2b10cb344","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 01:47:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2b0789264e1246c0b034ea42592e364a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/76b8e043a50943aaba54c95cde6d2884 : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"1b43be9236d9269eeed9818a520eebaa3bf5fb96","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 02:12:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a2f5520355040c6bc2d06bc639f8efa\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f16fc106d114581a66ce8a4abe09ab5 : SUCCESS in 10m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/e5b68c2687704d05a921f921cfe81ac6 : SUCCESS in 57m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3a29ff80167245b48aedb341a0b9cb79 : SUCCESS in 59m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c931ce2d0c34462baf6db3a457b59fc1 : SUCCESS in 57m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c64925e68d7543c7b04b178cc8c09262 : SUCCESS in 58m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/58966f00262b49a395347e6fd304b13c : SUCCESS in 19m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/41f0e2ff487147b9ba2e1346be042d66 : SUCCESS in 20m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fbcd80577c704d8b86971567ebf9c3c9 : SUCCESS in 21m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a10d7436fb204316a9f390a3369c51e3 : SUCCESS in 5m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fe99b9c16b3246cda4ad1bb24c27524e : SUCCESS in 8m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2afd518b62a14941bc82decb813e3ef9 : SUCCESS in 20m 07s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/462f46018daf428197e887695297f9bb : SUCCESS in 16m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7a2461eaebbe48e6bbc3636d83e42d72 : SUCCESS in 10m 32s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e8fc590ffa214156a234a912ba85a8bc : SUCCESS in 13m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c82b62644f95410c935d055b89bd23df : SUCCESS in 14m 04s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fdf71b8b457a4402b792653c4cc5e05c : SUCCESS in 7m 55s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8e0aecc5be35491088e8772683d13287 : SUCCESS in 4m 56s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/9ee5cdbdd437496fb8071c72b6e1eee2 : SUCCESS in 13m 40s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f17598fff3134c19944e8ff01087c9b8 : SUCCESS in 6m 05s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b9953851d394292b8cb19e65fd2d969 : SUCCESS in 33m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7daf5c8b14844934b9c257ff3425940f : SUCCESS in 44m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1b094e8292b048ea9810cbe9e3660983 : SUCCESS in 34m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/998ac6b939344d1791ef034d8341ba4e : SUCCESS in 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccebb2a092854571bca199ddfd1048c9 : SUCCESS in 3m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cd387a81622849a4b7121e0cfd38b2f5 : SUCCESS in 16m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3cf755fd90c44d6592895e3c761a03d0 : SUCCESS in 48m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4a4438ed69c14015a9ff964c426389cb : SUCCESS in 58m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d2d82cbd541a1353abf14cee5dcc34d1cd52e6b7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 02:13:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"98b17c43405a747499d251af0ef330c2ad4905b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 03:24:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62d7cfa9486749d5977b9f739c4e295f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/714ee11cfd274365a08ddb1ae9a68b97 : SUCCESS in 9m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/e579871a0b024af99c3fa6fd10bdeba6 : SUCCESS in 1h 00m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ca8ad213a1594b338276a24a4c140dcd : SUCCESS in 1h 03m 38s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/388cce0cdb10415189dbd7e0fbe4a880 : SUCCESS in 49m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/03fcf6027d95458f9b908000b90f558b : SUCCESS in 53m 53s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/70a4e6816b2a46ef82af70e2bd3f88fd : SUCCESS in 5m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/020965f4cae14f17bd1215f823c7686f : SUCCESS in 25m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/10c8d11976ec4ad2a89e0fc32dc648fe : SUCCESS in 19m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/02222d61a3d34780aae079a0f13335cc : SUCCESS in 23m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1db0522d49404974bca94b1f31093263 : SUCCESS in 7m 57s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/dafb258ef08149d6bf1d9edc8d387a43 : SUCCESS in 13m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0086613237ba4d5a9f49e10f7f3ad97f : SUCCESS in 21m 28s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6f93506dacf94849afb3a790ed308754 : SUCCESS in 7m 03s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/ff5fff0c6c344baab4cc30800d587f69 : SUCCESS in 7m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f32dc3a7095f451e983da2c37345bd81 : SUCCESS in 53m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/adceb84786ca4a858b7b890f2312df2d : SUCCESS in 43m 08s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9caad898c5374ef18411146e80036877 : SUCCESS in 49m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/16e5190212dd4be6bb09b5d40e474dc1 : SUCCESS in 22m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4920eab09e9c4839b17316f369fcc8f9 : SUCCESS in 4m 24s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/594e587de76d4653918f4659f4d6360c : SUCCESS in 17m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6abb62799a6c45b989dc24ffe16fc8db : SUCCESS in 48m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53bea01d57e146e7ab3940b112729e38 : SUCCESS in 48m 19s","accounts_in_message":[],"_revision_number":1},{"id":"a161a1f01b46c515928d64667ac27c315756a123","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 03:24:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9e0e9214dbcab6d2a8013abbd4f05ed77e915e56","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-07 03:26:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6032e4dbad504224ae9cfad3dbabd615\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4727ae5c1c6144d1ab46d5fea54f3179 : SUCCESS in 1m 06s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/4f24c905bddc400d889bd86b6095b80b : SUCCESS in 38s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49ae2aa9435acbcdea24e1ff941df7e9a8d770ee","revisions":{"49ae2aa9435acbcdea24e1ff941df7e9a8d770ee":{"kind":"REWORK","_number":1,"created":"2025-06-02 17:31:13.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/78/951578/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/951578/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/951578/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/951578/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/951578/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/951578/1"}}},"commit":{"parents":[{"commit":"184641a7545833a1566da348fe638ae435ec733c","subject":"Merge \"tests: Keep port number in valid range\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/184641a7545833a1566da348fe638ae435ec733c"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-02 17:31:05.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-02 17:31:05.000000000","tz":-420},"subject":"CI: Skip grenade-skip-level-always similar to tempest","message":"CI: Skip grenade-skip-level-always similar to tempest\n\nChange-Id: Iea1adfb93891e4bde62a11bfcef478d9b9696fd4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ae2aa9435acbcdea24e1ff941df7e9a8d770ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49ae2aa9435acbcdea24e1ff941df7e9a8d770ee"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
