)]}'
{"id":"openstack%2Fswift~951676","triplet_id":"openstack%2Fswift~feature%2Fmpu~I1191ef6805305b361ed6c341d2a5a530e3683442","project":"openstack/swift","branch":"feature/mpu","topic":"p-mpu-custom-async-updates","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-09-10 08:38:18.000000000","reason":"\u003cGERRIT_ACCOUNT_7847\u003e replied on the change","reason_account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}},"hashtags":[],"change_id":"I1191ef6805305b361ed6c341d2a5a530e3683442","subject":"WIP mpu: aligned sharding","status":"MERGED","created":"2025-06-03 16:11:10.000000000","updated":"2025-09-10 10:08:24.000000000","submitted":"2025-09-10 10:03:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951676-p-mpu-custom-async-updates","meta_rev_id":"abab23b5d3b9a00c92bd2787f6843bd230d3621c","_number":951676,"virtual_id_number":951676,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-10 10:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-09-10 08:38:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-09-10 08:38:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-03 17:15:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7f1ab2dba78f79629642cda7bb753493d461cfd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-03 16:11:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a0edeb8b410e81e0461fae65e45075c0c77d654","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 17:15:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e32d39a6d334671847e12ea806ce903\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b564fe6ae47b4e4cb25b99a01a184c9a : SUCCESS in 10m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e86880bcabf34dbc982e478d7870b3de : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfed04710e1a447887832dbfa760e361 : SUCCESS in 20m 58s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7528a341ff4144d7acde6ccb047c8b6e : SUCCESS in 22m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e065280a7d8949b38f66eeff51f7f8b7 : SUCCESS in 6m 34s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/00a51dbe52ba489cad7b36ecb3e70c37 : SUCCESS in 21m 39s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/72067875a7a04dc18152d0de624b59da : SUCCESS in 23m 35s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/b08d898bb22f4b50876d95960347be13 : SUCCESS in 22m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/33009a515b0d418d82635bdd8e56d2a3 : SUCCESS in 24m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/48218457c2b64a5683be970c6006b33a : SUCCESS in 18m 40s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9b8e4ee0da40401c9622fa3d512e1b45 : SUCCESS in 14m 56s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e3b4e548bf694a569515bbab2b3ce4fc : SUCCESS in 17m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f474603c25e3495f801ed03ad9b125ee : SUCCESS in 6m 24s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/7aec82006db24cba9735fda32ba34239 : SUCCESS in 13m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/37a75a7094bf4ea29b82ac23647b14f5 : SUCCESS in 9m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2fabfe45c624becaa18fd69835b0d01 : SUCCESS in 59m 41s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82fcba6ecdb048f3ae7d1d9ddd29a018 : SUCCESS in 19m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95aacb193fd64a5896ac4c9ef346d727 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b35b3d9cc49741799e14ddbcbaed6d2d : SUCCESS in 22m 46s","accounts_in_message":[],"_revision_number":1},{"id":"885877e54d60284dc38fe3e21984e4c924252466","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 01:05:48.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b73de750de3148b58ce82fe3d2ca3775\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a7273a21744a401f90b64fe6ba8b4e86 : TIMED_OUT in 3h 00m 51s","accounts_in_message":[],"_revision_number":1},{"id":"5087197cc8839b57d2e4fe7f84a2d9c8dd172c5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-10 17:35:40.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9aa61b8fdecad7bd86e01efa1d35fa4e695b040b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 18:49:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/147f227c5b064f3e968589e2d0f14270\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01cd3bbc60e54e23a2c85ead4f266ad0 : SUCCESS in 10m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/010135e0d0a34d2eb07ba445dce0a762 : SUCCESS in 20m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/07962869e9de46719554ac2c9c511aaf : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/49b0264fe4be4496a44e060791598d80 : SUCCESS in 20m 59s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/aaacda16aa4a4733b537f3fc692cdb4d : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a26f9ce876994498863fe828fe7fb93e : SUCCESS in 20m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0175f0f77da646cc9167f52d0ff004f5 : SUCCESS in 18m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/3d2e94ca2c2949a695be6677a268bc45 : SUCCESS in 11m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/011b8aa825a64c0bb77da996897870e8 : SUCCESS in 14m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f3aaaf097344426cbe2ef9fc51fe9a77 : SUCCESS in 8m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0539c99e2d91441fab0b88d58c338e19 : SUCCESS in 8m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/932e98cb96ed4024beaa672c55ffd01f : SUCCESS in 9m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c395a2d5abbe4a748898451a345ca7ef : SUCCESS in 5m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/877b08da5a8c4b0ca5bedc83123fe85b : SUCCESS in 10m 45s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b18a3cff803a4f01ac657c82e99d395b : SUCCESS in 9m 33s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/10c833fea7774abe83dcf4a066b89873 : SUCCESS in 46m 24s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e4b11de3f7224cf388aaa7c5138773a3 : SUCCESS in 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73f89d33dab14e60aba5feb00b6b1e1a : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d14d325e17804a8f94828bf019d93708 : SUCCESS in 21m 20s","accounts_in_message":[],"_revision_number":2},{"id":"686a1c52190182b8cb6e3fb01089773e1ae89705","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 19:17:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/93e1c808ebe04cb993dd2e193dd7ee34\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ebb2fd6aecca4958b3ab1827cdaadd1f : SUCCESS in 1h 32m 49s","accounts_in_message":[],"_revision_number":2},{"id":"4ddd72eccdf86fdd1e2b548a99a81e769751d7a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-25 12:28:43.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"2ac1f225d8685cba5d0eab6a5137499c72478006","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 13:27:35.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc9f8ff8fe3c4290a625453f697d094c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc8ef18f7e3c4bf585df0f9d5faa09b4 : SUCCESS in 8m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/08b5f1cc7392407497e2fb2129d1e0d7 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f267140b43fd4970971f4a139df273ee : SUCCESS in 20m 02s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5e1550d7e17940fe863357064eb27e22 : SUCCESS in 17m 20s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5679480cc4984198ab54f347a4759635 : SUCCESS in 5m 26s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2289d37ef26b4562ae9ee69bb5c73ef4 : SUCCESS in 20m 58s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2ab813c962db4ac38c9c90d5e1a49e83 : SUCCESS in 19m 47s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2f57eecec15449c588991c8487366f06 : SUCCESS in 21m 13s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5973cb9cedb6409fafea2b948af5134e : SUCCESS in 10m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e82fb33b9b744cebaa51906e6b37b899 : SUCCESS in 16m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d5b528156b0d44ba9e6ff34e154a8eb5 : SUCCESS in 15m 40s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5f96b2885b114560b41cd478df45ec48 : SUCCESS in 16m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/353339cb37484ad1a978bb54a4db7f3a : SUCCESS in 3m 31s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d0061262d4d14d578cac85b62377e432 : SUCCESS in 16m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bf3755d16df9415e8796ede7f833cb7b : SUCCESS in 11m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ac56254d66d479c9d6979660710ba76 : SUCCESS in 52m 55s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d600ccbd5d634b168d6a6e677a60e849 : SUCCESS in 20m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1a7f758731f84c63a0d42ac3412d162a : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2602d137c7e64531ad9369f5ddbc578c : SUCCESS in 8m 47s","accounts_in_message":[],"_revision_number":3},{"id":"135c2d0c6de72d4aa07da4941df1ff0e82089e66","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-25 14:47:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14a831364fb64087949153bcfd4c42ae\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4c107c3039634f309a692e902f051efb : SUCCESS in 1h 35m 09s","accounts_in_message":[],"_revision_number":3},{"id":"90dbfbb11aaf9dd61331b262bf05bfb3abc0a063","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-31 13:09:18.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e09f3fd674ccfb768d79880237012009a8fb0924","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 14:13:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e2778bfaea740e7aa7d6bb52b4e8e77\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d3115d3ffdba4692b5bb947ba70e4284 : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":4},{"id":"cabfcc1ac969dafb859d250a485e9fcd7464a7ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 16:12:32.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f1a660dba842bf9d1d51643528752f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1a5dc73652784ce29be6706952967da8 : SUCCESS in 12m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/73210e8bd9f04bca970754439f8f916f : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/3f47ec429964436cb98b39945b51ac8f : SUCCESS in 20m 20s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12d30f3e30df4b109f8c40acf0043b42 : SUCCESS in 21m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/04ef18e8351a4fe1ae73544276a36d19 : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e7985cb5dcc045c5b9acf5ea472dd688 : SUCCESS in 23m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/bc4f6a1dd3fe454699736898798bb44e : SUCCESS in 19m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f154b674c73042f2a56e8808d526b1c8 : SUCCESS in 17m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/79030a776ece49c09d82ac1554e71d69 : SUCCESS in 23m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/30d533ad21b541d99cacda798cfbf3ae : SUCCESS in 15m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6cfddae6ceeb478889e8f0d3b0be670c : SUCCESS in 14m 34s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/820c2e46e0ec4d29867d4868ad1c574d : SUCCESS in 14m 37s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c5ec4c8c834a472d8160cbe24a46a180 : SUCCESS in 8m 21s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2ca6565ca2df4418839c9de5ef5b8c08 : SUCCESS in 16m 50s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/82766a7da5224e4ca03b0a5619e174de : SUCCESS in 8m 24s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0cf27aca0554a20a5a89e711e5ea322 : SUCCESS in 58m 21s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a54c0b4afd0d4fb7a63b2d72e58fbc54 : SUCCESS in 19m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e718f4f6fbb47fca0ab8f8f3ef9fdfb : SUCCESS in 3m 49s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d382511407594982bc2fa1756dca514a : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":4},{"id":"5c3580d20ddde6b75faa2b091e3f79228668f536","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-15 18:05:39.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"04bb324ddabf752a9026cff17871de9a996e738d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 18:45:44.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/10377a855df04afe87fb42602510744d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76261faeedda41a3bf6acea883c7aa40 : SUCCESS in 10m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/751cc4bfe7b54c319952aadc6186f633 : FAILURE in 20m 05s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5b5c180cf52d43a0a822c974d30df782 : FAILURE in 21m 04s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a2c9a3c00fe342459cca1b9505cb8422 : FAILURE in 24m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5d79685c6a834c939a5514cf5ba69752 : SUCCESS in 7m 39s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/611643087f6e4306b8f34e367c9b6b2b : SUCCESS in 19m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e00498ffae704bbfb04fddf034cd242e : SUCCESS in 11m 03s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/904fc30e50384440960c86ef38858166 : SUCCESS in 13m 46s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/cec0b4fe2b5f4657846e55f9c0bab6bd : SUCCESS in 26m 12s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9b21aa97e9534ee1bb8e128605456310 : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/468b5a33262840598ad2d8936e46ff34 : SUCCESS in 16m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/929516d4a1c543eea190d9913e621cb0 : SUCCESS in 18m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/121b41739c604ecdab5f4982f04567a3 : SUCCESS in 8m 17s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8ac2f5f5f6d0426b88c5e7e150b0025e : FAILURE in 7m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b598e0411c6e4cbbbf5035bebae7c1e6 : FAILURE in 10m 58s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/043146f0cc5247f5ae0feaea3e8cb7ba : SUCCESS in 38m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e438bb41952b43db9a37d4c4cf3481fa : SUCCESS in 13m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b29f84748f724a59b5473fb6fe6baae9 : SUCCESS in 5m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7c1da06b77924054b18e0b2c72901247 : SUCCESS in 27m 39s","accounts_in_message":[],"_revision_number":5},{"id":"c640df799047be9d1e9a1b28ac4da72ca07e6b5a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 19:55:08.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90a8d90f90994ab1916aa58ae01ddd66\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ad9b312e57f44a7aa515e8ba0db4a48d : SUCCESS in 1h 39m 08s","accounts_in_message":[],"_revision_number":5},{"id":"cb3ebbd4d76164af54f3a1f041bde7a6cacd3bc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-22 12:31:25.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"163f6466c6a888b227b93201cad854e2dbd37b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 13:10:54.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/419e6d1c100049f69979b8a0535b827e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fecae72fe264bc283c793d38fb371de : SUCCESS in 5m 29s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/70319388dfc64d7b813d26c61fadff90 : SUCCESS in 12m 19s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6e589ab56c444a8ca60b0d0261df507e : SUCCESS in 13m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de5d228d5328445f8c78101274aa3657 : SUCCESS in 13m 02s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/54f1563a6c05409d965f85b266bab798 : SUCCESS in 4m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a262092b54fc4642ae0492bc887e3899 : SUCCESS in 12m 19s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/884dada50ee04f748a248df939df99a3 : SUCCESS in 11m 21s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ca1db3378dac4ff78041d7ff13fb0a7b : SUCCESS in 28m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f9de2588d1494d43985795256dc89de9 : SUCCESS in 20m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/522c81b6c9a54b95918009192e9d2db9 : SUCCESS in 15m 32s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4b4a38b97222430bbf79d56563073c03 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d4faacccad264979832323d9faa9ed97 : SUCCESS in 16m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5206ece058ae4ee7b4b558b4cc1b8154 : SUCCESS in 6m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/6112862084c04272abc3ce11ba6eac5d : FAILURE in 9m 06s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e0a918272cfe4ae58c4945b103583702 : SUCCESS in 5m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ffc189f296ec4c758634031dba210dcd : SUCCESS in 38m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b6c3403dd5048148ac3a801550ea193 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15877e39ef124e5d8ef06a6c8263ae90 : SUCCESS in 2m 50s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/29c702a774514faeb2b17c0452c96b3f : SUCCESS in 18m 22s","accounts_in_message":[],"_revision_number":6},{"id":"6ea9ad5d9b255c990860c3cd2f14f2d29954a785","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 14:42:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6643f85c4565467dbd9fd62e7dcb564f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/c68bcd2383454eccbd66692fa02d6e85 : SUCCESS in 1h 39m 02s","accounts_in_message":[],"_revision_number":6},{"id":"bbf2308305e99b9d247b4b4170246b171eadacc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-05 13:45:55.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2f2db4628e371bf0d0908c0632f2af7d1645e8a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 14:51:05.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7e7abd22a32428b958c1cb179ce186c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9675a3c184247ec88521c406acadad4 : SUCCESS in 11m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/163fb6a41bb34c7a84c1d539e9187ac2 : SUCCESS in 20m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7bdccc007ccf4a3b9cd8ff71aefee74a : SUCCESS in 11m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a673c2101465440eb56121fe46987f01 : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c1af2aa6d38477eb55714568d4256ef : SUCCESS in 6m 07s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/746dfea8abef4218bbc43bd10e403ec5 : SUCCESS in 25m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8eea813ea63249eeaf9e5ad6ce9b56a3 : SUCCESS in 23m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1b3e4a13a3b54b3f884b3fcae45bafd5 : SUCCESS in 28m 23s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f49dc519d5944254ab3508fa5b9ec581 : SUCCESS in 12m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0cbd12cd32d940349ff1b49a04a40bfe : SUCCESS in 14m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/185d0621cb9d4cb3a5bfb5a4f14bdc85 : SUCCESS in 15m 52s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/24a58e6b3dc44967b154aed656bf3950 : SUCCESS in 15m 57s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bf56de62d4a74e9bba238d62d421f979 : SUCCESS in 6m 12s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3e7c75582d654636bb749d2df8b8a93f : FAILURE in 14m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1d89b1a2c8d045e2b2b04aecca13b2e4 : SUCCESS in 9m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/366bffc0476645bf8780c324c9a526cc : SUCCESS in 59m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8603d7bd9abe4d0c9e6727d27ea33f3c : SUCCESS in 24m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8306c8cc96c342b0ba7090a68e50afd9 : SUCCESS in 4m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/47594e21789e4024bf8a8c1ba2f27314 : SUCCESS in 10m 02s","accounts_in_message":[],"_revision_number":7},{"id":"760c55e3e78eaace82ff5275979eae7d96d471f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-05 15:02:07.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8774cad9c77a4947bc7b191e25785ad8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0765f492ff5e4339b2f90901aeefaabb : SUCCESS in 1h 03m 08s","accounts_in_message":[],"_revision_number":7},{"id":"1149f7c510f53b0f704c7eb116f210b172c49a24","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-10 08:38:18.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"06f7f594530333a19e7ca39655badccdc856dfa6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 08:45:18.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"38764bda64c3698b6f58ad5196a513bd41bdc106","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:03:26.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fdd07b503d642329fb7eef885e24655\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5c1fac1de494d1aada1a69941fb27ce : SUCCESS in 9m 14s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/31a5e655acf5478882417374c4f90300 : SUCCESS in 5m 17s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/8d718b537f0c4c30966257688240b3df : SUCCESS in 28m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8c67698ab79d47939fb818a131bcb8ae : SUCCESS in 8m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/54b0ba8f0fb14dc7869e4ee75ef5551c : SUCCESS in 32m 45s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/0fdf7c6910ed4b8294b3047d68e093db : SUCCESS in 30m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/1792c93d02ea41199047d4859ded0ee6 : SUCCESS in 10m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a08d16dc8647489db550dc7386c1a840 : SUCCESS in 14m 05s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e399906a063543208b20d342be1ee4c4 : SUCCESS in 10m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a1ec804ed0e84134b7620edd5109f35d : SUCCESS in 6m 01s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1c6d2e7927094a1ba76fe55354e3a5f6 : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6073cf9f281844ac871a9117c535ea9a : SUCCESS in 54m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/82acd9a3e992446a9c6a23311c5b1dab : SUCCESS in 14m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/add8f9fb8c554f49af163a5ae9f76222 : SUCCESS in 3m 57s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/dd5311db270147b6ab19d275cc94d63a : SUCCESS in 12m 24s","accounts_in_message":[],"_revision_number":7},{"id":"c2bd1ff0b7faf35d8d180d013752d9bdc5822d99","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:03:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"abab23b5d3b9a00c92bd2787f6843bd230d3621c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-10 10:08:24.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87d4f0ad37e24328802a39dad576c101\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/963e5a27aa9b48fa8b3d2d14a274c3b6 : SUCCESS in 2m 08s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/021588d78c6946c5bdd71153d530dfac : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"59cd241b31a31790b34db5010b63f55866674610","revisions":{"8b4e2891a02d900f65bf9ff0a02747736fa77d6f":{"kind":"REWORK","_number":1,"created":"2025-06-03 16:11:10.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/1"}}},"commit":{"parents":[{"commit":"889ae251d3ed7cefc1891ef9ae366492dd6dc326","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/889ae251d3ed7cefc1891ef9ae366492dd6dc326"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b4e2891a02d900f65bf9ff0a02747736fa77d6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8b4e2891a02d900f65bf9ff0a02747736fa77d6f"}]},"branch":"refs/heads/feature/mpu"},"17095fb1f67071b8dabea17f76696d0f969abbbf":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-10 17:35:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/2"}}},"commit":{"parents":[{"commit":"d5adce5fb80309a3bfcf9e7763c9a08e50f4ecde","subject":"merge master to feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5adce5fb80309a3bfcf9e7763c9a08e50f4ecde"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-10 17:34:53.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17095fb1f67071b8dabea17f76696d0f969abbbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/17095fb1f67071b8dabea17f76696d0f969abbbf"}]},"branch":"refs/heads/feature/mpu"},"6770363ba7675c314268b60cd9fd4f9d93d2ae49":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-07-25 12:28:43.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/3"}}},"commit":{"parents":[{"commit":"20f84c1aff4c497c3d32773030fb2bb252060411","subject":"merge master to feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20f84c1aff4c497c3d32773030fb2bb252060411"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-25 12:28:30.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6770363ba7675c314268b60cd9fd4f9d93d2ae49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6770363ba7675c314268b60cd9fd4f9d93d2ae49"}]},"branch":"refs/heads/feature/mpu"},"d2b0c1b794319fed40d79ce4e45c1035ece8b664":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-07-31 13:09:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/4"}}},"commit":{"parents":[{"commit":"e792b16f61f4a7629615ed5ada1ebfd4c7cf78a2","subject":"mpu: add compat test for listing sessions","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e792b16f61f4a7629615ed5ada1ebfd4c7cf78a2"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-31 12:01:21.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2b0c1b794319fed40d79ce4e45c1035ece8b664"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2b0c1b794319fed40d79ce4e45c1035ece8b664"}]},"branch":"refs/heads/feature/mpu"},"9733b7db8f7be95658af21ec38d3f1ac4f832761":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-08-15 18:05:39.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/5"}}},"commit":{"parents":[{"commit":"a479551fce8ddbf7c02364bb2c11d704767bddb4","subject":"WIP mpu: introduce s3-compat mode for object-versioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a479551fce8ddbf7c02364bb2c11d704767bddb4"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-15 14:06:29.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9733b7db8f7be95658af21ec38d3f1ac4f832761"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9733b7db8f7be95658af21ec38d3f1ac4f832761"}]},"branch":"refs/heads/feature/mpu"},"d6781f30a6e7451798837e6400b622340f7666d6":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-08-22 12:31:25.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/6"}}},"commit":{"parents":[{"commit":"d615069b04084d6a4e5c92a3743fb12b4b78786b","subject":"WIP mpu: introduce s3-compat mode for object-versioning","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d615069b04084d6a4e5c92a3743fb12b4b78786b"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-22 11:48:21.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6781f30a6e7451798837e6400b622340f7666d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d6781f30a6e7451798837e6400b622340f7666d6"}]},"branch":"refs/heads/feature/mpu"},"59cd241b31a31790b34db5010b63f55866674610":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-09-05 13:45:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/76/951676/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/76/951676/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/76/951676/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/76/951676/7"}}},"commit":{"parents":[{"commit":"b75717b4139caa5ce7297bcfa4076b4db7dd6e6c","subject":"Merge \"WIP mpu: introduce s3-compat mode for object-versioning\" into feature/mpu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b75717b4139caa5ce7297bcfa4076b4db7dd6e6c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-03 16:10:07.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-09-02 13:37:42.000000000","tz":60},"subject":"WIP mpu: aligned sharding","message":"WIP mpu: aligned sharding\n\nPrototype mechanism to force shard bounds to keep names with common\nprefix in the same shard.\n\nChange-Id: I1191ef6805305b361ed6c341d2a5a530e3683442\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59cd241b31a31790b34db5010b63f55866674610"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/59cd241b31a31790b34db5010b63f55866674610"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
