)]}'
{"id":"openstack%2Fswift~952109","triplet_id":"openstack%2Fswift~feature%2Fmpu~I4c60356bcdf476f918d423f510a494d41095111d","project":"openstack/swift","branch":"feature/mpu","topic":"merge-master","hashtags":[],"change_id":"I4c60356bcdf476f918d423f510a494d41095111d","subject":"merge master to feature/mpu","status":"MERGED","created":"2025-06-09 09:23:24.000000000","updated":"2025-06-09 11:36:06.000000000","submitted":"2025-06-09 11:34:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952109-merge-master","meta_rev_id":"99d4d624019ba014a056cd01bfddd8f9c60a255a","_number":952109,"virtual_id_number":952109,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-09 11:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":2,"date":"2025-06-09 10:43:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":1,"date":"2025-06-09 10:43:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-09 10:13:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-09 10:26:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"972db570d56687ca4fdaa9402d067916b6a149d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-09 09:23:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"27c667e4506f1c10b564194739f7b5321c46ce35","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 10:13:52.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b839469daba6400faaf8d6be5a151ded\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7aff532cc2b74ce8b91ce840287f458c : SUCCESS in 45m 36s","accounts_in_message":[],"_revision_number":1},{"id":"03e43b0038972cb6f65d532042452140082f5113","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 10:26:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/abad656aeb7e471aacbc5f4ddf0982e1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11f0d1bf83e045939bc89f93a02cd412 : SUCCESS in 9m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/137cab1312a14f13b74b8fad0e2b239e : SUCCESS in 5m 04s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71285ed4164349f188c601df307e3bc1 : SUCCESS in 10m 26s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1539ba6a6b03407f9750c84149c1798e : SUCCESS in 12m 52s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/94b315cc417047c7869594939f0de15f : SUCCESS in 20m 07s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8589d69bd3fa4a72bbe51d7b97b7cf8f : SUCCESS in 6m 40s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7879e2dbeba4b928be1a8149101a439 : SUCCESS in 17m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4e911405bcfa4a24b490e6d7e6ed82f9 : SUCCESS in 17m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/564fbc45cc2d4908a6a4d17352787314 : SUCCESS in 22m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/984e263e6f5f4b5eb9556321ae429715 : SUCCESS in 19m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d88a6fe115b641ed97dd09135c259cb0 : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c1dc951767e04651ab168ec45ded141f : SUCCESS in 14m 18s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8c1601faf6e84457837792a9112688f1 : SUCCESS in 8m 41s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ce91e7bbcc04dd1942e1c6e73ba7944 : SUCCESS in 5m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/52f9e3c11b6942889843f2e02949d2a1 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/1ece9d2538964ea7930a555b29bccbef : SUCCESS in 9m 59s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3cccaa70365a4956bf0f5d51aeb00959 : SUCCESS in 1h 02m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/90a2842d71c1405da9520ddf93feace5 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4e30d80c9b04d4e96c24cd91918d0aa : SUCCESS in 2m 26s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/42359a858d6f4277942ac4a8d8b81159 : SUCCESS in 17m 04s","accounts_in_message":[],"_revision_number":1},{"id":"c365a453c55a10b215aed03299b758084ab920db","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-09 10:43:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2f91e778d489bf23726fa1b7f6d1d259101a2ad3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 10:43:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"85d606610bb7353c5f18a402fdf80e8962da0eb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:34:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9971d0e7bb8e41d684e27e6635b5ae36\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10e0eac3b66d429db26a69d8cd8a309d : SUCCESS in 10m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a21fb0ab6eff4189bd5059c352cfeafb : SUCCESS in 2m 46s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c2ffd304b1fc4e4eb2d5e1a937afef1f : SUCCESS in 3m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/763c39bf3f1649b5bfa0c95bf15e71ab : SUCCESS in 8m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/621907390eb04e9e8e54871dfb981199 : SUCCESS in 8m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/e1a03f9150de4e45b8c9596474dcfa49 : SUCCESS in 19m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/b30836d0d32945fc91b2a3ab5de5f66a : SUCCESS in 12m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ceed61b4378a4a1dbbb2f60a02f04a0f : SUCCESS in 15m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/474745b8e6b14cf9b24c3c4bd0bea95b : SUCCESS in 18m 48s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/dd1542fa33d644fbb5fbf8f18cf1f8cb : SUCCESS in 5m 42s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2fabd73cd8ea4b2ba05d84e2651ec6cb : SUCCESS in 7m 53s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/68abc4cab8014e9ea43d7a27851d6994 : SUCCESS in 49m 18s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f9924c304ad467e9a5c6a5241bc9e3f : SUCCESS in 21m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9affb5f422654cc792da405a065e93a6 : SUCCESS in 4m 08s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/417f2ed7b89a4008865895c1cbb64ad6 : SUCCESS in 20m 53s","accounts_in_message":[],"_revision_number":1},{"id":"46ac279525b6fcda90eb8865f46975da8ceec6f7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:34:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"99d4d624019ba014a056cd01bfddd8f9c60a255a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 11:36:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8db2d9b5239d4a608e8b74cd6d007fe5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d7450a52b8416b8da20313909a79e1 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5da5a2286aec424b879edee0a83c05ad : SUCCESS in 38s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/ec056106ae314d239dfa543cacc36496 : SUCCESS in 40s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d5adce5fb80309a3bfcf9e7763c9a08e50f4ecde","revisions":{"d5adce5fb80309a3bfcf9e7763c9a08e50f4ecde":{"kind":"REWORK","_number":1,"created":"2025-06-09 09:23:24.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/09/952109/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/09/952109/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/952109/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/952109/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/09/952109/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/09/952109/1"}}},"commit":{"parents":[{"commit":"889ae251d3ed7cefc1891ef9ae366492dd6dc326","subject":"mpu: append link to dependant mpu resource to override etag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/889ae251d3ed7cefc1891ef9ae366492dd6dc326"}]},{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-09 09:21:47.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-09 09:21:47.000000000","tz":60},"subject":"merge master to feature/mpu","message":"merge master to feature/mpu\n\nChange-Id: I4c60356bcdf476f918d423f510a494d41095111d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5adce5fb80309a3bfcf9e7763c9a08e50f4ecde"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d5adce5fb80309a3bfcf9e7763c9a08e50f4ecde"}]},"branch":"refs/heads/feature/mpu"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
