)]}'
{"id":"openstack%2Fswift~952165","triplet_id":"openstack%2Fswift~master~Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1","project":"openstack/swift","branch":"master","topic":"tmp_labeled-metrics_token","attention_set":{},"removed_from_attention_set":{"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2025-07-23 01:26:09.000000000","reason":"\u003cGERRIT_ACCOUNT_34930\u003e replied on the change","reason_account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}},"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-07-23 01:26:36.000000000","reason":"Change was abandoned"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2025-07-23 01:26:36.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1","subject":"sq? add new labeled counter metrics to cooperative token patch","status":"ABANDONED","created":"2025-06-09 19:58:36.000000000","updated":"2025-07-23 01:26:36.000000000","total_comment_count":88,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e3b340e0a7ad1dcf766093d53b44de2999b7b3a4","_number":952165,"virtual_id_number":952165,"owner":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-23 00:31:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"all":[{"value":1,"date":"2025-07-23 01:26:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-09 20:45:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-09 21:05:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-10 17:46:37.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"},{"updated":"2025-06-12 00:29:45.000000000","updated_by":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"reviewer":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"state":"REVIEWER"}],"messages":[{"id":"151fec1f4a7e7bf059c05b7173891b1bc9dcbd8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-09 19:58:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"162371aa336549aa96fc5659696b30c7c0345496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 20:45:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f1bb5e83b6c4721839dec136ff9a4c8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7bd29cfa7484434eb71336fae121a0cf : SUCCESS in 44m 41s","accounts_in_message":[],"_revision_number":1},{"id":"99a60efa5019603f06b90d989930e2a4f3486b45","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-09 21:05:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d40a1a9d2f2475eb3ae33a2fc324a66\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff449cbd28f4ad6b0a21e79482a3e95 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/9f665079f3e7420099015fd8ae847841 : SUCCESS in 56m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ef724e4f76a641d5adc31b84130e6927 : SUCCESS in 1h 00m 45s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/657f5ddc0ce645508bf25f55d8cc948f : SUCCESS in 38m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/81a80d4d52a14aa0bebcf39cfdbc6923 : SUCCESS in 37m 17s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/6621e680bf604ce2b2cc8709d1f06c24 : SUCCESS in 19m 58s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/23fa38bb08f249df8c866239012b5ae6 : SUCCESS in 12m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1b7ae5d4dcd496da7551907ef96ed6f : SUCCESS in 20m 42s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1c2b0314884343f78fae1b49f11dc1f2 : SUCCESS in 3m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b0dfe3daeaca486d969270ce94a8fe47 : SUCCESS in 20m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7ceb40b58092476ea5a88a1cc32afcbe : SUCCESS in 19m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/af805521a2644b3c9643cd8db624d4a4 : SUCCESS in 13m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/181468dbea274220a5808af8aca54035 : SUCCESS in 13m 43s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/31a11b6706774c3ab105e3477a675627 : SUCCESS in 11m 21s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/43926b2a764943f39f2cb630156a8b30 : SUCCESS in 8m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/168dfa51f5784edcacf014c9bd3de038 : SUCCESS in 8m 21s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e78947a80d14efd97cc0558ccfa85ef : SUCCESS in 3m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/bd3fcf3b821844b19253747c65a0ed23 : SUCCESS in 15m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6fe87e446cc941538d9083bf2fc2d40e : SUCCESS in 4m 16s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8bc3f5377e54d3eb0c28c77603464ed : SUCCESS in 42m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0ad91b4bab247fa842a2300f352b44c : SUCCESS in 20m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8edab99cc72746a089aa43a051799c17 : SUCCESS in 21m 14s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa08b39d99f148a6b11a00cfe3fe68f6 : SUCCESS in 8m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec1cc271a2a44980b7f1c7ba50df6a1f : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/58ec4ea932384c5b846e15efb5e6579b : SUCCESS in 15m 40s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/67a87447e55b4bb799edec9ec8ad7e98 : SUCCESS in 32m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cb7341f8ae7b42d097211d5ccf62c51a : SUCCESS in 57m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8c93e6ac51ed04853524c5f0163af06e308820c0","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-06-10 17:46:37.000000000","message":"Patch Set 1: Code-Review-1\n\n(18 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c6410eff541beb3bb13aed2e330a381c909647c5","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-17 15:23:33.000000000","message":"Patch Set 1:\n\n(17 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f42dae3546be4127e9180027f45da1c6601dee72","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-17 19:35:59.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"1fd8ba730e73c6e36817940971fef1266ae2e9ff","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 20:16:47.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97489f34f10a464195a0ef96693c493c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1db2137fd04c401097ba22c46784ee2f : POST_FAILURE in 38m 25s","accounts_in_message":[],"_revision_number":2},{"id":"7adee82cd022b7070ae44571eac457bea8683d64","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 21:15:34.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4184187b45174687aea4919525439322\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e58799a414846dba04a2924327ef86c : SUCCESS in 7m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/2df542905bca4cd293fa5d181baee0de : FAILURE in 7m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0599744c3b194727965814914bc195d6 : FAILURE in 6m 47s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5fe7d3a454234713aa6863abdbd4858a : FAILURE in 9m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19a4a0e6f73d4f4bbe11a2a711b99c37 : POST_FAILURE in 6m 56s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a93d322fb167410e992e1a746bfec5b7 : POST_FAILURE in 21m 53s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be5d8d67f516461f8f1e1a641e2d69b7 : SUCCESS in 17m 53s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a7e6524c1a94150a74da6fbd20ed616 : SUCCESS in 23m 45s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3d19654c03b84adaa353a47c4dba1e09 : POST_FAILURE in 3m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/295503357c4a438ebe7a270cd5bb0d50 : FAILURE in 17m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/97931831fce14cd8b46446a6875f25bc : FAILURE in 21m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ae0fec3eaa704eba9f30cbddceab8195 : FAILURE in 13m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/80c81f0bf8024bd39267a6f92fc6af36 : FAILURE in 31m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/ade288c971ef42b2a3840bf9c1c29ec5 : POST_FAILURE in 16m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/d336d8c5ae4646228f5e014646bbe1c4 : SUCCESS in 14m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fbd266a84b094311b59c863fe364079c : POST_FAILURE in 16m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c43af8c92ee84629a6b072e51c5c0720 : SUCCESS in 6m 07s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81c4d4ad420b4851bfa02cbb7c00f45f : POST_FAILURE in 15m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b1feea3c53a049d692c21df39fdd87d4 : SUCCESS in 9m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/a73acabb85014feb9a602f5614a03980 : POST_FAILURE in 51m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/069052bc5580477db36f2e098fada58d : POST_FAILURE in 49m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5869ed237b1b46b5a825ac1019264c0b : SUCCESS in 52m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/60bc1e40460147809558b38fbae01623 : POST_FAILURE in 30m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fc00138a259416a8aeccc2ea8e5efff : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/012b0f2275f84c2b8e3770a770044b84 : POST_FAILURE in 18m 20s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b9c826b487c34e8aae021546cd7729be : FAILURE in 6m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59b1ba539269449db12f0fb9b11b425c : FAILURE in 10m 17s","accounts_in_message":[],"_revision_number":2},{"id":"9b2f74cd12b629600f67c1538b977bec42cc6754","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-06-17 22:17:22.000000000","message":"Patch Set 2:\n\n(19 comments)","accounts_in_message":[],"_revision_number":2},{"id":"1c1d50032f4ee056ed87d9d751dc33fb3fa93a00","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-24 15:22:26.000000000","message":"Patch Set 2:\n\n(18 comments)","accounts_in_message":[],"_revision_number":2},{"id":"81dee93148f525b4d21c691e7aca314a66c823bf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-24 21:21:17.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"94a3e3f01515777b307e08a61cf86c28982f609e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-24 22:54:02.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54aaf5bdbab544989199d6b194f7706e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e71b5d1493d844568bf4e98cbf2e8552 : SUCCESS in 1h 29m 39s","accounts_in_message":[],"_revision_number":3},{"id":"b94011816c54142d80a6bc351804a6f840cea817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 00:25:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7b7ce0e3d19d421a86a56af5210ac943\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6612b00c2374a6d8de33932a7c81620 : SUCCESS in 4m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/f115f1f14c154cef9bd642a53e304ee1 : SUCCESS in 27m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b473f743939c4a91ac876ff28766be59 : TIMED_OUT in 3h 02m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9ce63f62e9644c0da0a9e4f7ad91af63 : SUCCESS in 28m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2bf1b73a6ea423598bc69ba081e7bd5 : POST_FAILURE in 9m 10s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/33b74be0e7b846139d99c581e9e3b49a : SUCCESS in 19m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ad2e20495587468cb3bddc62a9b703be : SUCCESS in 11m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dfa4e23832864ee7ae2959d05ea17220 : SUCCESS in 20m 06s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/43224cb06fb84e38844ad0c7c48beedf : SUCCESS in 4m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2050a97d04aa44ef8824590eca29ebb8 : SUCCESS in 20m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/305e3c3902fc4573b976df90b507d0b6 : SUCCESS in 20m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/63f21946d40c404daa1df7ccdb3b6ba3 : SUCCESS in 28m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5b3c25a2a63c4326b1897bb2ef6d621a : SUCCESS in 29m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5cb197dc349a4012a59ec02f5b237ce4 : SUCCESS in 11m 43s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e597ba61a95f43739e67985ae4899a04 : SUCCESS in 12m 32s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1117c14a08854d70bc9907546c3a46b2 : SUCCESS in 12m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ef0d7360d774d718bf08557d7c81394 : SUCCESS in 5m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4ee202caa8b64f618e83463fedfa8100 : SUCCESS in 13m 42s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6170887cefae4030b8b65fecfd75e611 : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/25faa1d7fed24c3c9da3c96f3b12f23a : SUCCESS in 59m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ed492cbb2bbb48a6b1829d668685efea : SUCCESS in 34m 54s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e610364825ee4a1c928e79b422d6eaf9 : SUCCESS in 51m 00s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4d77232427084ed1978a6d34c1c1ac3e : SUCCESS in 20m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a25409a08ea84d5fafe584d8b4529e33 : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/baaa15bf38744d158aef6ae080289823 : SUCCESS in 19m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8b3aea03d6b14eb5854f3c33fc697a2e : SUCCESS in 50m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a691e4df48bf436d815334134c20f9bb : SUCCESS in 53m 05s","accounts_in_message":[],"_revision_number":3},{"id":"9c46c6f5af744c003e0c13a54c2e730423531fc9","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-25 15:16:49.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"9cc7bcf7429f197d2c5ca4e3e2f915d1cd7d7d12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 16:53:42.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ad9eae3d5c844cba52c0125131882ae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8718b77f41174b858e21e99ed3e8167f : SUCCESS in 11m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/2507a1303ab24f89919361ee3d891dfb : SUCCESS in 59m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2b9454e1e53c41b69a50e77fe6fb9c92 : SUCCESS in 59m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/cc122c58f7b3459f9c8b38c3b8faefcf : SUCCESS in 46m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e79208b36bd43918a09148189debace : SUCCESS in 54m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a1ccaf61eda94b92ac22d3fb89644859 : SUCCESS in 19m 13s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1f5980ddc5e7457ca12e70d229b3a289 : SUCCESS in 21m 22s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba11c6ab4656457ba7f1608c16962f7e : SUCCESS in 20m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/75ea241d496044609b32b87e374103af : SUCCESS in 9m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3a7f278c41fd4d8da86d3171f2097aac : SUCCESS in 20m 59s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/7dacf03347b14491ba8d966bbc42f4cc : SUCCESS in 19m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7516260d432e4dcb93a9e46b922c0733 : SUCCESS in 23m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/1b9d281807b44e869b6010bf0e5fb1e6 : SUCCESS in 11m 45s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/906575762611498abca284157f2597cf : SUCCESS in 15m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ff062f5df1da4987becac9bee80b0dec : SUCCESS in 16m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/edf31f61bcfb449aa0fc30ecb02c4aca : SUCCESS in 16m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/aee8098b7e2c4d51a1a0dbe22adb3b3b : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/76fedb062fa840d987bb2c84658aa440 : SUCCESS in 18m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/721feb2488494611b72cdd3b0f95cb5d : SUCCESS in 7m 19s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/2c08f8e64df746bd818a436b4f0c7b95 : SUCCESS in 58m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/f559a7590b8c4910be8a1aed8e53e2ea : SUCCESS in 46m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8602e97dfb33488d8a3e4d81047c90db : SUCCESS in 46m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2824a2496d5641f3a8e8d49157f668a3 : SUCCESS in 20m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bfbfce0ad7024bf4832fb950014f5278 : SUCCESS in 4m 02s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d966590a64074ecbb0d42783a8442438 : SUCCESS in 21m 23s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/8ed8f059b3af419e99d0ba15e8afda8a : SUCCESS in 50m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b98add6ae9f1499ca643a837e56b3287 : SUCCESS in 55m 08s","accounts_in_message":[],"_revision_number":3},{"id":"7efe68bcbf44a5d9e28be24ce159fd26199c1f43","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-25 23:15:50.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/042a315fb70a40edb208080707bc5d73\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9f7d3bf3d0ca455796df08c4d7019ebe : SUCCESS in 1h 45m 44s","accounts_in_message":[],"_revision_number":3},{"id":"dc63ac186e7ae737357845da93f0620db22363ac","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-15 19:13:28.000000000","message":"Patch Set 3: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":3},{"id":"04cb1df06b3915a880813d07710004320b8aad39","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-22 17:30:51.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"750f3ce309ca8a0bbae837ac8017a0fe2ffab7d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-07-22 19:59:05.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"4efc58f281a2087bbd953a87bf961f6ef0fba390","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-07-22 19:59:54.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"f2dd9aea267f3e57c1525d629f740166f841a2c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 21:04:45.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/95d46da0441941e4bde9c7d44854b41c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52335e59f8a74c78b9fc2f8b44ce217f : SUCCESS in 10m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0da99eec4e1b4dc3baba97977b90d1ca : SUCCESS in 31m 20s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c9b6bfe39960480c912ad6bf562f7ed0 : SUCCESS in 44m 43s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1829cae5765c4d1baed87b3e5dc72f02 : SUCCESS in 36m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0cdb34b1ba54b5583283f93c9db2f37 : SUCCESS in 59m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/91a9a23e461f4e73aea8c4909f2cc36d : SUCCESS in 17m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/351b067916e043c3aee36316b8be3599 : SUCCESS in 20m 00s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/af4920572dd547cdba07660c364d4954 : SUCCESS in 10m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/af850d8facfd415a886fce6bb0481bd1 : SUCCESS in 8m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/190e8ca46ae14545be3b2506f88a44f5 : SUCCESS in 22m 16s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d1ee28a61bc04b2ea74807488c992e8e : SUCCESS in 19m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a7e2e692da6e40ed9bc57b284446c8b6 : SUCCESS in 18m 54s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6e61e69ce3494512aad21491e18fa5e9 : SUCCESS in 19m 39s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4966edbfd8e24d50b6d93818647fd0ad : SUCCESS in 14m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8c069ab8a9e745428c8a79b6c197ce55 : SUCCESS in 15m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e248ce304cee46c5b52eea2a192c17a3 : SUCCESS in 14m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c13bd084bc124a88820a60edbe3236fa : SUCCESS in 6m 50s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5370ede7698a43b9904ebb60512ab2b9 : SUCCESS in 7m 34s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/388c6d3bd82f40f8a7a9f0228c4a6113 : SUCCESS in 10m 40s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c785153a2d194f08bc78ec9a863b3e96 : SUCCESS in 59m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7e67459b83da44919a7254e6e3198379 : SUCCESS in 38m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d5ff9a3a1bf74936b0a494ecabaa4700 : SUCCESS in 35m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7f8e66f056a647f9a293bc569076035e : SUCCESS in 20m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/34a968d67ae7487aadb3e8a16edc0616 : FAILURE in 6m 11s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2f6618221e424dd3920870ab2be6fa54 : SUCCESS in 18m 41s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/66e0606b203c443c8348f33e73119ede : SUCCESS in 51m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/50de0a14fe784fcdb7af43df0ef8c6bf : SUCCESS in 39m 21s","accounts_in_message":[],"_revision_number":4},{"id":"a6c300951fcfe0b61431f083ba575126657d47a3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 21:41:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89d013eca0c84d1db1c387da1cb631be\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b746077daa3f47cca72baa5b775f4726 : SUCCESS in 1h 29m 36s","accounts_in_message":[],"_revision_number":4},{"id":"075731971cff8448f80576d6709be620b33e6b5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-07-22 21:58:50.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"0b4ae2a1222134d0a9dc0361f220bf7419c92a96","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-22 22:09:38.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"87c45e360730c1590c6f4e12e40c62ba20c8673c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 22:56:05.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/591cffa7ae8c430dbe47adcec06b8875\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2e3aace36b7646f6a9f9e155a4899a88 : SUCCESS in 55m 26s","accounts_in_message":[],"_revision_number":5},{"id":"a390357a506a61064e63a034e9af06a1dc7f2601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-22 23:02:26.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/85284b9b59a94563823c4b1961223f26\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/296b598607f94688a15708108bff629a : SUCCESS in 9m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/d0d2983c51de47779be04868f68a9dd0 : SUCCESS in 1h 01m 59s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/92a173b5272a45638b4bd120a6ef9dc7 : SUCCESS in 34m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21806ef5b3774a2f97131462c230820e : SUCCESS in 29m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/509ed623888948e5a71aed44a2f7c332 : SUCCESS in 44m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4218bc4db5a04abc8d664148edfc9afd : SUCCESS in 10m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7f82c2e7f1374664875e44e6e6f61de4 : SUCCESS in 17m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ffebff8ef3e4a0bb4b91a548a3bbaab : SUCCESS in 10m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c70cc576d41540cdab2fd7af26300777 : SUCCESS in 6m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/02b5e15cd62f45e891a213ede2b949e4 : SUCCESS in 19m 26s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e7daad2f9e8c4cab85ac424eef503182 : SUCCESS in 9m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/5328eca97a194cc2970168083dd23099 : SUCCESS in 19m 18s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb20ab179eda4708ba331a1e6c2f5aa4 : SUCCESS in 10m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9eb81511ae834c89b8334498f6b621aa : SUCCESS in 8m 27s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/e673a67ecd3644deabe155ef3b4e5701 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/b6290cf4cb704954b1651acbbcf92061 : SUCCESS in 8m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5a2d4dc167a64f2a944a2f81b9c8c5f0 : SUCCESS in 3m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/66bce2e62b41418db288cb55a3ca925e : SUCCESS in 14m 03s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7a6ba36c87fb4c649ffebe2e616df77f : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/88fbe38c7b2e42ddabd0883c2f0d172b : SUCCESS in 50m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/460f6642aaea4617b92496f337354bdf : SUCCESS in 53m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/9f912abaf16d4c46847670ce3abc03fa : SUCCESS in 34m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/408f657cb37b446f8a5b82cf06bc0b24 : SUCCESS in 8m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5f9fb29c477475992185725d17ae75c : FAILURE in 3m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a51b397f6c474dd2913a7e322afc2ddc : SUCCESS in 17m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/30eb084fe85f40afa6a3549d0638faf0 : SUCCESS in 53m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/559cdb4359f349998a6d47c99003c245 : SUCCESS in 43m 38s","accounts_in_message":[],"_revision_number":5},{"id":"033ba3e3e28604af05158296555644d409e80b26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-22 23:21:44.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"29dd2c4bd62c3ce400b79501d0ea8aa67729c043","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 00:31:53.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b80b089024a43b5b4ae7e030c986bc7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a95e9f448d2a4d92b53707595d8a4d6d : SUCCESS in 8m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/236f57627a99410d8741beac654f4524 : SUCCESS in 30m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e0d38494731a4d97a210366eb2534bad : SUCCESS in 34m 52s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/1eaf7d0350694a1aad4df580b1380faa : SUCCESS in 42m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f044572511248759abb276a1d5bda9a : SUCCESS in 1h 04m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c520d168f2ac4b5a9683eaf54d9b5c48 : SUCCESS in 18m 51s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fdf108dbaddb4b01be957b734412554e : SUCCESS in 10m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/acfc78be58524fe5b69079d9af49e16e : SUCCESS in 17m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b40f6c9507204f4da045002ec3bba7e7 : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6954e03837cd444ea1db9ce7e146cde4 : SUCCESS in 14m 14s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/37b703cc04744e0d86ace842f27e2624 : SUCCESS in 16m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/da0c508ccf2e42ef97508485216515b0 : SUCCESS in 10m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d77d80f09ff24390a11f2858168b6d7a : SUCCESS in 16m 49s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e5df548c55bb45dfb912357f32dba1c6 : SUCCESS in 7m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/5ef4dedb93f24b00a77ee3544a279e29 : SUCCESS in 19m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c34a9807d6dc4268a767382857332a9d : SUCCESS in 12m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cbd927a5fcf44de184675c46e6284b71 : SUCCESS in 5m 01s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/481c6d7d8d9d4f62a599c065b5072b48 : SUCCESS in 15m 01s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/173ba36c6d9747bc8099a2b0c206e702 : SUCCESS in 4m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/859d1dd25fa84b2db2930d4f2b293cc5 : SUCCESS in 57m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ac5a128e654348ecaa83acd1c3146833 : SUCCESS in 49m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/57fa504d3b23424596336dfaf6413d70 : SUCCESS in 50m 05s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d2d441b89fb0429a86e8328676fdc52c : SUCCESS in 19m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f9e69899fb74aa19eea0de2c24f0c97 : SUCCESS in 6m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/f73c5ddccd4344cb87a908cd7308c22d : SUCCESS in 13m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c3114ce9551b433292aa3428a198aecc : SUCCESS in 50m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c504f2ca550341a0a77903dd593bbbeb : SUCCESS in 27m 14s","accounts_in_message":[],"_revision_number":6},{"id":"c2287dd655e5344e0b2a53d0a6597da75b0d1b20","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 00:32:54.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ffbd1b0d9245c3bceaca44ede5169b\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5151c6f7448f4026a147f9c4e3462ce7 : SUCCESS in 1h 08m 42s","accounts_in_message":[],"_revision_number":6},{"id":"ef5dc2cfb2a37205cb4dbf53186c900813b7521a","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-23 01:26:09.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e3b340e0a7ad1dcf766093d53b44de2999b7b3a4","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-23 01:26:36.000000000","message":"Abandoned\n\nsquashed into the cooperative token base patches.","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"c7d0b7ccb5019cf6f6d81998e2b3eb8b84aa9895","revisions":{"facccdebb6b1b55d63999c5af5755eef56095d66":{"kind":"REWORK","_number":1,"created":"2025-06-09 19:58:36.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/65/952165/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/952165/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/952165/1"}}},"commit":{"parents":[{"commit":"165de47ee37a434b5ea420299740bdb4be706d39","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/165de47ee37a434b5ea420299740bdb4be706d39"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-02 19:22:29.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-09 19:29:28.000000000","tz":-240},"subject":"sq? add new labeled counter metrics to cooperative token patch","message":"sq? add new labeled counter metrics to cooperative token patch\n\nChange-Id: Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/facccdebb6b1b55d63999c5af5755eef56095d66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/facccdebb6b1b55d63999c5af5755eef56095d66"}]},"branch":"refs/heads/master"},"cfcebc15b296b6d5f1dd8230be7ba1e239d2afe1":{"kind":"REWORK","_number":2,"created":"2025-06-17 19:35:59.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/65/952165/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/952165/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/952165/2"}}},"commit":{"parents":[{"commit":"165de47ee37a434b5ea420299740bdb4be706d39","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/165de47ee37a434b5ea420299740bdb4be706d39"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-02 19:22:29.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-17 19:19:29.000000000","tz":-240},"subject":"sq? add new labeled counter metrics to cooperative token patch","message":"sq? add new labeled counter metrics to cooperative token patch\n\nChange-Id: Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfcebc15b296b6d5f1dd8230be7ba1e239d2afe1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cfcebc15b296b6d5f1dd8230be7ba1e239d2afe1"}]},"branch":"refs/heads/master"},"2295ac9e5f3eacdf122b1ee56d8936b6cc102efe":{"kind":"REWORK","_number":3,"created":"2025-06-24 21:21:17.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/65/952165/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/952165/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/952165/3"}}},"commit":{"parents":[{"commit":"df82f1963c26efc079bcdea7853fb402663177d6","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df82f1963c26efc079bcdea7853fb402663177d6"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-02 19:22:29.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-24 21:00:55.000000000","tz":-240},"subject":"sq? add new labeled counter metrics to cooperative token patch","message":"sq? add new labeled counter metrics to cooperative token patch\n\nChange-Id: Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2295ac9e5f3eacdf122b1ee56d8936b6cc102efe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2295ac9e5f3eacdf122b1ee56d8936b6cc102efe"}]},"branch":"refs/heads/master"},"c3c491f0ff4a0b262b5f32c9d25eb53bd9a905fd":{"kind":"REWORK","_number":4,"created":"2025-07-22 19:59:05.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/65/952165/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/952165/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/952165/4"}}},"commit":{"parents":[{"commit":"df82f1963c26efc079bcdea7853fb402663177d6","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df82f1963c26efc079bcdea7853fb402663177d6"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-02 19:22:29.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-07-22 19:52:02.000000000","tz":-240},"subject":"sq? add new labeled counter metrics to cooperative token patch","message":"sq? add new labeled counter metrics to cooperative token patch\n\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nChange-Id: Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c491f0ff4a0b262b5f32c9d25eb53bd9a905fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c3c491f0ff4a0b262b5f32c9d25eb53bd9a905fd"}]},"branch":"refs/heads/master"},"6933fae4981fc12a5238666b2738ddeb43a9de9a":{"kind":"REWORK","_number":5,"created":"2025-07-22 21:58:50.000000000","uploader":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"ref":"refs/changes/65/952165/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/952165/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/952165/5"}}},"commit":{"parents":[{"commit":"df82f1963c26efc079bcdea7853fb402663177d6","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df82f1963c26efc079bcdea7853fb402663177d6"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-02 19:22:29.000000000","tz":-240},"committer":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-07-22 21:53:28.000000000","tz":-240},"subject":"sq? add new labeled counter metrics to cooperative token patch","message":"sq? add new labeled counter metrics to cooperative token patch\n\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nChange-Id: Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6933fae4981fc12a5238666b2738ddeb43a9de9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/6933fae4981fc12a5238666b2738ddeb43a9de9a"}]},"branch":"refs/heads/master"},"c7d0b7ccb5019cf6f6d81998e2b3eb8b84aa9895":{"kind":"REWORK","_number":6,"created":"2025-07-22 23:21:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/65/952165/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/65/952165/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/65/952165/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/65/952165/6"}}},"commit":{"parents":[{"commit":"df82f1963c26efc079bcdea7853fb402663177d6","subject":"proxy: use cooperative tokens to coalesce updating shard range requests into backend","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/df82f1963c26efc079bcdea7853fb402663177d6"}]}],"author":{"name":"Yan Xiao","email":"yanxiao@nvidia.com","date":"2025-06-02 19:22:29.000000000","tz":-240},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-22 23:21:27.000000000","tz":-420},"subject":"sq? add new labeled counter metrics to cooperative token patch","message":"sq? add new labeled counter metrics to cooperative token patch\n\nSigned-off-by: Yan Xiao \u003cyanxiao@nvidia.com\u003e\nChange-Id: Icc01e95ddd2189a6c336b0a9a8dbf0615d6e88e1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7d0b7ccb5019cf6f6d81998e2b3eb8b84aa9895"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7d0b7ccb5019cf6f6d81998e2b3eb8b84aa9895"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
