)]}'
{"id":"openstack%2Fswift~952260","triplet_id":"openstack%2Fswift~master~I73f1a03818157dc76938cbffd520c5c38941e2b6","project":"openstack/swift","branch":"master","topic":"p-timestamp-collision-patch","attention_set":{},"removed_from_attention_set":{"7233":{"account":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"last_update":"2026-06-25 11:58:14.000000000","reason":"Change was submitted"},"34930":{"account":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"last_update":"2026-06-25 11:58:14.000000000","reason":"Change was submitted"},"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2026-06-25 11:58:14.000000000","reason":"Change was submitted"},"36606":{"account":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"last_update":"2026-06-25 11:58:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I73f1a03818157dc76938cbffd520c5c38941e2b6","subject":"Common: a new decorator to log timing breakdown for slow requests.","status":"MERGED","created":"2025-06-10 15:08:36.000000000","updated":"2026-06-25 12:01:08.000000000","submitted":"2026-06-25 11:58:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":70,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"952260-p-timestamp-collision-patch","meta_rev_id":"42afafdefeed11fe73980d87d29d05d1b2cd17ca","_number":952260,"virtual_id_number":952260,"owner":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":0,"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-25 11:58:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":2,"date":"2026-05-06 10:50:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":2,"date":"2026-06-25 09:53:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"all":[{"value":0,"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"value":0,"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"value":1,"date":"2026-06-25 09:53:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"}],"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-10 16:24:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-10 16:26:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-16 15:29:55.000000000","updated_by":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"reviewer":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"state":"CC"},{"updated":"2025-07-03 01:51:05.000000000","updated_by":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"reviewer":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"state":"REVIEWER"},{"updated":"2025-07-04 17:48:33.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"CC"},{"updated":"2025-07-04 18:05:17.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"state":"REVIEWER"},{"updated":"2026-03-10 20:01:53.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2026-06-25 09:53:38.000000000","updated_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"7d80ab115127f3fdc073af9010a989bb99bb8c29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-10 15:08:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e46afe92c5136d95ebac6650e3cc30c99ca8844b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:24:26.000000000","message":"Patch Set 1:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1faaea60e424bfc85160b1432ca7c23\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2a267ab5679e44f4b4def05be36ac3f3 : FAILURE in 50m 49s","accounts_in_message":[],"_revision_number":1},{"id":"4248b902c15a3d672eb0a411d4b94fd1121acb6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 16:26:16.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5a7ea27a5a844a6b7920d92136e82a0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b545371a16424100b0592267877c0201 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/0ab479d0ca384c8587d684894ab73655 : SUCCESS in 1h 02m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/556133f36cfb4ff38352b77057bb42e7 : SUCCESS in 1h 03m 48s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ee7b5fd133ec4f0595436ac4eb5d5a71 : SUCCESS in 54m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8caea97550184b50a27d2cfaf6f54164 : SUCCESS in 1h 05m 02s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a7602977c748445fadd27fcaa60d1dfc : SUCCESS in 10m 35s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0bfc1c3f42c1464cb93b256fe8788410 : SUCCESS in 19m 44s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ab71122e16444b2bcf7708792d646bf : SUCCESS in 20m 26s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/9d7f7fb3ef674a9aad0e2067ce97d8e0 : SUCCESS in 5m 50s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b37f15c93984b5cbbf81c5d7452fa2d : FAILURE in 22m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/c94684daac964c969d977abb909a9299 : FAILURE in 22m 30s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/8ed85fd564e041deb2f0e6efa63b202b : FAILURE in 10m 16s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/68c093259bf44d40bb9cd31d22f9ee7e : FAILURE in 27m 42s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4204f414dc2e458481d1514ce42070ea : SUCCESS in 15m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6a52157ea2584b0a9ed5b38e5274a0d6 : SUCCESS in 16m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6d6d9747c454483ca50263fcd5547431 : SUCCESS in 18m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f728af6f96b64ce592e41d7febe59e5b : SUCCESS in 5m 46s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b68a44be3e234c05ac90a77194b6d819 : SUCCESS in 16m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/afab2c7247364ca6aee848ef3a24d39a : SUCCESS in 7m 38s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/af087fa6f2464d19a5f367611474e9d6 : SUCCESS in 1h 05m 41s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1a2d9703a82c408489811f149ff9f544 : SUCCESS in 47m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/51aec89218c34a9bb55dc78af0dbd1e8 : SUCCESS in 45m 54s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2adab946189a44b0a9dc29a3ac1dfdd9 : FAILURE in 20m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7372568003f5438a98ccbc16d65c47d3 : FAILURE in 2m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/657204ce5a3142d7855cd53e2bf8c748 : SUCCESS in 20m 49s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/90f0f519133044f9a28d1e13474ce160 : SUCCESS in 29m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ae74168cfc844d1a0d987fcf6a8cb48 : SUCCESS in 47m 52s","accounts_in_message":[],"_revision_number":1},{"id":"04a0fcc9d4394d2aa2ba659312768ad9571be19f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-10 17:34:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"964094bec9e8a597076f51a1ad03f49cbd934825","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-10 17:40:47.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"76ec247f01640aa71e23c9de7af032be2e9c941d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 19:19:44.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaf21046b73b4e18b570e0386d9b3dfe\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a43aa097450448b86ec7b7f4cbe05df : SUCCESS in 8m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/7a9d5550c1d547dea70fc204774b627a : SUCCESS in 57m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1abf5e9a979b4ac7972423550014cb4f : SUCCESS in 1h 01m 51s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4e659fca34584e4fb04fe4b267a7fe6d : SUCCESS in 41m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e2ee663a48e4fcc8c820ead7079d36a : SUCCESS in 1h 02m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/117e1b67960d4401851aa710b86b45d6 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d418aeafac304851be4e7c28bf8329c7 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1b6c8c22a59847e3af9f3705dce1525d : SUCCESS in 19m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/1f6167e4c2174eaab8598dc9c0dceb39 : SUCCESS in 5m 22s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/6ede4abb8e374ee4a082161caaea23f5 : SUCCESS in 22m 30s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0cab405fd1f74dee9060276b620acbe5 : SUCCESS in 22m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/12bdfda424a6497dace9dc96660d1726 : SUCCESS in 20m 31s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8201974561dd491084a20cdb75a36c25 : SUCCESS in 22m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/4473be1bee874e9d81f04954e0652497 : SUCCESS in 8m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6cf1d37a6b0643fb859a3816bdc3aa1a : SUCCESS in 14m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/028190e340c644c1b908e9fa2ffd2832 : SUCCESS in 13m 05s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6fefc90828fa47b6a43cb3d8c573eb06 : SUCCESS in 6m 00s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2befae11ca88475599b8e16046155ebf : SUCCESS in 15m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/17a1631d6381431ab58df904f916f996 : SUCCESS in 7m 28s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/08f4a2453fd540cc96f5556c8021875f : SUCCESS in 57m 34s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/53f7fb91356546bb9879cf5e367e8af1 : SUCCESS in 53m 30s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4d34dd91cf9f4571ac3c866446fd4f1e : SUCCESS in 40m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5035235543864a81899033f88e1d21a1 : SUCCESS in 19m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf2178ae2083406db113a94284cff129 : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/40a02cc22ffe4f379c3d8ec83d81153c : SUCCESS in 21m 31s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/6802b5e7c552474d859366095347959f : SUCCESS in 50m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aafd8dc00aeb498295687e110077e2d8 : SUCCESS in 1h 04m 12s","accounts_in_message":[],"_revision_number":3},{"id":"dd3b5ed578b8986a7341b8251db10570a30d9224","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-10 23:42:57.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2f7bf83ef50e40459486af2f51c58db4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a45cd806604c4c5d804e5e52af1b3f19 : SUCCESS in 1h 08m 02s","accounts_in_message":[],"_revision_number":3},{"id":"9de11839be757a0ad88241a2304785aa9aa8956e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-06-12 00:26:46.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"2fb7677fa0b87f9a3b3842b666514bbfc33fd512","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 01:12:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c0f2d9bf1ef485baf3b488b8eab4e7a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0eef34164eb6474cb7137547e5851d0c : SUCCESS in 43m 49s","accounts_in_message":[],"_revision_number":4},{"id":"027815f8919461684dcc57ed8551ae01409e62e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 01:39:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8f53e5986c14f85aca38a382dfd032b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1affd4293ad74017a2c634a0a561f56e : SUCCESS in 7m 20s\n- grenade https://zuul.opendev.org/t/openstack/build/0f9f1087937c4010bea1512ac7819965 : SUCCESS in 43m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17a2f35831494590a699d0219ef36ea6 : SUCCESS in 39m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/9bc7fa5825f446d8909feee561e24cc8 : SUCCESS in 50m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1609a2970b5d4e9d923cb2519eddc252 : SUCCESS in 1h 07m 05s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f284a8a3f53e4d9680e957007c69d575 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ebb840c85aa4e609ff11cca98c26fca : SUCCESS in 12m 35s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b7a1ff6d88ba4adc848545527851eebf : SUCCESS in 28m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/5a1a9b9b46e840008d9bed9856b043fb : SUCCESS in 4m 51s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a13b28c6b55843739050ba30006ae990 : SUCCESS in 24m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/a40de8e221d24cf9867b9affeea319c3 : SUCCESS in 10m 28s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/46d2a9204689400e80695fb116749b1d : SUCCESS in 38m 55s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/73d4bc7fb3a34ba4a5176144eea5c5ac : SUCCESS in 28m 04s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/82cb077a87f34eea832cdcf8620dfef4 : SUCCESS in 11m 46s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1660e8e3842842bebc524e2182527ed2 : SUCCESS in 14m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cebbc4fe667f495c8cdfa865bb7e076d : SUCCESS in 9m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/d63e0dc724b64d1aba6c9f1b6e424fd9 : SUCCESS in 5m 54s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ae943e0b1ff1436697fdb5edb40397a5 : SUCCESS in 15m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5e6918e2d5104a13a2486b0593c6cc2f : SUCCESS in 6m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/7f0f62f4bd114e8f8f5a0b18bd427ba3 : SUCCESS in 53m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4cbdbb37ff63429d86250cfd472ea6ef : SUCCESS in 46m 44s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8d72ecb3826c43da8e54feb1a00f23d0 : SUCCESS in 34m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/031ea953233d4218b5c8cf2907d15dd9 : SUCCESS in 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46c6bf400b254a868708b009cbf872ed : SUCCESS in 2m 54s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/6e5fc923ad624ad5b142b5bde0c3bc79 : SUCCESS in 18m 38s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/d1010956d988474b889e37a82df331ca : SUCCESS in 50m 31s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2eebec2f69084550bd5cc014571dba5d : SUCCESS in 46m 57s","accounts_in_message":[],"_revision_number":4},{"id":"92462e16ba606cb4de5c718d0c88bee7612e3c81","author":{"_account_id":36606,"name":"Yan Xiao","display_name":"Yan","email":"yanxiao@nvidia.com","username":"yanxiao"},"date":"2025-06-16 15:29:55.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7129f2a79cb1f1e4bb74b68cde3a30f28deefbcf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-02 21:03:41.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e2a8a86c87b00a26cd07c5ea4d739c0f98466394","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-02 21:06:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"92a8cfd76e76fb393bfc35779b589eca0f1a9612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 22:10:18.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d2520198ffb4330b3917b06e61fe14c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9913d88073e84378abf01e8a81a1d749 : SUCCESS in 10m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/9d11596a5d1f42ef8fe85bd269208912 : SUCCESS in 43m 50s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/15a8215a57214ca38fcd4ed2262a44a1 : SUCCESS in 1h 00m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a111e52c70b44bb966d63f385bbfc64 : SUCCESS in 41m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cc462927b94467ea624046717685c39 : SUCCESS in 1h 01m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a9c426d0684d475c9e38f152395d070b : SUCCESS in 17m 49s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/dd4ee8b4ca8e43dcb128d743f0d1906e : SUCCESS in 19m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/40798096aa5046c0ad66fc0476b98774 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ee8add9b564d4f5e9f9478017f0632fa : SUCCESS in 4m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f7d0103a535f42f9b1146b7db83a3f8f : SUCCESS in 15m 03s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1635815ed1704f1c8a8a4d4b85eb40af : SUCCESS in 13m 38s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/1d83a3f74e9e4cd08e48b06a28904016 : SUCCESS in 23m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3d3a07bf61a349babb4f647e4a643625 : SUCCESS in 14m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/97969e88288a453fa016ca24583da2f3 : SUCCESS in 9m 34s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3e7287511d9c4a5aac0989b64ef6f3ce : SUCCESS in 15m 13s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/160ad6fffdbd4e3a9ac88a7249a06f7e : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/57746f94c8a14bd7a56866a6d3ec11f9 : SUCCESS in 4m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/251c3899fef74d9d8c3398e51558e781 : SUCCESS in 10m 21s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cae9b09968104e6983daea6ea567c37b : SUCCESS in 9m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfc88c162b494329b4ba483726885c52 : SUCCESS in 56m 45s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a3b1db89d50047608f0171bbc103807c : SUCCESS in 50m 02s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0af7f219850e49e1ba4236ba54301ad5 : SUCCESS in 54m 45s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/d152f7ddf85a4fc8834e8260d87cd879 : SUCCESS in 7m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12281cf26574c438040c25bf8b7d286 : SUCCESS in 5m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cf700583af30452bae1664263f6eb589 : SUCCESS in 26m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/bcbe9c24ae704e72917d9a91bb30426b : SUCCESS in 52m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b0ad825c9bc340228d12b812254eefa2 : SUCCESS in 54m 13s","accounts_in_message":[],"_revision_number":5},{"id":"de94d68babff96e04ed4444511a04098fd7b52bb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 23:10:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e5ac2b930494fa99646a6946d8459ea\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64d5b958ee964ab9930de5e79408e1fb : SUCCESS in 38m 48s","accounts_in_message":[],"_revision_number":5},{"id":"5a4cb22b4fcae0a91557e1ba77f8ec1900903cbc","author":{"_account_id":7233,"name":"Matthew Oliver","email":"matt@oliver.net.au","username":"mattoliverau"},"date":"2025-07-03 01:51:05.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"f07670d18ff9310b4098d79d1e4e5aa72bd8aa06","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-03 19:20:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"bc6f1da524dbf948e05a8bf2a366437587b74610","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-03 19:20:42.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fd93336ca3e5c0ddee9553b03403bc07cb8e4334","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 20:20:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ac169b1803848e9af0ee69461afac6e\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2861dffd09fc49e08279297f3c9b4239 : SUCCESS in 58m 27s","accounts_in_message":[],"_revision_number":6},{"id":"7ac0f7e78c539f38f07d8936cfcc91897d260cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 20:30:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/344b50a8f54943b391f0abe345e90400\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7836355080f84051bda0ddc0953f24c0 : SUCCESS in 9m 35s\n- grenade https://zuul.opendev.org/t/openstack/build/ccb6cd27a1764fe69aefb22a3983edd9 : SUCCESS in 27m 48s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8b84c4c9fc57469584aedf828347e5aa : SUCCESS in 42m 27s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e3f7767151804be6bed0c1e9c31cfcd4 : SUCCESS in 57m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/adaa635acaac4bf09c1ad97ea603d571 : SUCCESS in 1h 00m 28s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6713f614386442b9dc2ab6f5e79d5c9 : SUCCESS in 18m 06s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/458583a5adba439e9de18c5f61952881 : SUCCESS in 16m 43s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71ac216f8bad4d54963e0b93986c37d2 : SUCCESS in 14m 51s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/0bb5ba5020b046ffb4e7388ef54fc1c9 : SUCCESS in 6m 42s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/316ba405711b4b4daceaae13af4dd964 : SUCCESS in 7m 56s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e775a3cd50164036b1789e39795f35f4 : SUCCESS in 7m 40s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/58047b6a1e814286bacd0fb650428a1e : SUCCESS in 18m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/5915e02ad95c4a49a9141a2827c35337 : SUCCESS in 10m 29s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8c1d53d15c784d24a58ddf221a1c2711 : SUCCESS in 16m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6be994c1ea764ba6b9ff098df2898900 : SUCCESS in 14m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4ad54cecf3cc4198af6dc5c6c28eb97f : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6732e36614924cb587cf9c3251f6990c : SUCCESS in 8m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/d7a5bed771044576aade0272e3af67c0 : SUCCESS in 12m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/723fb3192af24b91b6507a381cf5b0d4 : SUCCESS in 10m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ab86d90ec14c4868a477210831ef1dc0 : SUCCESS in 58m 53s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/74c6634d0c8945a4a27a9b8bb292ce2b : SUCCESS in 51m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/0b4b93fde14a4c6b80a102f5c2d1a218 : SUCCESS in 51m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a7833cccd93840c1a759bc322e010474 : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/840c8136e4ff4a6aa9b1ecbc503a6824 : SUCCESS in 5m 21s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/400304cd46f54589b555e716d8747676 : SUCCESS in 25m 55s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c69c8b0eb3f14124909b62907a02ee61 : SUCCESS in 59m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ee027d9c95a48b983ce6f589c662786 : SUCCESS in 51m 11s","accounts_in_message":[],"_revision_number":6},{"id":"33ddb814297928880d804d4e0e54f2b42e3ee48c","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-04 17:48:33.000000000","message":"Patch Set 6:\n\n(15 comments)","accounts_in_message":[],"_revision_number":6},{"id":"27913927c64f849704f0152a70c9dc9b0fb43394","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-04 18:05:17.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"d4b77ebfaba6667bf8b038d9788c9f672963ac5b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-08 03:18:34.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0d476aa4dd842305d0fd497fad638a1895f7d275","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-07-14 18:58:34.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"f46aa6bfa0ba18e5fc039a24c6e0814b779e55ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 20:09:19.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d6022e326824c3da47a0f89447a20c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74d5654d52ec40738013e48ff2087216 : SUCCESS in 11m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/b04b81036476419faa24553dfad004aa : SUCCESS in 59m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/429bada1965e4b3b91b7ed7f75508463 : SUCCESS in 1h 00m 39s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/e908042059604d68af62b767c41a625b : SUCCESS in 45m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3e2b39454fcb4ec49ae59b571f5fc5c8 : SUCCESS in 1h 02m 19s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e0a5ca56ca4645deaf7a37d346892b76 : SUCCESS in 18m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d3febdf388fb4263b42bccf67bb8db18 : SUCCESS in 18m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/77b51f8ac72544b382951ca832b39a99 : SUCCESS in 23m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f8dfdc9ed18f49cf91d80430792497b0 : SUCCESS in 7m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b70b912ce1e64965876889c07ca99daf : SUCCESS in 20m 44s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4f6bc7e22cf84a8a8fbb8d00924f79aa : SUCCESS in 19m 39s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a1ac5ae8c0184f9db75572abd1bf9e53 : SUCCESS in 23m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e3a0a245732b436fa076efe0bc8f1e56 : SUCCESS in 22m 57s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/55e8746471db454f92030d0b0466f25c : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/39afda2138f24e69849ec44ef31d2067 : SUCCESS in 13m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ed12283025994ff5855b2d0c56175b8d : SUCCESS in 14m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b53f3e5b7a9b438da25861d8424dec33 : SUCCESS in 6m 08s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/41747a14c1b142aba3ae14015483813d : SUCCESS in 18m 29s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/58e6b25806d443c39b0fe356118a9745 : SUCCESS in 8m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/196bb5212d784c1490999cb36271a709 : SUCCESS in 55m 09s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e0649d2c4a7c40e8ba1ec7778b22cfa3 : SUCCESS in 48m 09s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/92fea11e394f4e1d992296c8357757e1 : SUCCESS in 52m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/57fbe02aae0f435e98d710653f8437da : SUCCESS in 17m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882d3a27e4e84c7082943574825db869 : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/0a9679ddb8b742deaced28a655b27d9a : FAILURE in 37m 26s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/59107d7ac27a404ba6d7724267178046 : SUCCESS in 52m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/420d9c4cc79a47819bbfc9858f35eb16 : SUCCESS in 29m 41s","accounts_in_message":[],"_revision_number":7},{"id":"c303a31ea8698835714422e413b1a429e894c138","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-14 20:10:52.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2d29959a2df54e518f17d626c9ea8037\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/49cd32042dcf444fa4ca96f68a77a8a4 : SUCCESS in 1h 08m 59s","accounts_in_message":[],"_revision_number":7},{"id":"9cef0c9733888eb33427cd2b711b4e02d5e6a35f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:28:06.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"1178f170b95b6ba5ef7e709305706eca876f65b2","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-20 05:30:48.000000000","message":"Patch Set 8:\n\n(15 comments)","accounts_in_message":[],"_revision_number":8},{"id":"7f205116fc4b15d9586e16c729d2f0d3e3fb382c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 06:53:36.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d86b4e1149c6450c95858749a09fadd9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9edad2570e4c7bad3c787feeb7948f : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c7c9cb3d9b9b4947b6726b30414e8cf7 : SUCCESS in 44m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/513f70932b3d47d9a3ea13d9b63f7f24 : SUCCESS in 45m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/21824bc31542438f800ff1cfbd32f5f9 : SUCCESS in 52m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/695cf8ca7cd14b41922cc171fb3b2da6 : SUCCESS in 1h 07m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d006b9d859d2449e9ef3278297058b1d : SUCCESS in 20m 50s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d245edb73c24450280dae49fb103901a : SUCCESS in 20m 27s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f1b194692ff64fa6935e6ae1aedfd628 : SUCCESS in 21m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/33e09019e4914ecfb6b052c77d0c7306 : SUCCESS in 4m 23s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/fac0314378ec4fd2b45bc88714628a02 : SUCCESS in 29m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/6e01e5341a534d9b900e656b1d40a18b : SUCCESS in 22m 32s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/d48e8e4bcb954eabaae83656bf049c7e : SUCCESS in 30m 09s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d1402a4bb6d940d18e78298db0926778 : SUCCESS in 24m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/995dc22dfc98434e95b1a4e186748926 : SUCCESS in 9m 08s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8ea80f0afc4844ce8f5ef132744c3be0 : SUCCESS in 8m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/8188715b8ae84351b7eff234db838873 : SUCCESS in 16m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/01ed479c039345c2a349563f73cc25a0 : SUCCESS in 6m 59s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/1bc0e3b721d94372809f4e60700da376 : SUCCESS in 19m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b527a6ce28214170a4f7f28937bd8e76 : SUCCESS in 8m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9b0db2368bcc4612873c0325ab495c07 : SUCCESS in 50m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/d77b183b1d4c4c9d846a105a890383a1 : SUCCESS in 48m 59s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ba7191e50f0043f3b3d9572e9c884c7e : SUCCESS in 22m 30s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8363c2a30f2c4e499df8ce3494b3036c : SUCCESS in 28m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d0f1e87d87b4f8b8d11610b93130c71 : SUCCESS in 2m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/e18ad9d347604f1b9635fb66b144af05 : SUCCESS in 17m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9751b137e3ce4438897531a6be5bc0f5 : SUCCESS in 40m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/07c833e4125744ba89a52e329892270e : SUCCESS in 1h 00m 26s","accounts_in_message":[],"_revision_number":8},{"id":"39488b3fa7e62f42cff62d50c99b7152652276a9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 08:51:22.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/057eb9537c6745fe9cfc70e3d8bd14dd\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c031f48ecc341d4a70e7cb3888d174d : SUCCESS in 1h 58m 49s","accounts_in_message":[],"_revision_number":8},{"id":"2a7dcbf23bccf000f1f7bbd38edc0ed1435e14f7","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-21 17:20:36.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ebd4d7de5c10ad04bd1b8db2018f9a3cfb386259","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-22 05:20:44.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ccc060d3daab125d83d66fe243907dbd527d9879","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-22 05:21:51.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"3fcef6791790aaa72629d10643ae1d18213d4a36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 06:30:21.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/beda0b7ba1d04c4ca05ba330184fd14b\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/adbed21e982d4c44b9390ec8252a45a3 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/514cbe318a5e46bc82421783b7a48fde : SUCCESS in 42m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ccc4e18414c3462dac23f6ecfe456f09 : SUCCESS in 31m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/52afc627abbe4327b2e62f608907afab : SUCCESS in 53m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98ff8e9a04a946608f5741400b7391af : SUCCESS in 1h 03m 36s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7fab0115d58a46f38eb97f0ba1774e93 : SUCCESS in 13m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/efbec4fed7234c938917aeb91513af45 : SUCCESS in 15m 57s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2d69966da514b4f8b0e04a281a79b74 : SUCCESS in 13m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/677546c0587b415eb93ae158cbef3cf8 : SUCCESS in 6m 05s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/2b09ec7b0a2a4c4aa8e1fe157abd5318 : SUCCESS in 28m 23s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/2acd898e3cf041f08930d4d36e973807 : SUCCESS in 21m 33s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/66611ab050944e6e959eb5fb58dbcc3b : SUCCESS in 20m 57s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/91a36c8bff8649058e9aedb81841c06d : SUCCESS in 32m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/80e5fd7ed51c413ab8df0fa86062bff7 : SUCCESS in 14m 49s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/cb5ff3a4b9134da8a0ffe80fc227aafa : SUCCESS in 8m 53s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c9b33fc64ce247f79d5e0994db5cddbe : SUCCESS in 14m 43s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5aab5b91066a4b8581588c64f717fd81 : SUCCESS in 4m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/dbb3a094c0394f1488cd25555c212ba4 : SUCCESS in 8m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9041d6297e684b1f80e65c5b163f2127 : SUCCESS in 8m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/abb4373acccb47088fcfedcc8d9ab910 : SUCCESS in 58m 23s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1284582a4fba4748995723b728516e9f : SUCCESS in 50m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/88e4ed84e8f8473cb4abc64caab36d53 : SUCCESS in 54m 04s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/77fe9395c4154103929fc85083834fee : SUCCESS in 27m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4685c3f3e594177b5364f19ed872b9d : SUCCESS in 4m 35s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/650d815e135142fc88bc68cc8186738c : SUCCESS in 19m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/15eb1ede2a354ec099cba6c537b247e9 : SUCCESS in 50m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e31c10f689645ff8734f23e405f76f8 : SUCCESS in 35m 53s","accounts_in_message":[],"_revision_number":9},{"id":"b52196646ab6684e81ddeec000bd703871115ddc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-22 07:01:28.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4bed7336311421a9fdf6f4a83e547c6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e903b1ea987a4d5fa0a6d049e1772660 : SUCCESS in 1h 37m 51s","accounts_in_message":[],"_revision_number":9},{"id":"3eeb7c2429f52a00d08c51538ef74f660e13f775","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-22 09:50:54.000000000","message":"Patch Set 9: Code-Review+2\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"14f9fb8835a657e82f6ccebbadaa5bb31d80efe6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-22 09:52:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"6d74a70d2d87793f80a5f83895d98317e4bc71c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-23 00:04:17.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"92b142e3797f668655a045485de00164a4e9c617","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2025-08-23 00:04:27.000000000","message":"Patch Set 9:\n\n(7 comments)","accounts_in_message":[],"_revision_number":9},{"id":"4631ef8df6e6ebb8d8a2f8876a0a9d5f36810c7f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 01:06:49.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73c50f6676b04b35839a55b0d2f761cc\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/dcc700ba61e648e7894ef4fe0ad0566d : SUCCESS in 1h 00m 12s","accounts_in_message":[],"_revision_number":10},{"id":"044df214c3630affeff3745f40e8aba9850d7e0e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-23 01:12:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71896ce27c984373a94058540fad45a7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/892d08318df3419188fb79df75e023e5 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/d89893269d924f57b4a380039d0d08f7 : SUCCESS in 28m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/10b9014e34aa480d93fdd104d521c432 : SUCCESS in 1h 03m 02s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f3328c4ca6b84d3ba5ab39e1886d6901 : SUCCESS in 37m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19f58714722f4319a449c8d717fa2a7b : SUCCESS in 1h 00m 51s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be4e740b4f954e8ebc505124323eeefd : SUCCESS in 15m 36s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a01d5235d2b34476b50bf3a030b9a817 : SUCCESS in 21m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f981a4e8e647417c88a4137e5c8c6085 : SUCCESS in 11m 10s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/38b9e19587f1405892ceabfdc1ebd113 : SUCCESS in 5m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f13e01e0dcea4a89ac65d1db646d4a58 : SUCCESS in 30m 20s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/8a8a2feba7fd4c80ba6a7a8a13f878c0 : SUCCESS in 10m 34s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/31169b91b02940f493e1587d8a36a4d8 : SUCCESS in 21m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f0b6b52639924edbbefba2dc15d263ff : SUCCESS in 11m 31s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/09851ca03a69422983f8e75d5c7faeac : SUCCESS in 15m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/72412a3c66034e4aac9c4542e7e7f332 : SUCCESS in 8m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/895b1facf22e48c189f020b2d5319fe8 : SUCCESS in 8m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/703366b28ea844ee926bea9939b49cab : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e97dfd5b2bd2402984b5315dde740d00 : SUCCESS in 9m 37s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/29dbd714a43c4a5ca64be0a19adef3a9 : SUCCESS in 8m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8242d1eac9bb44f1887fb1d1286299fa : SUCCESS in 55m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/5e05024b7fbf421c9f8afb12825c5005 : SUCCESS in 48m 05s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a086a2d739a34f0283b5c157ff8c73a9 : SUCCESS in 33m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5c38c7d557fb4ce1970d10000af954f8 : SUCCESS in 20m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc09c0b4c32462f858c8a9ba3ea68ad : SUCCESS in 4m 29s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d559f406c8044129bf13da2eadd6e247 : SUCCESS in 15m 19s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8d326c7b1f844278c90f14c1414d2ed : SUCCESS in 51m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ffbf0a652074a2b93f83039dfd6652b : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":10},{"id":"66171b681483fc902aa7382f9c124f78bd405ace","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-09-01 13:34:23.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"8069a318ca8e0601fc8fed3a377b8236933d014c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-02-10 04:43:10.000000000","message":"Uploaded patch set 11: Patch Set 10 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"c427ee0ebcbc0d8fa7449a4ac9c6e73a0b5dc89a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:15:29.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77db298e47894107aa8848dc64b61a11\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/64037feeef9b45e9a3b77f995585d364 : SUCCESS in 29m 29s","accounts_in_message":[],"_revision_number":11},{"id":"c35ddc0e3842fc068a3d4551684c483aca392992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-10 05:48:30.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ce26f0dfa9cc4a1cb5481473d8372c71\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68416112b592471d952c104a23986f9f : SUCCESS in 8m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/548961cdd2744c558764731cfa61556c : FAILURE in 6m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/759cdf2741cf4732865edacce6d8d50a : FAILURE in 7m 59s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/d068a0b5845c44869385f8cdb1b878cb : SUCCESS in 31m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e18d9ad5f894dc0825eab96b745743c : FAILURE in 36m 33s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/727613e6ceaa4e689584259de6d94ab7 : SUCCESS in 17m 22s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/be5b63bd98ae4c94ace72272dce0fd6b : SUCCESS in 10m 47s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8c62319f4c56420da1dbff49770f5f2c : SUCCESS in 10m 44s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/fcc9f60d184d416cb5dc5d43b99b72cc : FAILURE in 4m 31s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/086b09b045c043fa80a7a8ad4206df4c : SUCCESS in 10m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61205468c71448bbb213bd948a99ed91 : SUCCESS in 14m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/dd0738499c2e493c939b2b480be3768f : SUCCESS in 27m 59s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/093053b686734a249612bae9551b2380 : SUCCESS in 27m 35s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c83ac3c7a795405cbd065c9166e750ed : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/23c556be7d47477898826e22e95c47ca : SUCCESS in 12m 49s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d5afc482f27c42608df7bf5abb26abcb : SUCCESS in 9m 14s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b1870ff11f3a44e988d2d859127fc83a : SUCCESS in 5m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/71d89b912d0140dc91a4428a49cd3505 : SUCCESS in 18m 56s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/cfa016b2568740dd820e885420bd43bb : SUCCESS in 10m 15s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee3c11ee584043f1b561775c799eca67 : SUCCESS in 59m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2c2448ef1d45498dae21f9fdf0037f4f : SUCCESS in 25m 11s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/7ac56320be9445f7ba1e82c62930f25b : SUCCESS in 55m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bd71459f3bef40e4a4537a06a02bf3f3 : SUCCESS in 20m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a09473de9e244141a202ec8fb5918c2e : SUCCESS in 3m 47s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/01500df3b5ec4e2f954e260a892eb840 : SUCCESS in 18m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b7ce43e3a59d4354b0f53cb450530aeb : FAILURE in 43m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1524c293ceea4229952de8d345daf7e3 : SUCCESS in 43m 03s","accounts_in_message":[],"_revision_number":11},{"id":"9e5cd56f7a059330103d9ed475cf4501068f398b","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-02-13 23:43:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"42dc00399eccbc9e754fb3650e39b5e31dd7b824","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 00:15:46.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9686edf9de244576b0161f339e2349a7\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/892fafb267994a148c2d78ace18eace4 : SUCCESS in 30m 29s","accounts_in_message":[],"_revision_number":11},{"id":"f32be9a105111967f3b14fb1d45ae24f9ef358db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-14 00:51:44.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58ba4a12a0f24221bdf47f993bd6231d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0baddbf9ef584d96a9bf2cbb138e1252 : SUCCESS in 9m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/e269880f723e4da9877e5b1530f16e03 : SUCCESS in 40m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/80087d58f050464eb2e4c3bf110da8e2 : SUCCESS in 1h 01m 22s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/343ef1661d2d4223b5d11d4a8e6637d0 : SUCCESS in 53m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/19fca704c44542219bae1ba5f9ae5d88 : SUCCESS in 59m 37s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ae58b7f622904a61bfc4b7833124413e : SUCCESS in 22m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2bcc2185b3e4758aaada9898680c37a : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ea8c87538d3549238323f080bb681358 : SUCCESS in 20m 19s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/a4e1bce00e904e3d9d8b4280fbb1ec98 : SUCCESS in 4m 35s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/c382591f40b74444936277366eee5c8a : SUCCESS in 24m 31s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/61f3dc166a7a41e6a79f65f474a8900e : SUCCESS in 13m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/c666a0e5a9c940cd81abd93062a195c6 : SUCCESS in 21m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/19313d2294dc4f8394f563d97b5acb63 : SUCCESS in 27m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/33ff55c72ba44a2a8777f583731fbecf : SUCCESS in 14m 31s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ce56d6e20c054f3d98dc6b703d6b2605 : SUCCESS in 14m 37s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/4dc490cc9f244042b617ef12774d1cdb : SUCCESS in 17m 24s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/8ad30347c8924b13962c9fac4bbb7220 : SUCCESS in 5m 39s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/49c0f0debd61439a976c0132f8ce3462 : SUCCESS in 15m 17s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e1d66257e1e24b08950368d6a7943d79 : SUCCESS in 7m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31a723d38c9845398b16d9358ae8479e : SUCCESS in 56m 50s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/fcc35f649f4d4fce8cbf43f13f7ff660 : SUCCESS in 56m 35s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1f7e3e3bfe3d49d5b32ac985d61649e3 : SUCCESS in 50m 28s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/78f48232d79f4451ae64fca54d96f543 : SUCCESS in 11m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8b1b73130a245a9929a950e3875f945 : SUCCESS in 3m 39s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2ec3603e22fb4e39989498bd976b334e : SUCCESS in 8m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5f4f57a69a84fa5a10c28134848c5af : SUCCESS in 57m 33s","accounts_in_message":[],"_revision_number":11},{"id":"39e1e69185ca0c9967144d2b0ede82c9bd8c38de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"date":"2026-03-03 05:39:12.000000000","message":"Uploaded patch set 12: Patch Set 11 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"32d4d1bed8ef83669819457d7f6512d2c4fb7b36","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:10:15.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0fdacb05a2fa46d48d1e9735646ca4b1\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e425bf5b5663468585f97e72fa390a30 : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":12},{"id":"0b47d60f2cb37d3477f70ac8f4c8231a9fac66f6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 06:57:29.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ba400cb83d4475a8d46ff675c146c1c\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c23e7a8d40ca40f5b3443f17f8f32b8f : SUCCESS in 4m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/d6fa142f67db4a39a713ca8881ebfd0d : SUCCESS in 31m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4ea0ad3021574309ad669aac1d877a39 : SUCCESS in 1h 12m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/547e0def71ca4573bda6e892545a6ed3 : SUCCESS in 1h 02m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b0d94a065ce94d04b7447f8cb8009dc2 : SUCCESS in 37m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f7f52688594a47a8a78d68b2af65b4c6 : SUCCESS in 11m 16s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0293b838a60a43c292e8c0cb7bfd524a : SUCCESS in 15m 33s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/839997b9d5624c95ac61a2196a5488c8 : SUCCESS in 11m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/f9b0bef6c56f48b9880e73a8884602df : SUCCESS in 5m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/eed017f3bdd64dddb1f0c9472ab15913 : SUCCESS in 18m 42s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/0b830073d94144ddb4a66ed4f5917e87 : SUCCESS in 20m 54s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/89af086ffd77482cbe5fae701d8bcc70 : SUCCESS in 10m 25s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7be0b902769f4101b81b1e58c292139d : SUCCESS in 29m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c3dfb013abaf41538368b431d49ad9ac : SUCCESS in 16m 30s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/39e8c324b475411eb5e23afb448aa224 : SUCCESS in 11m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/59935a3aa3ad463da9efb01f55f3cc5d : SUCCESS in 10m 16s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2791989e5cee491c8310d148dfa80c22 : SUCCESS in 3m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e488af139cd141349cedb1e39e1ffb99 : SUCCESS in 16m 31s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/98725ce354c84eb3b76e5e8b11d736bd : SUCCESS in 7m 30s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/12b1fa90e2fc4144a87cde200770433c : SUCCESS in 37m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c115dea8d9774ea88f19bb8b5cbbed94 : SUCCESS in 25m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e4460d5a4b294279bdbbac905db28993 : SUCCESS in 57m 48s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/3eaea645771d46c4a70257d0ef20e631 : SUCCESS in 28m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c7fcabe132c3434eaba72920fd648ccf : SUCCESS in 3m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9129de8efba34467acbea91fab78d8db : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11e61b6f44b240a2a25c19ff4600961f : SUCCESS in 1h 02m 12s","accounts_in_message":[],"_revision_number":12},{"id":"49d3063d71a2d75667c60d4d56bb8e00efe3b92b","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2026-03-10 20:01:53.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"cf80bf2b81a8938e1b0dfa1489127cdcb900d5a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 13:35:59.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"47e8926071c0f92bc97f9094bbba6a105ad13706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:57:57.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7a93d6e3241744dd9b6faddb276bc853\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c33837cfbb3345b99abc84b07321a757 : SUCCESS in 10m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f4f0afe360854cd29b713bff77a2ff3b : SUCCESS in 56m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4b1978f125c54af38af59d06006b42d2 : SUCCESS in 1h 02m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/82a787e835d34c7583f3464de4d30743 : SUCCESS in 36m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21961e43f81440de9c75a6cbd819882d : SUCCESS in 1h 05m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f726af38de140f180f18c7b5a33f35b : SUCCESS in 20m 01s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e77b3978c6c54b44b6086003e9fd8156 : SUCCESS in 10m 51s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5ecf1eb4ef024c82a29c8290231dd212 : SUCCESS in 11m 30s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/8941ef82c4624b00bb55ec3a4c2f6a36 : SUCCESS in 7m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/e365e01787404a409a131fdd8fdd4171 : SUCCESS in 25m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/3e25a611be9b480cb16aec25a8d53cd9 : SUCCESS in 22m 33s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7f5e23509ce040699f49331eaea6da42 : SUCCESS in 25m 09s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f9b1521e7f524b4bb07a69edfa353bb3 : SUCCESS in 21m 05s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/2d94756cdc074a15b43fae016fe466e1 : SUCCESS in 9m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/624cd2975c9a48fe8f2d3848d4310c68 : SUCCESS in 14m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/21a67719f245483397e925a575b8a7c5 : SUCCESS in 8m 47s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cd1a4aed56784f708d181d22fa92a8d6 : SUCCESS in 6m 15s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/81a21876059a4a61aece192279ec6cd8 : SUCCESS in 16m 49s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/df8183560e4c4f4fa67a24d4e2944e2d : SUCCESS in 7m 47s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9ca18b00f3f44931922c9f66798179b9 : SUCCESS in 1h 03m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2ba06450a4374a5ca51beb5948b17408 : SUCCESS in 56m 19s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c4c80f31b0634e27940def9204c5f6f5 : SUCCESS in 51m 17s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f2b1e455a4c946ca9786434623046a57 : SUCCESS in 31m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4dd3a17d793d4f618a65d7477744fcf5 : SUCCESS in 4m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/df72ee3fc2c642cb8e66b56596d378d6 : SUCCESS in 19m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c72f264831044ac1a34b93f2570f3eee : SUCCESS in 1h 01m 59s","accounts_in_message":[],"_revision_number":13},{"id":"37ffdff01136e7b1e8d9b0fe0d16bc58e538d378","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-18 15:18:47.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"eb256f0fd662a9e0b61c3cec51aecda271ed888d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:35:36.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac25c09d1c6146a882d1eb0a6861ee6d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6449396b7b1c49c0af9bcd099f894f78 : SUCCESS in 9m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/6ce92bf0ae7e4ced9d0fa7c802bf30fc : SUCCESS in 33m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e1616b8e574b41c5a856c02d7987750e : SUCCESS in 1h 01m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/86f7030eac06414ca2a336ea9f11cafa : SUCCESS in 37m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cad8b0ebf4bc49c59be4bcc57587ec1f : SUCCESS in 1h 04m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/edce69acdedb4e11acb3244175a6f480 : SUCCESS in 11m 17s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c4c04e36d6574c0d830abc8fc3495950 : SUCCESS in 20m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed268cbd24314773994063351f028139 : SUCCESS in 15m 35s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e1475c51614543a09f36cd040567c494 : SUCCESS in 7m 38s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3dfc338cbd8f4d24968c568702916f87 : SUCCESS in 13m 51s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/9c94991f674d48a29eb9be254d032a45 : SUCCESS in 24m 21s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/01490a5652c1484695b93bf1eb50960f : SUCCESS in 25m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/b90a00ad729641c39d830a098bb992f3 : SUCCESS in 21m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d29be13cbcd84d27b64bb6346879f812 : SUCCESS in 8m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0830626444c34cb7a676b7e82311ea70 : SUCCESS in 15m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a628f9db4a304f1e80cce690938b490d : SUCCESS in 15m 33s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a746546798974a788da665a4ac2cc21f : SUCCESS in 7m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/cf7162ea9932417bb14c3bca0b26790c : SUCCESS in 9m 35s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/319fd73d06614bfebedec441054dcb38 : SUCCESS in 5m 41s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9804ebfb726d4b73a0d7d23deb629d4b : SUCCESS in 43m 24s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/740f2fa823794a9ea1791184fcfb02f2 : SUCCESS in 57m 34s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97643a93683b41acb69723ae32f8af0f : SUCCESS in 50m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b48068d6a8a843b49f4cac51c44c7ca2 : SUCCESS in 28m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4cf5bbec5c941f3b3348dc3c43091ea : SUCCESS in 5m 07s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/21c56c84e02e453a88d74288687dfc16 : SUCCESS in 10m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/74d014c825b54e64908e0e748e952eae : SUCCESS in 1h 04m 25s","accounts_in_message":[],"_revision_number":14},{"id":"b73f09ab2763ac24b26b85157103b56d584699e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:23:49.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1040026fe3854bc6902d34b9df2e7553\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/305b2aede188458f9bd1bb97f0ca6599 : SUCCESS in 32m 49s","accounts_in_message":[],"_revision_number":14},{"id":"a814b4638d6b2e74675e16d987c8eb4ea8ab70b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-24 17:58:55.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"14da48a02a6177412fc925c5b6fa93a81f18a9e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:13:57.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3e0d57154bc4a05a7b1c060afe2723f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7f6451791094f1cba8965fb921b712f : SUCCESS in 11m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9942ba0dd2c547a7af5dc35499e755b5 : SUCCESS in 1h 00m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cad33e594c734ef68653fb3558204c30 : SUCCESS in 1h 05m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4154ee8bcb0b466c9eacdeb2711dce3a : SUCCESS in 58m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/57811941b19143b9ab65a5be0dcf0423 : SUCCESS in 1h 02m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ba22d4992a374469b782de1f028e6c19 : SUCCESS in 19m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8a01baa0164c41ef9597d3cdba78c306 : SUCCESS in 20m 12s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/858784430da94451955813e85bb362a6 : SUCCESS in 21m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/21430674782846f5a0ea4c253803b78e : SUCCESS in 6m 02s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3c6328344f1441dc8534442848f89692 : SUCCESS in 25m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/966c61fda5904107bc95fdc423904bdd : SUCCESS in 9m 53s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/b31296e612184337855a18167d1a096e : SUCCESS in 21m 00s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/f1a45b0f8a20451499758d60750766c7 : SUCCESS in 20m 00s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/04e720a943fd464bba50c548cd6c0b3b : SUCCESS in 16m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/374e604089f2476f9331803cb2987939 : SUCCESS in 14m 45s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/459953c28922404f849cfe0a72645a01 : SUCCESS in 15m 04s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/ce9db7255f974989841dd0dbe90d7616 : SUCCESS in 8m 06s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/56d68451f9014a52bb84fc409c307046 : SUCCESS in 15m 32s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9dc2877f927345d38d7047a7cf7930df : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/3406f0e7c0d14844a715e1cee09146c8 : SUCCESS in 59m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/2aa8d1f7a7fd43a0a30a1cd1aae05763 : SUCCESS in 23m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1cff46cb7acc4ae683819ee6f832d760 : SUCCESS in 53m 40s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8cf4100e9c134669866011b5b9285fb7 : SUCCESS in 24m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb2fae8878734d739c270e441bc0e53c : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/4918fc18753e49059e3bc26187fd24aa : SUCCESS in 18m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/08f69bf257f148aab08e1c08682a4e53 : SUCCESS in 46m 28s","accounts_in_message":[],"_revision_number":15},{"id":"f2cff10d25ac543cd39d1249325c8394158b4a59","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-24 19:35:49.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d6297496461488f97a673cfbdc2c345\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4bf82d054f694c24b7cf09fddfcc6f24 : SUCCESS in 1h 05m 02s","accounts_in_message":[],"_revision_number":15},{"id":"8446ef93e5552784f92c0a03bf642571aa608932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-25 16:30:13.000000000","message":"Uploaded patch set 16: Patch Set 15 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"999e48470f170d876513b6f9cab57a96d5e1f857","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 17:43:48.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f132801a9a744038b5efb6b94147791a\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ff23cbaf24740828cb671599d13a493 : SUCCESS in 10m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/10771d8255f344939f1d3ac4f10fe709 : SUCCESS in 1h 06m 46s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d190d54a2f70493cbb3921535fcb9736 : SUCCESS in 1h 02m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/fdae825458fb4a468ea4a06375e7560f : SUCCESS in 55m 25s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b03f8ad98c4a40a4be203283e7fdd17c : SUCCESS in 37m 24s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0dc40345cab441c8bcdddc4fe8c11b6a : SUCCESS in 11m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/38e59832dea84c63ba39b8e3a11fd1ef : SUCCESS in 22m 32s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ce3f3b9e30446b99839e6cea1b4bea7 : SUCCESS in 17m 50s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7bbf6c3a7e4e4c6eb3ec4ee78ff1b3fa : SUCCESS in 8m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/15fa6e62d9a1487bb463fb1efe8a2d21 : SUCCESS in 24m 27s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dc444cb5ab584cf1b5e2867d21a818ce : SUCCESS in 23m 44s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/ac1ef4f01e99403fa9b029dc2d1aecdf : SUCCESS in 32m 16s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/2058d0125a2944c99340e2c2fb345e77 : SUCCESS in 10m 28s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/e9f729eebc604b78a9e64df1d4f848c3 : SUCCESS in 9m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/86e588053f3242a8b21da304021ad204 : SUCCESS in 9m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c91f06677b21490d9da51034ca1672e4 : SUCCESS in 10m 36s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/06844272285d4463aeed177b84e1cf6d : SUCCESS in 7m 33s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/24b351674f694911a68ff5d56aa2c537 : SUCCESS in 17m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f89573923f04404daecd33264518d7af : SUCCESS in 9m 27s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ad9d812676e4f0dbfe45ef74c51fd80 : SUCCESS in 35m 22s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/641afa0f6b82463d9af79347316e41e8 : SUCCESS in 25m 37s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/e7f06602353a40458f3763651dc8741c : SUCCESS in 40m 56s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/ed56d2db1a8249cbaf9d595cfbbfcda8 : SUCCESS in 26m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06a4b8c0faf042e6a845a58ce43a024c : SUCCESS in 2m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/24de7cbb70d14f9ca56b1310b8841790 : SUCCESS in 25m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba373510f77745ac9297837c4a9ab1f1 : SUCCESS in 1h 06m 23s","accounts_in_message":[],"_revision_number":16},{"id":"3ea5fc3f4b099110b8c28c11ee61b14cf9ecac47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-25 19:29:14.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38994ac4a87d467daf2851216cc4ee72\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/804c2ef20e014af68dbeab7715c586e5 : SUCCESS in 37m 06s","accounts_in_message":[],"_revision_number":16},{"id":"fac67975b8e9795f0ab6b7cbc0f089c204978aa5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-03-31 14:54:16.000000000","message":"Uploaded patch set 17: Patch Set 16 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"223ca8c6c0f8bc586d81ec421c31fc2d6a78051d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 16:09:12.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/645daecc5490494e98a866b968430501\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f736b40dc5ba49ff8be4170b3240aa80 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/6e6155aa6c3847f69d017e5a5123698b : SUCCESS in 52m 03s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6798f475548e4cf68f6d1736bc7b359b : SUCCESS in 1h 05m 19s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/589d8c5e1c6a41f7b0ded0c3dd4ed6da : SUCCESS in 37m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/320732fb78044147931fe76baf6ac8a9 : SUCCESS in 1h 04m 45s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fae3531b927457c90cc6aef60917481 : SUCCESS in 11m 09s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/98ac38bb181a4e63b387f3ceed343585 : SUCCESS in 20m 18s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/56689694e7434757a8027fb10e4678f4 : SUCCESS in 20m 01s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e8c2e6ea9a994bbb8cf47e2bb2aa49a8 : SUCCESS in 6m 54s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/f56dcb9791e843f1851238fd92a0a98b : SUCCESS in 27m 39s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/5e2df62983774ecdb1fd5242c2f2a9cf : SUCCESS in 20m 41s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d1ae1af4b79847e7bbe5c98aeaa19a85 : SUCCESS in 31m 04s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/7a9a68336b574b68a56c6f65f4fdd4da : SUCCESS in 20m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/a03f3823c93b4e9b8232cc9330aa90fb : SUCCESS in 14m 36s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/87dce6453f93457cbe03a0f922996798 : SUCCESS in 16m 14s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f9199d7dd0074106aec391cc8a6c8076 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2e1540b9e16d40f0a4b21026899960e5 : SUCCESS in 6m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/200e56f36241427ca33cfa19f16e453f : SUCCESS in 11m 18s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/af8a91bdf0b24b8da2b1b3b88049f845 : SUCCESS in 8m 56s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e3ef95613a5a4d498f6ca5fd766e8632 : SUCCESS in 58m 29s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c0fa228cd7bd429790d3f7f9221069bb : SUCCESS in 36m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/b577ab3a91d341ebbddca3864ad39ef0 : SUCCESS in 36m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/97993cf3bee545cda831b6e2a5a5cddc : SUCCESS in 29m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89f54a4217314a04a826af0fe4fd6244 : SUCCESS in 3m 04s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/019482fca86e439c9a23ca842f7a74f4 : SUCCESS in 19m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/18300607e42b4d3691d264ff619cac71 : SUCCESS in 54m 59s","accounts_in_message":[],"_revision_number":17},{"id":"1ce1b8ec49a400e643f4fce8ddd6d20f6e615ebf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 17:16:50.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1692d30308734f52953354502e1abd5a\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6a25aadca1d0412d95d69d5fdd2b3769 : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":17},{"id":"4fc58a37e37a366791abb24b352579d0467e9ef9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-09 09:21:56.000000000","message":"Uploaded patch set 18: Patch Set 17 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"98ab102fbcc46f47bf9658439ec80c533069dff6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:37:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c08c7d00bd542bf8572d4d52c2f50ee\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23038e9bc9f543789965727a03061748 : SUCCESS in 10m 26s\n- grenade https://zuul.opendev.org/t/openstack/build/0b8dd695112641558aa29a955eb2c679 : SUCCESS in 1h 10m 26s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ead70d085c79416cb2bf5f76037bbc3d : SUCCESS in 1h 03m 49s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/576689163df24584b3d1b091ec3fbc53 : SUCCESS in 33m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fa951e2de92434abcae10c6c8785ee4 : SUCCESS in 1h 09m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ecc971ac36e54399babfda90081dae68 : SUCCESS in 18m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/62cf2195397f4228a984b4bcaabf809c : SUCCESS in 12m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fa57d24f9df444e5b8d8d6afbc5b4b5b : SUCCESS in 20m 41s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/64a106160f324399833f1b016767ab13 : SUCCESS in 6m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ec8f88b59d954ba5acb95bdc9aade633 : SUCCESS in 27m 19s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/909ccf5dc8ee44568cd8051d54ffd29a : SUCCESS in 9m 04s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/d06c645337434ee492045a27db36d1a4 : SUCCESS in 14m 47s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/72e65d392b67456ea42176e9f9b135c7 : SUCCESS in 17m 34s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7059735e8b34427899708ac283ae9467 : SUCCESS in 15m 14s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b1c0d7819e804181bf79fe00e375d0d9 : SUCCESS in 15m 16s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0bb38ab2b32643ca84907164b5b2c6a0 : SUCCESS in 16m 11s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/1bfb95e793fe49a686a118b3a7a9cf90 : SUCCESS in 4m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f51cb4d8b6c54f9bace104b33991dc5c : SUCCESS in 8m 13s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/292c503b16d046dcaca1776a51d58641 : SUCCESS in 8m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/63349dd69b444371b3a16cda80d1750a : SUCCESS in 1h 02m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c5499487228c4f13b507d8a682ddb8f0 : SUCCESS in 22m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/79545a83cbde4ddab1c94c089ddcd908 : SUCCESS in 52m 16s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/af697fe13be141d5ac5095c98edb6eac : SUCCESS in 23m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/15b6c1340fb244018a062ee831bc6885 : SUCCESS in 3m 37s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/8bcaae98d45e4093bfb1b7c75b9a7f02 : SUCCESS in 18m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e8b96b6bf5ec4a618232441f293297da : SUCCESS in 28m 23s","accounts_in_message":[],"_revision_number":18},{"id":"f6734504060c6744af6d59c896d73ac5c2fe3fcf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 11:33:05.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb7e9b7a86de4cfb8bec98e006493cc4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/bd1a8fa1522140be8d9e9c0865f7799c : SUCCESS in 33m 08s","accounts_in_message":[],"_revision_number":18},{"id":"d4db41cb3e7d3a418b810cf110f8f19da0803962","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-13 09:47:13.000000000","message":"Uploaded patch set 19: Patch Set 18 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":19},{"id":"7ec169e961c76ada8aeb7f9ec357aa3785a6ba57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 12:09:34.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/90bdc2047ab54785b89f683a082d9eb3\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/53d9264976c549ef86988a7642aa9705 : SUCCESS in 5m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/2149fdbd0bc44026870193890db270f2 : SUCCESS in 1h 20m 54s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7b3f438a2a084e6bbc76f36ba3247d13 : SUCCESS in 1h 12m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/c482a502ba174149a3136d7fc29667a8 : SUCCESS in 34m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/525560747cfa4bf09766f5a1383952f7 : SUCCESS in 1h 08m 43s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c6332d03f9eb44ab9ed8f392ca3d7f20 : SUCCESS in 20m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1297d79caade42a38cd4980ddbdaf335 : SUCCESS in 21m 55s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eeff08fffecf452993b114b46b1bf5d0 : SUCCESS in 12m 14s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51dd4c769ebb4ec8921e7cb0a0752916 : SUCCESS in 6m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3e282c59e947496ba99d5814eafe9df3 : SUCCESS in 26m 54s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c60955b74b4a4505911430fe5bb1daa3 : SUCCESS in 23m 15s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/81914c28593c4a1db37d4885c61b55db : SUCCESS in 25m 22s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/0971232f588f433cb7bf2b424e461bb0 : SUCCESS in 27m 11s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/204cadf0b587473eb4bc607d5f49ae13 : SUCCESS in 8m 20s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ec1c24f6e9a44bb8818d33f600a1616b : SUCCESS in 14m 21s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/da7a9f598003482983b6ea4bbdf26732 : SUCCESS in 10m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/615210d2845940ecadb11d043ac59d27 : SUCCESS in 5m 57s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/acf4aa6f66db47aa92123e7004a51afd : SUCCESS in 20m 43s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/2fda8f8b9aa0499595acded4ac1f91f9 : SUCCESS in 7m 21s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/1c1a95735bdb4d7db32b1e4d452920d1 : SUCCESS in 53m 27s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/e6530ede72644353b05fffb3ac6aa681 : SUCCESS in 55m 55s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/c6dc894207d44dee931e1ec5febe47f1 : SUCCESS in 42m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6da4f69df9d0444dbf8ed707a2c2967f : SUCCESS in 28m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daaba5bc860f45bba506acd8f9eb114b : SUCCESS in 6m 06s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ad9e4efde3f945b085448bd61e9e8d81 : SUCCESS in 17m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c2dc5cef428b4333b62cf47409991a3b : SUCCESS in 1h 01m 40s","accounts_in_message":[],"_revision_number":19},{"id":"ef6a5026882c25bf63a7cbc53a093885a608e7c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-13 15:42:45.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e7c4eb83c141fa9fc62cd17c0f87d4\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/e5cfa4c9fbdf4caeafd8d5227ad2ab27 : SUCCESS in 36m 12s","accounts_in_message":[],"_revision_number":19},{"id":"2ac4bd06a87c6fa8f28a152ff2f09f33918fe3f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-04-14 15:09:20.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":20},{"id":"4a78d5fa62ea693cb118010378393e0cec0f8ed0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:19:33.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7352dedc6204d69a9b501cb366439c6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97f306373ef944c9a592fa889f5f5d73 : SUCCESS in 13m 09s\n- grenade https://zuul.opendev.org/t/openstack/build/316ba2aa04f04b94a9a1ef78a5296ce1 : SUCCESS in 53m 11s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0632bbd91a804e648ca0530a7004aef1 : SUCCESS in 34m 25s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f9b2b8e3ba92433abf4dfcc9824aad59 : SUCCESS in 54m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0aa6b7af45a441feb4f98eef93e7ed55 : SUCCESS in 1h 02m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4ef28d7cb9c648a0a6d424f74b148ac8 : SUCCESS in 16m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/7a5e5cbde22645f4925dcfe0df9eba29 : SUCCESS in 20m 38s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2dffaf33c93a4dd3bffbaec68842c9b3 : SUCCESS in 14m 27s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/31cb4aa9300843b9a00f0d61650e9346 : SUCCESS in 9m 32s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9d70a4df2a50452d9f687dd539777146 : SUCCESS in 26m 37s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/bea7e28a9ad34f9e97ce721c1168a9d0 : SUCCESS in 24m 38s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f01f68b230074ab5b3d0cc3aad2efdf6 : SUCCESS in 24m 50s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/5265920e4a7b4bfcbe6959a935748ebc : SUCCESS in 21m 10s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0a259d31a5f4707a9b2770a32d94756 : SUCCESS in 15m 33s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4ce92c6f44f545fea3d785fb07d0be4a : SUCCESS in 12m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/f88bed62f71644c28751dfbb6dc9e205 : SUCCESS in 16m 13s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/306894f745be42d9b43dda8dd1a5b381 : SUCCESS in 7m 04s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/3ce7313fadce437196a5e767c45db4fe : SUCCESS in 16m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/43d52e38ad924612b7000e38d1f2dae6 : SUCCESS in 12m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6e252d59bac4346bf89d1d80a9dad80 : SUCCESS in 41m 21s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/9125900fc1f54adf91f1bda43726442b : SUCCESS in 29m 25s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/388df1d9eebf4d9ea94591cd182b0aac : SUCCESS in 54m 25s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/12aa048da13d4d0b83a464e325968445 : SUCCESS in 13m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1b84244ca934d3e940ee416f654f689 : SUCCESS in 6m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c45b93befdf247cd86bc8c495cf82be1 : SUCCESS in 20m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3f16aa36126f41ffbbb34f87074874e8 : SUCCESS in 29m 54s","accounts_in_message":[],"_revision_number":20},{"id":"8fbc214027a39dcb71f04a61842e612850f7ba44","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-14 16:55:06.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43f54f25cac14e68b407f632d597d038\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b498d2226b2b4ad9a7d434e4cb3f445d : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":20},{"id":"7f2804b1d5bd0de37358d92dfa54db35308f9094","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2026-05-06 10:50:40.000000000","message":"Uploaded patch set 21: Patch Set 20 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"d6a6de92f9df486b44f7cc75be9e8db46d66f3b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 11:40:20.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eaa82537754447349b4b6f406fb7e732\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5bfa4b318a34649a4400fcc4744ee87 : SUCCESS in 31m 20s","accounts_in_message":[],"_revision_number":21},{"id":"9ee914f5c7d209234b0fbe7b57437874ffb694c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-06 12:14:50.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/150b0eeceeb341c18245c1aea3fcead0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af849280a4b94746bed5827e254ff71d : SUCCESS in 10m 16s\n- grenade https://zuul.opendev.org/t/openstack/build/b7f42a37958f4c569aa0ac992598fcec : SUCCESS in 1h 07m 56s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/debd8582e8a6438d923374472798b3e7 : SUCCESS in 1h 12m 11s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5c5b705d3fff47898add7af4c6f9203a : SUCCESS in 54m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a3ef2c0d7bc84bcebcf70023866f277a : SUCCESS in 1h 03m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f84b8fbfbcd1447eb35ce86e07935988 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/26926b7b8d8c47bdbd0751e28f4904e5 : SUCCESS in 19m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4b8e16e622bf4c30b58aa07e8178df12 : SUCCESS in 21m 22s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3a9c1bbc3cd94f4aaddce6814da48ed9 : SUCCESS in 6m 52s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cb842afe1eef445794816bd2d71f843c : SUCCESS in 15m 18s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/726ce0c7597540af87fba94cbf820a98 : SUCCESS in 16m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/8d2d00c80502466c8314123c2bf43d50 : SUCCESS in 16m 31s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/283a5ebfa0224eb484bc1842621b7aa9 : SUCCESS in 24m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/be2ead6dea28453594e21273a1599a4a : SUCCESS in 16m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/33dfde8ae1314a7788e36abf2638be31 : SUCCESS in 11m 02s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/200ac91ef7cf492b9606297a50a670e1 : SUCCESS in 17m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5107d5f1603c4feb8155d333f5b30d4c : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/fd69dbd41cf64c878d435833b43502d7 : SUCCESS in 23m 15s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b82ef30add3c423391731e8007ffb3fb : SUCCESS in 9m 13s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b32eee7f74e24edfabb0a79569c53ca1 : SUCCESS in 36m 40s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a1f38011de5a46beba6fb34612140efc : SUCCESS in 54m 46s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a56e167b17bb468a8be313c97a57fa54 : SUCCESS in 59m 58s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/be683de0925348f6b5a8141ce818b12d : SUCCESS in 29m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9400939c30b4cb093cb45322af886dc : SUCCESS in 4m 41s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/903986e252d442e4b110ebe58a2c9cba : SUCCESS in 19m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c599003ea350418cb3ac90f04f97d663 : SUCCESS in 44m 14s","accounts_in_message":[],"_revision_number":21},{"id":"a34c8c46fb2512adcafd4df1c79eb5193f245dbb","author":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"date":"2026-06-25 09:53:38.000000000","message":"Patch Set 21: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"c9e6e521974f945b2c72a830a361f9c1c1021ad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 09:54:17.000000000","message":"Patch Set 21: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":21},{"id":"8c3f6f002b8ef0ff0a181f21bc649024211f5354","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:58:14.000000000","message":"Patch Set 21: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b63682c93aae4290bc79f03fb8c6c817\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ffbbc0bdeb647739b9b588d27a6088c : SUCCESS in 12m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/15d0065976604345970b32aabfe11c04 : SUCCESS in 8m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/8e7c5d8c15304d92a1eba1f73ee590c5 : SUCCESS in 1h 09m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/64a68ef219b544e5b2f4796d962f7a55 : SUCCESS in 34m 01s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/76b105799b5a407089e4cdfbf6af3f50 : SUCCESS in 58m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2a10ee8917694b01bed23c3ab0d33212 : SUCCESS in 59m 28s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/c188e699ff0541759f9410b364936a4c : SUCCESS in 8m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/cf338ca6168a4a91a58b0b3477ae4388 : SUCCESS in 28m 16s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/b8fe269091d5424e901227b601025a33 : SUCCESS in 22m 22s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/bb767ae06a8f4dc2afd8147cddace066 : SUCCESS in 31m 13s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/311f49d9c78d43fb8cc7431cac624dd1 : SUCCESS in 17m 27s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/baa6271e3fc44b77b22bcc73632c32e9 : SUCCESS in 8m 37s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1651969e329b46ff85c9bbedcdb17582 : SUCCESS in 9m 35s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/983fb462463044e5875a995889b5192b : SUCCESS in 16m 25s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/474b041a82f74214b7e5032a9b0ddcee : SUCCESS in 7m 51s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/bce8df0550dc401ab7b2335b596ac81c : SUCCESS in 6m 17s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc48031f412746c69fa0504759983dec : SUCCESS in 1h 05m 15s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c2abc23eabd943fcacb49b4078d49aae : SUCCESS in 51m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ac3f861a7fb4a23abfb109ff280dd36 : SUCCESS in 41m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f46242eb86144409a84fd349b3397bde : SUCCESS in 30m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a292ad204ee24cd0a33d5552105c351c : SUCCESS in 6m 13s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/61384737d58b4e79ab55264245ef5316 : SUCCESS in 21m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f0edde9e25c0469aa72ed172c24abad8 : SUCCESS in 1h 04m 20s","accounts_in_message":[],"_revision_number":21},{"id":"fcb828bd5dfd6de3fc2ba6ab54c3c63b0f14012f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 11:58:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":21},{"id":"42afafdefeed11fe73980d87d29d05d1b2cd17ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 12:01:08.000000000","message":"Patch Set 21:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b163ffb22ab4b3a8ed451805e759707\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b4ad3f753294befa4758830642f4223 : SUCCESS in 47s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/aa88a7089f304afea2cd9240a1d47e58 : SUCCESS in 52s (non-voting)","accounts_in_message":[],"_revision_number":21}],"current_revision_number":21,"current_revision":"5a16b75b6bd8bcf40c0ba3f2502d3506d03a2e06","revisions":{"366916b21f209096238e3b77e0bb9f511d426026":{"kind":"REWORK","_number":1,"created":"2025-06-10 15:08:36.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/1"}}},"commit":{"parents":[{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"subject":"Common: add sub-operation timing breakdown in labeled_timing_stats","message":"Common: add sub-operation timing breakdown in labeled_timing_stats\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/366916b21f209096238e3b77e0bb9f511d426026"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/366916b21f209096238e3b77e0bb9f511d426026"}]},"branch":"refs/heads/master"},"818013c0c95c09129bf78160d0598d6a7f558f24":{"kind":"REWORK","_number":2,"created":"2025-06-10 17:34:58.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/2"}}},"commit":{"parents":[{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 17:34:48.000000000","tz":-420},"subject":"Common: add sub-operation timing breakdown in labeled_timing_stats","message":"Common: add sub-operation timing breakdown in labeled_timing_stats\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/818013c0c95c09129bf78160d0598d6a7f558f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/818013c0c95c09129bf78160d0598d6a7f558f24"}]},"branch":"refs/heads/master"},"3da1dd9523965d934f1f1f32455624de31e2a58a":{"kind":"REWORK","_number":3,"created":"2025-06-10 17:40:47.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/3"}}},"commit":{"parents":[{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 17:40:20.000000000","tz":-420},"subject":"Common: add sub-operation timing breakdown in labeled_timing_stats","message":"Common: add sub-operation timing breakdown in labeled_timing_stats\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3da1dd9523965d934f1f1f32455624de31e2a58a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3da1dd9523965d934f1f1f32455624de31e2a58a"}]},"branch":"refs/heads/master"},"155bc4c58848538ec6c4458204f915e072ce2f67":{"kind":"REWORK","_number":4,"created":"2025-06-12 00:26:46.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/4"}}},"commit":{"parents":[{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-12 00:24:27.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155bc4c58848538ec6c4458204f915e072ce2f67"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/155bc4c58848538ec6c4458204f915e072ce2f67"}]},"branch":"refs/heads/master"},"80f8b5418bbdda3cd00d0c7161edeb02581d0e28":{"kind":"REWORK","_number":5,"created":"2025-07-02 21:03:41.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/5"}}},"commit":{"parents":[{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-02 21:02:22.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80f8b5418bbdda3cd00d0c7161edeb02581d0e28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/80f8b5418bbdda3cd00d0c7161edeb02581d0e28"}]},"branch":"refs/heads/master"},"00fd734b25bb5eec1653b86df1c606a20221efc4":{"kind":"REWORK","_number":6,"created":"2025-07-03 19:20:37.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/6"}}},"commit":{"parents":[{"commit":"9106b27fe31e09880096bf20f19c219e3a7a1934","subject":"Merge \"CI: Skip grenade-skip-level-always similar to tempest\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9106b27fe31e09880096bf20f19c219e3a7a1934"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-03 19:20:27.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00fd734b25bb5eec1653b86df1c606a20221efc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/00fd734b25bb5eec1653b86df1c606a20221efc4"}]},"branch":"refs/heads/master"},"f368f48bb1d4b9084e72f67ce5c9a02d05589981":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-07-14 18:58:34.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/7"}}},"commit":{"parents":[{"commit":"9ffecb7d1945d2eb8b2d53555d8304fa8937aa1b","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9ffecb7d1945d2eb8b2d53555d8304fa8937aa1b"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-07-14 18:58:09.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f368f48bb1d4b9084e72f67ce5c9a02d05589981"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/f368f48bb1d4b9084e72f67ce5c9a02d05589981"}]},"branch":"refs/heads/master"},"92542e7c31003021aa49faed40ebe29c32a33ed7":{"kind":"REWORK","_number":8,"created":"2025-08-20 05:28:06.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/8"}}},"commit":{"parents":[{"commit":"797adfa88803127a372d3e47ef7ad5f3e625413d","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/797adfa88803127a372d3e47ef7ad5f3e625413d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-20 05:18:31.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92542e7c31003021aa49faed40ebe29c32a33ed7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92542e7c31003021aa49faed40ebe29c32a33ed7"}]},"branch":"refs/heads/master"},"1066e05a5dfd611f2f763c2ecea27cd6d60e9c1b":{"kind":"REWORK","_number":9,"created":"2025-08-22 05:20:44.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/9"}}},"commit":{"parents":[{"commit":"797adfa88803127a372d3e47ef7ad5f3e625413d","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/797adfa88803127a372d3e47ef7ad5f3e625413d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-22 05:17:45.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1066e05a5dfd611f2f763c2ecea27cd6d60e9c1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1066e05a5dfd611f2f763c2ecea27cd6d60e9c1b"}]},"branch":"refs/heads/master"},"21a72996eef209baad0a333531f2518bda12667d":{"kind":"REWORK","_number":10,"created":"2025-08-23 00:04:17.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/10"}}},"commit":{"parents":[{"commit":"797adfa88803127a372d3e47ef7ad5f3e625413d","subject":"Delete s3api MPU segments when expiring the manifest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/797adfa88803127a372d3e47ef7ad5f3e625413d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-08-23 00:04:05.000000000","tz":-420},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a72996eef209baad0a333531f2518bda12667d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/21a72996eef209baad0a333531f2518bda12667d"}]},"branch":"refs/heads/master"},"4c3e95ef100f8b7c2f3c60beb7ab78c2c25972be":{"kind":"TRIVIAL_REBASE","_number":11,"created":"2026-02-10 04:43:10.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/11"}}},"commit":{"parents":[{"commit":"c7bee623658b464072783096b98adaddfb1005bc","subject":"Merge \"s3api: Support for object expiration time\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c7bee623658b464072783096b98adaddfb1005bc"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-02-10 04:12:43.000000000","tz":-480},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3e95ef100f8b7c2f3c60beb7ab78c2c25972be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4c3e95ef100f8b7c2f3c60beb7ab78c2c25972be"}]},"branch":"refs/heads/master"},"9a1adb5c67fa0022ed4880c0202aca5f3d0e2b21":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-03-03 05:39:12.000000000","uploader":{"_account_id":34930,"name":"Jianjian Huo","email":"jhuo@nvidia.com","username":"jhuo"},"ref":"refs/changes/60/952260/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/12"}}},"commit":{"parents":[{"commit":"51b068309a626d2228ab69b5158ee857e96d101d","subject":"tests: more timestamp iter refactoring","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/51b068309a626d2228ab69b5158ee857e96d101d"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2026-03-03 05:34:31.000000000","tz":-480},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a1adb5c67fa0022ed4880c0202aca5f3d0e2b21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9a1adb5c67fa0022ed4880c0202aca5f3d0e2b21"}]},"branch":"refs/heads/master"},"1d4c87ca9121c856804ab328becbf81cc3b75474":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2026-03-18 13:35:59.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/13"}}},"commit":{"parents":[{"commit":"e63f0b3d68770d4792e1606ff8fe924b3f25b7f9","subject":"Merge \"tests: Make self.ts reliably a method (if present)\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/e63f0b3d68770d4792e1606ff8fe924b3f25b7f9"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 13:19:02.000000000","tz":0},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d4c87ca9121c856804ab328becbf81cc3b75474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d4c87ca9121c856804ab328becbf81cc3b75474"}]},"branch":"refs/heads/master"},"92a8f1bcd8d97e2291ef8366a4d473ef75c148a9":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2026-03-18 15:18:47.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/14"}}},"commit":{"parents":[{"commit":"1d6813c656835bc81acba9504c428c5c41e988c8","subject":"Introduce a jitter component to Timestamp","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1d6813c656835bc81acba9504c428c5c41e988c8"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-18 14:57:12.000000000","tz":0},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92a8f1bcd8d97e2291ef8366a4d473ef75c148a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/92a8f1bcd8d97e2291ef8366a4d473ef75c148a9"}]},"branch":"refs/heads/master"},"0bd26531142205b06418786dec82c08ec95f3b12":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2026-03-24 17:58:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/15"}}},"commit":{"parents":[{"commit":"ae0ef5d59cc824b2ac05c717a79aff61f86999de","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ae0ef5d59cc824b2ac05c717a79aff61f86999de"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-24 17:40:00.000000000","tz":0},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd26531142205b06418786dec82c08ec95f3b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0bd26531142205b06418786dec82c08ec95f3b12"}]},"branch":"refs/heads/master"},"3941f498fd456c193a07488e9088ab876db96da3":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-03-25 16:30:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/16"}}},"commit":{"parents":[{"commit":"49dc0711dac95caefd9bb857440eaf1effa2384b","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/49dc0711dac95caefd9bb857440eaf1effa2384b"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-25 16:24:45.000000000","tz":0},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3941f498fd456c193a07488e9088ab876db96da3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3941f498fd456c193a07488e9088ab876db96da3"}]},"branch":"refs/heads/master"},"9786fb07f0377a4a19cd6564d19a271fc11faa58":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-03-31 14:54:16.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/17"}}},"commit":{"parents":[{"commit":"1e818bdd5f757de1411e685f5d2baf04018d0259","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1e818bdd5f757de1411e685f5d2baf04018d0259"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-03-31 14:53:54.000000000","tz":60},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9786fb07f0377a4a19cd6564d19a271fc11faa58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9786fb07f0377a4a19cd6564d19a271fc11faa58"}]},"branch":"refs/heads/master"},"de984474c9f7142b4a4106c57d978c6f8cc6be40":{"kind":"TRIVIAL_REBASE","_number":18,"created":"2026-04-09 09:21:56.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/18"}}},"commit":{"parents":[{"commit":"fa792e5ad5eec34d69c624136d995c5f9ff689b1","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/fa792e5ad5eec34d69c624136d995c5f9ff689b1"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-09 09:16:31.000000000","tz":60},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de984474c9f7142b4a4106c57d978c6f8cc6be40"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/de984474c9f7142b4a4106c57d978c6f8cc6be40"}]},"branch":"refs/heads/master"},"8ca043f23388852b9e2f828bd4d3575a41d7411b":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-04-13 09:47:13.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/19"}}},"commit":{"parents":[{"commit":"5c8af87ca882ef822fe675f5a2cead1180a210b3","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5c8af87ca882ef822fe675f5a2cead1180a210b3"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-13 09:46:53.000000000","tz":60},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ca043f23388852b9e2f828bd4d3575a41d7411b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ca043f23388852b9e2f828bd4d3575a41d7411b"}]},"branch":"refs/heads/master"},"93f0abc8e3c6494bccb79c5842ada2950f8f2cad":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2026-04-14 15:09:20.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/20"}}},"commit":{"parents":[{"commit":"2710b1ff9fc434b41f2d8427e864451bd2c064d7","subject":"timestamps: add NormalTimestamp for ShardRanges","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/2710b1ff9fc434b41f2d8427e864451bd2c064d7"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-04-14 15:08:58.000000000","tz":60},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93f0abc8e3c6494bccb79c5842ada2950f8f2cad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/93f0abc8e3c6494bccb79c5842ada2950f8f2cad"}]},"branch":"refs/heads/master"},"5a16b75b6bd8bcf40c0ba3f2502d3506d03a2e06":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2026-05-06 10:50:40.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/60/952260/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/60/952260/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/60/952260/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/60/952260/21"}}},"commit":{"parents":[{"commit":"7abd896027f40902068f3d5e075de62bf0db0dd2","subject":"Merge \"container-sync: Use internal-client.conf if it exists\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7abd896027f40902068f3d5e075de62bf0db0dd2"}]}],"author":{"name":"Jianjian Huo","email":"jhuo@nvidia.com","date":"2025-06-10 14:59:08.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2026-05-06 10:50:06.000000000","tz":60},"subject":"Common: a new decorator to log timing breakdown for slow requests.","message":"Common: a new decorator to log timing breakdown for slow requests.\n\nChange-Id: I73f1a03818157dc76938cbffd520c5c38941e2b6\nSigned-off-by: Jianjian Huo \u003cjhuo@nvidia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a16b75b6bd8bcf40c0ba3f2502d3506d03a2e06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5a16b75b6bd8bcf40c0ba3f2502d3506d03a2e06"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
