)]}'
{"id":"openstack%2Fswift~952311","triplet_id":"openstack%2Fswift~master~I3167a539b3e26ceb35976fbd7a2356ba59d4a5e4","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-13 02:48:14.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3167a539b3e26ceb35976fbd7a2356ba59d4a5e4","subject":"Drop remaining skip check for Python \u003c 3","status":"MERGED","created":"2025-06-11 03:03:02.000000000","updated":"2025-06-13 02:49:46.000000000","submitted":"2025-06-13 02:48:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"952311","meta_rev_id":"73f0442885de5b0ad7b1748db144285da286c363","_number":952311,"virtual_id_number":952311,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-13 02:48:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-06-13 01:17:01.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2025-06-13 01:17:01.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 03:56:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-13 01:17:01.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"2c813191775d2ac36ae4590777ce5d9c2ec9cbaf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-11 03:03:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7649853b68080058675a1f15ebb288697081cc72","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 03:56:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8006fad7c284829a87b7a349b777ec7\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/783ce7ae611d4c7dba4564342717cb6f : SUCCESS in 9m 07s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ee4fbd85512e4dbdbae7bf4cbde72d46 : SUCCESS in 19m 45s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/53c839211f9b499a954368a8c2d15fce : SUCCESS in 18m 42s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/51f0210293f14839a93955077820fecf : SUCCESS in 19m 46s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/518ec0ed4c294870b6a50486330e2e9e : SUCCESS in 5m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/277957c865aa4cd1a34fa27e9dca2bdb : SUCCESS in 21m 11s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/79122178cda8402686d65dd680966c0b : SUCCESS in 21m 44s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/4467625ce9e24cc390ca5f6ba0e4ce2f : SUCCESS in 19m 37s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e7480a65130446d2b85674d20d84f8e0 : SUCCESS in 25m 24s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c0cd84070ef44c14b69314e35b53aac5 : SUCCESS in 16m 16s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ae326bfe894741a690bc4f5d981e9e39 : SUCCESS in 9m 51s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/04a89003dc00440bb29d1376d5aeeb10 : SUCCESS in 15m 12s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/de0e8fe2ff9b49dd9fd21261b75f7ab2 : SUCCESS in 48m 57s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/1589cb3ed8a3402783c83d81cceec761 : SUCCESS in 34m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0a71185237764fdeac14e79372b56da3 : SUCCESS in 20m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae932c1adb0f44a790ec65aba4ab178d : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b0b6c2f53bcf481f967908b83c229a7f : SUCCESS in 24m 51s","accounts_in_message":[],"_revision_number":1},{"id":"4dab3b18c3ae4b8ca336180994c9676dae5cb87c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 04:53:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8c70888fd2d42c79d44cd39585bf7c3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4b35def09bed476385e5da7bdda609c2 : SUCCESS in 1h 05m 13s","accounts_in_message":[],"_revision_number":1},{"id":"565a78148280b4d413db4b506922e00396717e1f","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-06-13 01:17:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a6ef10f38a9af30238830d9ee42a0caad298f5f8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 01:17:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f5722448641b6ac73fe5fb1b47c99c5a69af1291","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 02:48:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec2bfd8cf20242e48b2ae74f6cc53d72\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e299957a25a4ffcb609abd4e5ebe45b : SUCCESS in 8m 10s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/9d7affa243224f918049b8e6c3066968 : SUCCESS in 4m 30s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/1b4f0610b01545349359505bcc9fffaf : SUCCESS in 14m 29s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/558e46be595c401c80bbd2785ca3b519 : SUCCESS in 21m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/996e62e0f9c4421e9bd0a3d93a9301cb : SUCCESS in 9m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/074fb7684bb342fbb2f6aee69475c26d : SUCCESS in 14m 23s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/01c714f9155e40c69f1ec7f6821040a6 : SUCCESS in 14m 36s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/e376463383fc4d49867cc578a65afaf6 : SUCCESS in 12m 32s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/afe4d0b44c0c470ea29f13ff11afdc4a : SUCCESS in 6m 39s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/73775e6984a64797b4a316386f621692 : SUCCESS in 9m 28s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/c9ccf9d3742e4f4ea2115989773f6deb : SUCCESS in 46m 39s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4a4543572337427b95c6a05e48bf7734 : SUCCESS in 33m 43s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a79b17aabf6c45dfbe37e78e9830f286 : SUCCESS in 16m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/704367d3c9d1455d96a6d474c4f3052e : SUCCESS in 4m 10s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/bf3dac73bfcf4e4898ef266e1df1a415 : SUCCESS in 17m 23s","accounts_in_message":[],"_revision_number":1},{"id":"0ba5f2d55d2bc7622246fe636c0c78ca3b300cab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 02:48:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"73f0442885de5b0ad7b1748db144285da286c363","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-13 02:49:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36ee7a7bcc2946d0a809570814ab0e99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903a334e38a24a978af4e238e9dd6dd6 : SUCCESS in 46s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/2f6b57042946475aa1cadd8a34a55d33 : SUCCESS in 35s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"005d69d1a9f3eb355901b135c854a4c4984e3540","revisions":{"005d69d1a9f3eb355901b135c854a4c4984e3540":{"kind":"REWORK","_number":1,"created":"2025-06-11 03:03:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/11/952311/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/11/952311/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/952311/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/952311/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/11/952311/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/11/952311/1"}}},"commit":{"parents":[{"commit":"d2272833fe8fea82b0321e0b990d265449089d59","subject":"Merge \"tests: Fix some connection-closed testing on OS X\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d2272833fe8fea82b0321e0b990d265449089d59"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 03:02:40.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-06-11 03:02:40.000000000","tz":540},"subject":"Drop remaining skip check for Python \u003c 3","message":"Drop remaining skip check for Python \u003c 3\n\n... because Python 2.x is no longer supported.\n\nChange-Id: I3167a539b3e26ceb35976fbd7a2356ba59d4a5e4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/005d69d1a9f3eb355901b135c854a4c4984e3540"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/005d69d1a9f3eb355901b135c854a4c4984e3540"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
