)]}'
{"id":"openstack%2Fswift~952462","triplet_id":"openstack%2Fswift~master~If25a8f0a3079558544ab15c874eda666a9f69933","project":"openstack/swift","branch":"master","topic":"p-s3api-multi-delete-checksum","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-11-17 20:23:45.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-11-17 17:13:54.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"If25a8f0a3079558544ab15c874eda666a9f69933","subject":"s3api: verify checksums on DeleteObjects POSTs","status":"MERGED","created":"2025-06-12 13:58:44.000000000","updated":"2025-11-17 20:24:50.000000000","submitted":"2025-11-17 20:23:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":25,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"952462-p-s3api-multi-delete-checksum","meta_rev_id":"426443308d318684dbb8017f7ea32351600613b1","_number":952462,"virtual_id_number":952462,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-17 20:23:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-11-17 17:13:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-11-17 17:13:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-12 15:24:06.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-12 17:03:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-01 19:52:26.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"}],"messages":[{"id":"bacbc92bc3ca44c7937b1100559b1a5359a156e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-12 13:58:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7a32a5b533c515cf827e73e6ed19381a09379337","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-12 14:04:54.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6d9e59789f4d56712cac0633691da5a30b710edc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 15:24:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4250539f2b049928474765b55b461b8\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f97dbf4bf2c24267ba880f93d8455f46 : SUCCESS in 1h 19m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fba6ae65d4416ee60d9cfd816b375fe8d808fc0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-12 17:03:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59cbe497c413467c868ba77361458089\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/859f4eae9e834d8dbcd61b9338867c33 : SUCCESS in 9m 10s\n- grenade https://zuul.opendev.org/t/openstack/build/4c4b99768f064c18b0fb69da21e346d4 : SUCCESS in 50m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7337ac186b77467fb8133ffef27d0563 : TIMED_OUT in 3h 02m 32s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/2398fdaf5ef94c9c95432aa7a859cfc7 : SUCCESS in 41m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bcbc097a278411bbd2461124d6aeee5 : SUCCESS in 55m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/57a36f967a7945589b216ef085ddb235 : SUCCESS in 13m 38s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fb6f5b9ae0da45979ce3f0888abfa73e : SUCCESS in 18m 01s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5f58b76e7c44f8695855b90ba36af08 : SUCCESS in 13m 49s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2d4f8a8827ba4e4db7dbe211c3a5a74e : SUCCESS in 3m 58s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/bb30b294e442449886eb3e4c014b4428 : SUCCESS in 12m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9dd77edc70d54f7c849e7f507c1a931b : SUCCESS in 20m 16s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/ca90e1ca60dd4335bc13999319b68b4f : SUCCESS in 32m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/6ef43003011f4689a658188f77d41ac6 : SUCCESS in 26m 16s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c76fab7786a2431f9d6a0c9e8de52862 : SUCCESS in 15m 39s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/2254774937c54cc0a9691f2b81a01a8a : SUCCESS in 14m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/29c7d6e6870f45a0b845cd7629729b1f : SUCCESS in 16m 52s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/34752ce747294bdeaaf855356c39f811 : SUCCESS in 4m 19s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/c760d785c71b4f98bf9a84b8737399c6 : SUCCESS in 14m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/33980b5d08054d688e868668fdc9f4fa : SUCCESS in 8m 54s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f2dc004e83564aeaaa9de044e6b217c5 : SUCCESS in 1h 03m 11s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a31f5242b28b4b14be704ff8eac5798b : SUCCESS in 55m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/34d1a6c1028d4915b8b307189fc01e0f : SUCCESS in 52m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de8ceab887f0404c8f47a0f3451a3dad : SUCCESS in 21m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec6d21064e3c4d9bb6aca450374844bd : SUCCESS in 2m 51s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/398e6c069ea84d5dade5b45451c65e15 : SUCCESS in 23m 32s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/024c72e5bb2448fca695bb81b2d66b78 : SUCCESS in 41m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a0650e91374746919f16ac07ba96a3fa : SUCCESS in 53m 09s","accounts_in_message":[],"_revision_number":1},{"id":"c2ffbf075e6ca2d9eabb8136425c9cf3db91a6cc","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-06-13 16:18:19.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f9a95df2d921c1e83fe7cad0faee9ac98dbd59b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-11 13:54:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"288527f7d0712ebb284dab7041decab94aed3027","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 15:06:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c548eb3c0b14a6389fdd92bafe29d18\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b326fa3a5acf44459d1bde430211e9b2 : SUCCESS in 8m 53s\n- grenade https://zuul.opendev.org/t/openstack/build/40993653a02c403f8a41709ceb651dd5 : SUCCESS in 40m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c8ae7bf1e4ca4245bdf441156516e0af : SUCCESS in 49m 55s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3fe4695f23054552a8bebd6ea571d073 : SUCCESS in 1h 00m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/959d6024bfb34dab878c680cc3fa7c12 : SUCCESS in 50m 49s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ed704377e20249ae806b2872e5d10c53 : SUCCESS in 23m 52s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e274b61862114540acdea0b8c0c097cf : SUCCESS in 20m 48s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81c46e33808542f9a143b9ba950c4b50 : SUCCESS in 24m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/55d0a038eb2f4fdc8ba524ac62d85787 : SUCCESS in 5m 09s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/20778bbf192d4a8aa22de86f110fb756 : SUCCESS in 9m 15s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/433faf3188b6469aa8b7d2262cf58cfa : SUCCESS in 20m 57s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/f290a0dedfcb4009ab3b6bc8b0340371 : SUCCESS in 31m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/f5ecc9b7391f4aa5ac4acaa5459bd922 : SUCCESS in 26m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cdc394f8605c40b8af6a5d295914cec6 : SUCCESS in 11m 10s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4b56a21253c54630a2f46b0abef327a5 : SUCCESS in 13m 43s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/5069f80d13414bf297c0d6e7114c2bf6 : SUCCESS in 16m 54s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/e23ae793c087401886430e8ba5d91cb9 : SUCCESS in 9m 09s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f8ca9d60c9a744efa7137a0b96966ba4 : SUCCESS in 17m 28s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e4627df560e044d29e90192bf497bfbd : SUCCESS in 11m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e2fef4068ae54ea0b188d2d2e390f800 : SUCCESS in 54m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/14349ddd4b7746b9a6c3bbd1b20d30c8 : SUCCESS in 22m 45s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/49b5fcc64ef749889c1712251ba61452 : SUCCESS in 45m 23s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b6dfdf6b9b824ef3b58912e74cf47424 : SUCCESS in 18m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8cf3f83f4d3d40f6b98c77e3099bb4a7 : SUCCESS in 4m 25s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/d8873392c3dd4190888b2e7cefe359ea : SUCCESS in 18m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e5a6e276e7024a62b16cd9037bef8f80 : SUCCESS in 35m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/159a6d12dd0442328100e1ab67869910 : SUCCESS in 28m 11s","accounts_in_message":[],"_revision_number":2},{"id":"db36203420b39c303549ad95757108d50d19f985","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-11 22:05:35.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87162b20abea4f2fbb50cce76c05093c\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fdf9bc8f9fa409b9c62b2e7416e7525 : SUCCESS in 1h 27m 27s","accounts_in_message":[],"_revision_number":2},{"id":"1a549e66d09df2f72c4d7fd4b483dfadf5e68a3f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-18 09:19:18.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"6fed60fa9afc48794343ea8a0b772770606bca09","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-18 09:37:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c748666f9c20a1b7a9283976937f610fbb470df6","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-18 09:38:09.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"dca3822c81a77ab66dd3b8f37eb1b9b0f8955405","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 10:50:20.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a1f91e68035408b8f7eaaf6541cf742\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a79de833b0d94c95b58598fbbf382b29 : SUCCESS in 11m 37s\n- grenade https://zuul.opendev.org/t/openstack/build/53412246cdba47bf8e209d5507623d3c : SUCCESS in 1h 02m 21s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6663958c7f9346a4bcf1ae56ebc8df9a : SUCCESS in 1h 00m 30s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/265714c678c445439fce61b003544d28 : SUCCESS in 53m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e59a3abdb9a54b969202db679047f3c9 : SUCCESS in 1h 01m 16s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4374907b3f474a3f9bf8f4da15b23a3e : SUCCESS in 20m 55s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/845dd560b01c451889c4352009327209 : SUCCESS in 21m 07s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c56c55da154b4602aa50765dba3bc6b5 : SUCCESS in 21m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e95e6c82f280468db709a897105cb518 : SUCCESS in 6m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/29b2d98c48284eeab4c81c62e3fc5069 : SUCCESS in 18m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/dda2c05d2fff4926897b578d3fbd1342 : SUCCESS in 20m 01s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/926e083debf44ed9aba79b2eafc3966b : SUCCESS in 22m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/53b5c23b3790487ba7354aef362bb9ba : SUCCESS in 29m 07s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c2aeaa626da64ce585aa3659c7d86934 : SUCCESS in 14m 48s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/804b5679973843bbb244613dfc81aebe : SUCCESS in 14m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a74626d3d147427db73e115e4f0751ee : SUCCESS in 14m 09s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/bc415a8a3aaf4217ab26af770556b4b7 : SUCCESS in 6m 42s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/f6988a77bd2f4c3ebdfc5bdba78fac51 : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3acbab3f0b51442daf54e828307b6ce9 : SUCCESS in 7m 51s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/8786386cb1ef49ee8d4e7a4c20b3b450 : SUCCESS in 37m 59s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6c2c41d717dd4419b2d379d119686163 : SUCCESS in 56m 23s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/97bdf4dcef8b44b99d6f65aef226578e : SUCCESS in 49m 51s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dd8460e3a9b24790811b38622ccd6f80 : SUCCESS in 19m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b27fb72fb94485ba63fedf8bfd3eeb3 : SUCCESS in 4m 12s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b6f0997b36e44953a403d20434981957 : SUCCESS in 18m 24s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c63cba8434a3433cbc2bc2a4262f7ba5 : SUCCESS in 52m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cfb6c048a3364c859b5a9df17b7cf6a3 : SUCCESS in 34m 21s","accounts_in_message":[],"_revision_number":4},{"id":"de25a117268a67147f2ed93745b073f5b2ae2ff4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-18 11:08:12.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ca28fabac0243f3bbf566feb45e3402\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8810af49a9d449658b6d8552f0ca4585 : SUCCESS in 1h 03m 52s","accounts_in_message":[],"_revision_number":4},{"id":"320c068cb266743d33817b7cea2ab8c851d004cd","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-10-01 19:52:26.000000000","message":"Patch Set 4: Code-Review+1\n\n(8 comments)","accounts_in_message":[],"_revision_number":4},{"id":"9cfd3d371c09a5f6ba9c0435e028e7cc5f401e45","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-13 17:44:44.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"8bd060ac2438a45183f63c60738295b633dd3da6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 14:56:34.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7ff9765c6c89e08391f42a840fa80ef252991a40","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 15:04:46.000000000","message":"Uploaded patch set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"6404b7736486f67dd34ea4bee08a60071c745b38","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 15:09:11.000000000","message":"Patch Set 6:\n\n(5 comments)","accounts_in_message":[],"_revision_number":6},{"id":"7a63f9eac3cca6b02c72f633d3a11e73da60aa6c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 15:37:15.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/84d11fd498354e65bb722a6dd0286fd0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b9c488d00a5048c797b9428b5b9300b2 : SUCCESS in 30m 41s","accounts_in_message":[],"_revision_number":6},{"id":"6ef1919958ce66d18350b3b0db82a715e8d1035d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 16:08:27.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5e7c3e41df5a4a8c91f390a80bc083d8\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1e1c6b66a5b44e1a5f115b639076f10 : SUCCESS in 4m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/bcf6084f8a4c4a6bbfba4e1c2e774e89 : SUCCESS in 58m 15s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5c7fc6a2097a429d89e3b1fcc5f92056 : SUCCESS in 39m 29s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/74159c73e3d94a33aa5a12cc9dba7f64 : SUCCESS in 43m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6c739c5d67f5459d9f2c11f4e811fea2 : SUCCESS in 36m 41s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/072878492fd34f8d958dfd11051ab687 : SUCCESS in 19m 57s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a5c951a77dbd4d1c9f2308aa0cd8ecdf : SUCCESS in 13m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0905b84e881c4d89b57bf3fe331030f7 : SUCCESS in 20m 29s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/73b4f35862654a98bba1ca9b6386ad8d : SUCCESS in 6m 55s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/d3304ce4cbbe4d5296cd20042a080ea5 : SUCCESS in 24m 10s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/39a0d4fc719e49968f0ebd74101d9e66 : SUCCESS in 27m 53s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0a37187bbcbc4f6bb5011fc082e23984 : FAILURE in 34m 05s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/081e447f49414186939ae36df7013bd0 : SUCCESS in 21m 46s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7f20cc93093b4ed7842f5c585efe1dca : SUCCESS in 14m 26s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/69a6560d7b624f36856ac87e8a960078 : SUCCESS in 15m 31s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/6fa9b72a7ac54839bbfa70a8a95af6d3 : SUCCESS in 15m 29s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/15ddb5e0da154d509914281997afda8b : SUCCESS in 7m 52s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/03544798e0364021b4d55ec77457ed3d : SUCCESS in 8m 53s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/3537a99a5d564a02bf9d3ed6f5cc8d9f : SUCCESS in 8m 18s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/d5392c2f17974ab18856833cf04697b9 : SUCCESS in 55m 52s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/dbc1d9c0d9774b69b2687658c7c2a725 : SUCCESS in 50m 38s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/179c5fe82fbe4c8088ea68218913551b : SUCCESS in 25m 38s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7d432a484a1a41ff82a0d2c1dd5c5e6e : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d34caec3a84e3998086377280ed06c : SUCCESS in 4m 17s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/5acd58405c154848babe7ee0db705687 : SUCCESS in 19m 47s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/79f5f30fadcd4629ba0a21f433d133f9 : SUCCESS in 53m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c1e826d64a2d43e4b92998e3c4caccdf : SUCCESS in 33m 07s","accounts_in_message":[],"_revision_number":6},{"id":"f3303947c7e88ba649c945f73b94ed27bcf6b11d","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-11-17 16:45:38.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0bf09b04ea33eca00f39533332d845fa6a0d373f","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-11-17 17:13:54.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"0a3a16037c8e04339263e7556438d1694eaea08c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:16:56.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4caa55120faa427ca79e47aeb77fa258\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8504f0225ea54ebc9bfe31104d96b5d1 : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":6},{"id":"ff2e18db99eab5afcec6b6ecc615a29e7d3b4692","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:51:43.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/911b2646d5c248619d300a5a00eb12bc\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c01353631294766afeed257d611a7ee : SUCCESS in 5m 54s\n- grenade https://zuul.opendev.org/t/openstack/build/0329ad06bcce4156b5e773cc8c34a5cd : SUCCESS in 54m 53s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/27e12ec2405544bcbb6b2a04a30bb3fb : SUCCESS in 28m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/847dea1c034442bfbbb291abe3f77e2d : SUCCESS in 41m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6bd6d99c30034c8f8d1301da7f087e2f : SUCCESS in 59m 53s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0172d1c6ab6a46cebe53434c59ea62a8 : SUCCESS in 20m 04s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1751b3f8720e49be92c5e11179dda728 : SUCCESS in 16m 11s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f38efa3dd07a4068a5f1816bb0061ed6 : SUCCESS in 17m 39s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/7c6ed17e3bd84a9e954042854965dcaa : SUCCESS in 6m 56s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9a0a0684aebc4096bbab44bb87d8f77d : SUCCESS in 18m 28s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d84e80a01d4a49b5b50497438705df29 : SUCCESS in 24m 15s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/66866eedb3b740c389b161a3bb4c5b34 : SUCCESS in 20m 26s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/39bcaeeff3f14cfcb1bbd765c77e0e75 : SUCCESS in 28m 52s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c5e292320c6e4332b117bb190752bfda : SUCCESS in 14m 52s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/c9ea24459f7f47af90b5e021f8ff5b37 : SUCCESS in 15m 26s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/11b2fd3e0af843c69bb9adbdc4eca8c3 : SUCCESS in 12m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/5d7245327e1f488f8da895e8760dd8d6 : SUCCESS in 6m 40s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/2a576cfd6ec4433f971e036847725841 : SUCCESS in 17m 09s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/751870fbe2eb44dfbae5eb422dd54108 : SUCCESS in 8m 03s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/31e75193ac3d4ca79234edf7047850a0 : SUCCESS in 46m 31s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/a5435afe3d9a414ca2308bbb2ff54beb : SUCCESS in 43m 18s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/846bde0bdc824e2c8902b447d7ac936f : SUCCESS in 34m 33s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1497bb37edd84541b3b84475bc742295 : SUCCESS in 31m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3f42314731834cb3bd0fb49c5fe1d601 : SUCCESS in 5m 33s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2107981b522140a59e080b144f1ee325 : SUCCESS in 15m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fa6d29483eb245afb2d0a557f387f243 : SUCCESS in 54m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c0a38e60c1914986a782a3d3389890c0 : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":6},{"id":"22373ee5c4011fe15f55076a7eab9ede57406736","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:52:31.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"741ee9f7bd2bfcdb3fed815fd10305c505fd1a10","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:23:45.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a427fbc71aa48a388153a27e0c30ea4\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5882a0e166fd47eaa5ac7553dd5cbeb0 : SUCCESS in 10m 47s\n- grenade https://zuul.opendev.org/t/openstack/build/48a69f5d121b4710ae8303ea7fce217e : SUCCESS in 57m 47s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8d70f7cfd4af4b669d1a22c4e89b7572 : SUCCESS in 43m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/3a17affa77884460a80e12519c1450e4 : SUCCESS in 56m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/927fbc990a41488084f5fc3b7d8737d7 : SUCCESS in 39m 23s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/8f5c68b9ae104868a4b53bda9dec1803 : SUCCESS in 10m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/dfe42be3d69d4f2bb50f85831c99552f : SUCCESS in 13m 21s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b977eb373f1948a8b21c8359aceff422 : SUCCESS in 16m 43s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/500a415d931b4e39b21acdc2e721bf4f : SUCCESS in 14m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7e990252704d44f0befe5cc73c205dd2 : SUCCESS in 28m 36s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/cc14b68350be49a78cd07677f06bb01a : SUCCESS in 15m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4157d34e4b2e49b286d2ed8116c44583 : SUCCESS in 17m 55s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df2125516ec444e6ad1ab894f596a1d5 : SUCCESS in 15m 10s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/18214a6b44e54ce58fe137d4066a5d7e : SUCCESS in 4m 00s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/9f954d2f2ff4450b8907406f36937c47 : SUCCESS in 5m 08s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ad74a2364ef4a00972a7f210f469c2a : SUCCESS in 56m 37s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ff99cc4608d44aa7a539f1f918a042a3 : SUCCESS in 46m 56s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/07cceaf1c6e64d61a5857cac84109c67 : SUCCESS in 27m 31s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/472090ba4c234215a82a9d0cbf8407a6 : SUCCESS in 12m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/456739272a6c44b69ac1770dde05da3b : SUCCESS in 4m 22s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/95d9d396881d4b9b90dd5e7ed14985d2 : SUCCESS in 15m 11s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c215d407ec2d4bbc80541d5bb8fe8c50 : SUCCESS in 40m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/59e9634c700a426292d26413dab9fe12 : SUCCESS in 57m 42s","accounts_in_message":[],"_revision_number":6},{"id":"1692cfa92c2e873db437a2e232809015b38c016c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:23:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"426443308d318684dbb8017f7ea32351600613b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 20:24:50.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36b7513d02f946d190c4061908b15fc7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c78ddd97d4746e2b910b62a6972e353 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7761a37f6e8847239264ef98840483c2 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"cbc3b42d7d8d9280b415f909bce05332689cdf21","revisions":{"d65b17e03c474112ab26c6d4457c67d2ed19c2ea":{"kind":"REWORK","_number":1,"created":"2025-06-12 13:58:44.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/952462/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/952462/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/952462/1"}}},"commit":{"parents":[{"commit":"bc41a2c1b51836ed50085b7398f0fe77b9a1518c","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bc41a2c1b51836ed50085b7398f0fe77b9a1518c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"subject":"s3api: verify checksums on DeleteObjects POSTs","message":"s3api: verify checksums on DeleteObjects POSTs\n\nS3 will verify a checksum given with a DeleteObjects POST request,\nunlike a CompleteMultipartUpload POST [0].\n\nAlso, although not required [1], a Content-MD5 value will be verified\nif given with a DeleteObjects POST.\n\n[0] Related-Change: Id39fd71bc59875a5b88d1d012542136acf880019\n[1] Related-Change: Ifbcde9820bee72d80cab0fe3e67ea0f5817df949\n\nChange-Id: If25a8f0a3079558544ab15c874eda666a9f69933\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d65b17e03c474112ab26c6d4457c67d2ed19c2ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/d65b17e03c474112ab26c6d4457c67d2ed19c2ea"}]},"branch":"refs/heads/master"},"3ea3ba30f2dae41cac88acd0d4304d2ab5496d13":{"kind":"REWORK","_number":2,"created":"2025-07-11 13:54:49.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/952462/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/952462/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/952462/2"}}},"commit":{"parents":[{"commit":"be56c1e25822a23a6f6ad30d190782a395f3149f","subject":"s3api: Validate additional checksums on upload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/be56c1e25822a23a6f6ad30d190782a395f3149f"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-11 13:54:37.000000000","tz":60},"subject":"s3api: verify checksums on DeleteObjects POSTs","message":"s3api: verify checksums on DeleteObjects POSTs\n\nS3 will verify a checksum given with a DeleteObjects POST request,\nunlike a CompleteMultipartUpload POST [0].\n\nAlso, although not required [1], a Content-MD5 value will be verified\nif given with a DeleteObjects POST.\n\n[0] Related-Change: Id39fd71bc59875a5b88d1d012542136acf880019\n[1] Related-Change: Ifbcde9820bee72d80cab0fe3e67ea0f5817df949\n\nChange-Id: If25a8f0a3079558544ab15c874eda666a9f69933\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ea3ba30f2dae41cac88acd0d4304d2ab5496d13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3ea3ba30f2dae41cac88acd0d4304d2ab5496d13"}]},"branch":"refs/heads/master"},"9449588d83004d86a90867196130081e9bc6008f":{"kind":"REWORK","_number":3,"created":"2025-07-18 09:19:18.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/952462/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/952462/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/952462/3"}}},"commit":{"parents":[{"commit":"8af485775acc9ad97a6cbdca64279aac9f647f8c","subject":"Merge \"s3api: Add support for crc64nvme checksum calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af485775acc9ad97a6cbdca64279aac9f647f8c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-18 09:19:05.000000000","tz":60},"subject":"s3api: verify checksums on DeleteObjects POSTs","message":"s3api: verify checksums on DeleteObjects POSTs\n\nS3 will verify a checksum given with a DeleteObjects POST request,\nunlike a CompleteMultipartUpload POST [0].\n\nAlso, although not required [1], a Content-MD5 value will be verified\nif given with a DeleteObjects POST.\n\n[0] Related-Change: Id39fd71bc59875a5b88d1d012542136acf880019\n[1] Related-Change: Ifbcde9820bee72d80cab0fe3e67ea0f5817df949\n\nChange-Id: If25a8f0a3079558544ab15c874eda666a9f69933\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9449588d83004d86a90867196130081e9bc6008f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/9449588d83004d86a90867196130081e9bc6008f"}]},"branch":"refs/heads/master"},"dd9b21195e777b820fc32c7976ea5f68315e5521":{"kind":"REWORK","_number":4,"created":"2025-07-18 09:37:02.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/952462/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/952462/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/952462/4"}}},"commit":{"parents":[{"commit":"8af485775acc9ad97a6cbdca64279aac9f647f8c","subject":"Merge \"s3api: Add support for crc64nvme checksum calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af485775acc9ad97a6cbdca64279aac9f647f8c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-18 09:36:52.000000000","tz":60},"subject":"s3api: verify checksums on DeleteObjects POSTs","message":"s3api: verify checksums on DeleteObjects POSTs\n\nS3 will verify a checksum given with a DeleteObjects POST request,\nunlike a CompleteMultipartUpload POST [0].\n\nAlso, although not required [1], a Content-MD5 value will be verified\nif given with a DeleteObjects POST.\n\n[0] Related-Change: Id39fd71bc59875a5b88d1d012542136acf880019\n[1] Related-Change: Ifbcde9820bee72d80cab0fe3e67ea0f5817df949\n\nChange-Id: If25a8f0a3079558544ab15c874eda666a9f69933\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9b21195e777b820fc32c7976ea5f68315e5521"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/dd9b21195e777b820fc32c7976ea5f68315e5521"}]},"branch":"refs/heads/master"},"8d0a2d48e72fef4770ec991d283b4c53e59a6713":{"kind":"REWORK","_number":5,"created":"2025-11-17 14:56:34.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/952462/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/952462/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/952462/5"}}},"commit":{"parents":[{"commit":"8af485775acc9ad97a6cbdca64279aac9f647f8c","subject":"Merge \"s3api: Add support for crc64nvme checksum calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af485775acc9ad97a6cbdca64279aac9f647f8c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-17 14:46:36.000000000","tz":0},"subject":"s3api: verify checksums on DeleteObjects POSTs","message":"s3api: verify checksums on DeleteObjects POSTs\n\nIf a checksum header is sent with an S3 API DeleteObjects POST request\nthen it will be verified (unlike a CompleteMultipartUpload POST) [0].\n\nIf a checksum header is not sent with an S3 API DeleteObjects POST\nrequest then a Content-MD5 header is required.  Even when not\n*required* [1], a Content-MD5 value will be verified if sent.\n\nPreviously we could return 200 when we should have sent 400 (eg, when\nthe client sent a sha256 but no md5 or additional checksum), or return\n400 when we should have sent 200 (eg, when an additional checksum was\nsent with no sha256 and the auth protocol didn\u0027t require a sha256).\n\n[0] Related-Change: Id39fd71bc59875a5b88d1d012542136acf880019\n[1] Related-Change: Ifbcde9820bee72d80cab0fe3e67ea0f5817df949\n\nChange-Id: If25a8f0a3079558544ab15c874eda666a9f69933\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-authored-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d0a2d48e72fef4770ec991d283b4c53e59a6713"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8d0a2d48e72fef4770ec991d283b4c53e59a6713"}]},"branch":"refs/heads/master"},"cbc3b42d7d8d9280b415f909bce05332689cdf21":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-11-17 15:04:46.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/62/952462/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/62/952462/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/62/952462/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/62/952462/6"}}},"commit":{"parents":[{"commit":"8af485775acc9ad97a6cbdca64279aac9f647f8c","subject":"Merge \"s3api: Add support for crc64nvme checksum calculation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8af485775acc9ad97a6cbdca64279aac9f647f8c"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-06-12 13:51:34.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-11-17 15:04:25.000000000","tz":0},"subject":"s3api: verify checksums on DeleteObjects POSTs","message":"s3api: verify checksums on DeleteObjects POSTs\n\nIf a checksum header is sent with an S3 API DeleteObjects POST request\nthen it will be verified (unlike a CompleteMultipartUpload POST) [0].\n\nIf a checksum header is not sent with an S3 API DeleteObjects POST\nrequest then a Content-MD5 header is required.  Even when not\n*required* [1], a Content-MD5 value will be verified if sent.\n\nPreviously we could return 200 when we should have sent 400 (eg, when\nthe client sent a sha256 but no md5 or additional checksum), or return\n400 when we should have sent 200 (eg, when an additional checksum was\nsent with no sha256 and the auth protocol didn\u0027t require a sha256).\n\n[0] Related-Change: Id39fd71bc59875a5b88d1d012542136acf880019\n[1] Related-Change: Ifbcde9820bee72d80cab0fe3e67ea0f5817df949\n\nRelated-Bug: #2131671\nChange-Id: If25a8f0a3079558544ab15c874eda666a9f69933\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\nCo-authored-by: Tim Burke \u003ctim.burke@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbc3b42d7d8d9280b415f909bce05332689cdf21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/cbc3b42d7d8d9280b415f909bce05332689cdf21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
