)]}'
{"id":"openstack%2Fswift~952725","triplet_id":"openstack%2Fswift~master~I4eff901aaa334c8a73cebfc578cea14d23e6c365","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"1179":{"account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"last_update":"2025-07-31 18:56:21.000000000","reason":"\u003cGERRIT_ACCOUNT_1179\u003e replied on the change","reason_account":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-07-31 20:32:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4eff901aaa334c8a73cebfc578cea14d23e6c365","subject":"Clarify why there\u0027s a ShardRange.__hash__","status":"MERGED","created":"2025-06-16 20:56:54.000000000","updated":"2025-07-31 20:33:40.000000000","submitted":"2025-07-31 20:32:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"952725","meta_rev_id":"c2490ac645e02439031b884db8981e2da5bce810","_number":952725,"virtual_id_number":952725,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-31 20:32:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-07-31 18:56:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-07-31 18:56:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-16 21:41:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-06-16 22:01:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-30 21:33:58.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"CC"},{"updated":"2025-07-31 18:56:21.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"84033a5ce7bedddc1fe2aa7f7d64938a056a44cd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-16 20:56:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58dce551ed7f0c55c0d47962d6917c762511990d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 21:41:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4cb50068a3c64bebb1fbe5c602d57e38\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/ab2c3bdefe74497cab968c63154351ac : SUCCESS in 42m 41s","accounts_in_message":[],"_revision_number":1},{"id":"f188d78eb1a3f75ec665890b238222ef4ac78e79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 22:01:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/edc01a587bab4121895d6939c0b94545\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94dd8b24309548e89c85c0a6c8b90658 : SUCCESS in 10m 18s\n- grenade https://zuul.opendev.org/t/openstack/build/c2c82b610efa44e9b025f9f93789d335 : SUCCESS in 55m 52s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f43754a26f03481f9a2d7a638d0bd02d : SUCCESS in 56m 35s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8185d4f17f814aabbc0f65a5fb6d15fb : SUCCESS in 50m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5721f98070243e9a4ac672780b6f548 : SUCCESS in 59m 57s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b8a52bd3a0474158a35f4b4e94f48fe5 : SUCCESS in 18m 34s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d2bdafb29f544913aed89aec96e7d40c : SUCCESS in 15m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/fd2b9b6df01741f6a0f143540b4998dc : SUCCESS in 19m 57s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/c3d583de49f84041b0e8d632de2f9480 : SUCCESS in 4m 14s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a3dc66ce6c8c4155b05bcb0be8810681 : SUCCESS in 16m 38s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/aa4c4b3cf1334562ad7ced8b1cecabcf : SUCCESS in 23m 27s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9676249c07bf4ce08d05b6b73b02fab4 : SUCCESS in 26m 52s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/7c2aab91bb194a4380e1d75d78008076 : SUCCESS in 19m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9f385d06f613472ca9f22d9ed61ce045 : SUCCESS in 11m 59s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/772c3dff1c1f4e7ba727b434ec38ebf9 : SUCCESS in 12m 33s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/7a9f577a2d2e4caabf177ed44fe80c06 : SUCCESS in 14m 02s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/b8f8eb9fdad84dc895c85ebed70d5955 : SUCCESS in 5m 38s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/8614c4d1033e4f9dba8ebab18c6d7b5c : SUCCESS in 14m 30s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/177727a7bcf2484f8694a136195b8d08 : SUCCESS in 6m 25s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/b35df35dcb734d3ca7c1139011a0802f : SUCCESS in 58m 58s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/db1d80761021437f928a80faf0078036 : SUCCESS in 36m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/32dab256f7934e06a86993e99da92cae : SUCCESS in 49m 46s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/518ba96883a149c3b713b649290bb986 : SUCCESS in 20m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a4b6b93d18a43ad82f5be8e4076aa0f : SUCCESS in 3m 53s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9d50565b02cc4c78ad83ac56b5dd66a3 : SUCCESS in 10m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/2100a383a6ac4e9a9039429700fb43a3 : SUCCESS in 36m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d06220d31a44772b796a323cff3a429 : SUCCESS in 53m 47s","accounts_in_message":[],"_revision_number":1},{"id":"9686e79cf566691296d39670f14fcb8275487fea","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-30 21:33:58.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"84ea48724fc546c5542fb6614f279166c6b40901","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-30 22:19:17.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3400f716e27c36e44bd668afac16f44ad3b3f8c7","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-07-31 18:56:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"ead2171a88e3c4d5b04eb000a23de78f3c9e11ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 18:57:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2ef44e1ce3f64cffe2473bcb10c6080b50aed4d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 20:32:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01de701e7da9497ea27569dbf471c489\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/547dc671ac2d4d13a9e47cdaacd1fa75 : SUCCESS in 6m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/9cc6338e48bb45c59b4a3b0151f24635 : SUCCESS in 54m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/18f0be85d9244ba4804de3c7c1981bac : SUCCESS in 36m 46s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/0d9563dff669404ca81991a7ecfb50f6 : SUCCESS in 46m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72671fb24b6e42c2a35373bb56480043 : SUCCESS in 37m 01s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/5116a5765dab4fe6b54a1a79c73a6058 : SUCCESS in 5m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3dbfbec0b05246ee80e512190bceca04 : SUCCESS in 10m 02s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/5f75a7d70aa94ddcb7197b8946425671 : SUCCESS in 10m 52s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/9d7267cdcec54632aa5078dbf48e4c2c : SUCCESS in 26m 03s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/20b01251f1084bfaa46daba2177ba7d4 : SUCCESS in 9m 22s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/0585a707fc3c43fd8528d4d143d379e5 : SUCCESS in 9m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/dfbcad9004e9406f94d827ddadb0fbaf : SUCCESS in 9m 46s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/601bda199f6541138b95a39c5c56e0ef : SUCCESS in 5m 44s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7246dc421a24417ea2a2662a1f899e4b : SUCCESS in 6m 07s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/006880d41fc04f97926f294cc6588c93 : SUCCESS in 49m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/3fb26e84ac9d4e6792143730dcbcdfab : SUCCESS in 26m 10s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/8412cde88c784e2ebee5723cc5cf59ec : SUCCESS in 26m 12s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/9f5264022bc542dabb3091ba8a10ee29 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/59437c9f66fe499fbafe41302745300a : SUCCESS in 3m 58s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b97895320cbc4cabadffab06d7b81b2e : SUCCESS in 15m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/07024079fd2a48a4ae1637fbd77ff2bd : SUCCESS in 43m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ed813a7f62a244ed993932272e46f0ed : SUCCESS in 52m 07s","accounts_in_message":[],"_revision_number":1},{"id":"660db2d0aa343a5798557f75e08f39a04986d705","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 20:32:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c2490ac645e02439031b884db8981e2da5bce810","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 20:33:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/27a66e4955e14fa598cf11e2866307cb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86a025f061a749018960ec092d9e0dd0 : SUCCESS in 45s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/0c39adbc380a44528eed00b14abb4923 : SUCCESS in 39s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c177e7dcff31671ea2bd4de1da206bb286296568","revisions":{"c177e7dcff31671ea2bd4de1da206bb286296568":{"kind":"REWORK","_number":1,"created":"2025-06-16 20:56:54.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/25/952725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/25/952725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/952725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/952725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/25/952725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/25/952725/1"}}},"commit":{"parents":[{"commit":"005d69d1a9f3eb355901b135c854a4c4984e3540","subject":"Drop remaining skip check for Python \u003c 3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/005d69d1a9f3eb355901b135c854a4c4984e3540"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-16 20:28:56.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-16 20:28:56.000000000","tz":-420},"subject":"Clarify why there\u0027s a ShardRange.__hash__","message":"Clarify why there\u0027s a ShardRange.__hash__\n\nChange-Id: I4eff901aaa334c8a73cebfc578cea14d23e6c365\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c177e7dcff31671ea2bd4de1da206bb286296568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/c177e7dcff31671ea2bd4de1da206bb286296568"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
