)]}'
{"id":"openstack%2Fswift~953615","triplet_id":"openstack%2Fswift~master~I80142c6c0654b65b5755e7e828bcc4969a10f4f1","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"597":{"account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"last_update":"2025-06-27 23:56:06.000000000","reason":"\u003cGERRIT_ACCOUNT_597\u003e replied on the change","reason_account":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-07-07 19:15:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I80142c6c0654b65b5755e7e828bcc4969a10f4f1","subject":"Fix traceback in invalidate_hash","status":"MERGED","created":"2025-06-27 16:16:52.000000000","updated":"2025-07-07 19:16:47.000000000","submitted":"2025-07-07 19:15:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"953615","meta_rev_id":"ca6334fae103c0dcfb7f1813a429ccc2d5399dca","_number":953615,"virtual_id_number":953615,"owner":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-07 19:15:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":2,"date":"2025-06-27 23:56:06.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"all":[{"value":1,"date":"2025-07-07 18:24:20.000000000","_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-27 16:47:26.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"CC"},{"updated":"2025-06-27 17:24:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-27 23:56:06.000000000","updated_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"reviewer":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"state":"REVIEWER"}],"messages":[{"id":"c415b7c84d885c0e2ece56548a5752e0dbdcffd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-27 16:16:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"84fc2ea6e29bbe61333dfb4db72387efb72a980d","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-06-27 16:47:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b46f766085e583091ae6d9c8211d68356ca5d43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 17:24:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/93490d2f21d74022bad18fa382f9a214\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4081f33102824012a2b4318660e1d785 : SUCCESS in 11m 03s\n- grenade https://zuul.opendev.org/t/openstack/build/3a23ddec3a9240fdaa35fff61ce5ce48 : SUCCESS in 1h 03m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e3d58b58d0bb452bb00b5e56f615c37e : SUCCESS in 1h 05m 17s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/ceae1ba65896410287013ae74652d8ef : SUCCESS in 48m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fbc0628e0a924507830fc832d4225956 : SUCCESS in 1h 02m 40s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/afcbd10468f14cad8ed77962a1b90e80 : SUCCESS in 9m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/747e9481f2b5422a80c7a0372dff313b : SUCCESS in 12m 26s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/2f82e55904754deab3d05ba5009040d9 : SUCCESS in 25m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/32eb866812fa4a239276851475a58b6c : SUCCESS in 5m 16s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a49bd753df4b42319aaed815cfabb90c : SUCCESS in 23m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/eb110ddd5b33442ca6570c846918eb54 : SUCCESS in 8m 59s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a56024a0ca1042ab86e9252571441bb6 : SUCCESS in 28m 00s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/043233215b5d4385a59622b715dbb22d : SUCCESS in 25m 15s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/9a2be81a1c484273a96dc0f52d06ac55 : SUCCESS in 14m 47s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/59e338eb18c64f8c8cd1ef77b7d29d91 : SUCCESS in 13m 41s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/0d73c62dae6843138b9600e0714146ab : SUCCESS in 15m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59e7a245534f47b09aa1c260254c48e6 : SUCCESS in 8m 10s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/ded28b4d9d544f789d9af521565ac66e : SUCCESS in 7m 26s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/13dba6284bf744ec850defcf05ddeddf : FAILURE in 10m 00s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/784cdb808d5a4a5c8ad12d1af75e8f41 : SUCCESS in 57m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b1771f0c79e4611ac74abb984de8f9d : SUCCESS in 46m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/acc11d5492694eaf946ec8a5adce8d70 : SUCCESS in 20m 06s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/a94784f29e4744b4bbc99ffb8fc1dacf : SUCCESS in 21m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ded43625f8f54fc19e8edb3c3293f0f8 : SUCCESS in 4m 40s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/94753ba405334128a61b5a4a5037f658 : SUCCESS in 20m 59s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/98ed4b3b26d24bf5a9992de8dd45fd95 : SUCCESS in 50m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/336e503c6b084bb5950632b97960d15b : SUCCESS in 58m 56s","accounts_in_message":[],"_revision_number":1},{"id":"7853be9f9092accbec466279a3b54284678f5e9e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 18:04:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d06ddcdfbe1455fac2c2d54841c9693\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/40a97225cbc746cf993036d3c2fe3269 : SUCCESS in 1h 42m 45s","accounts_in_message":[],"_revision_number":1},{"id":"033e283e786e9561a4621e1aafb298a3664e7884","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-27 18:11:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da36b069638677982b4f311020328059e4ae35b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-06-27 18:12:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2765b34cecbc4d88a0959faa73fccca7a9d40015","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 19:17:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1600b007ce4600b37b44ac4fd46967\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e4acfce64c64651bba05bca1b4c0eeb : SUCCESS in 1h 03m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4c12e78de9bf7fc5dbd2b9ad643c29f6d0680a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-27 19:29:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/48a227a96298431998f80310b4735d5f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da074a04e4eb427d956ff5d1a08a877d : SUCCESS in 8m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/0fddabe3573b4473a2f0ccf196f7b07f : SUCCESS in 59m 29s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/7f16805138a545bd9950a18d69a2929d : SUCCESS in 1h 13m 57s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/8692a200c23c45669f9914160a677d70 : SUCCESS in 44m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/593c09545478478684461edbbb552528 : SUCCESS in 1h 08m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e86dbbc5e4504eb096ac3c56e9dbbccc : SUCCESS in 19m 00s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/81414715196c43908641141338e57356 : SUCCESS in 10m 31s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/4003c8b15ee046539a78291594d46e92 : SUCCESS in 19m 12s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/6c1e2cbc595340929a51b1aa697d81ff : SUCCESS in 4m 27s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5d6395813daf4317a543a3d969e12568 : SUCCESS in 20m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/1bdf0ac4f42747bbafb3232b15aa9bcc : SUCCESS in 18m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/230141c7ed984b1693d6dfe25347943a : SUCCESS in 16m 35s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/51f87e3289a1408792901386dca077ed : SUCCESS in 24m 54s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/3cadc6943f1742229476f08147e306dd : SUCCESS in 13m 09s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/4895731e1b3e4ca39600b9ab9c6cb50e : SUCCESS in 9m 03s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/86465a42c4464058bb4390c27726d8ae : SUCCESS in 17m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/354f7fec6aa94e3888d5a864b3c84493 : SUCCESS in 3m 30s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/522687419fcd4a87a4d2401c451c0f06 : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/6432cb8ec62943b8a990c7aaa625f2b9 : SUCCESS in 4m 44s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/5e357efd8b9541438219307b3d72afe4 : SUCCESS in 37m 55s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/892deacb9bef46d28a7748d921a91375 : SUCCESS in 50m 33s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/fbcd0aef1bc94178b6bcf719e4b8a8c7 : SUCCESS in 40m 36s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/58e279a7ce2e41f1b998834c1119639d : SUCCESS in 21m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67a68d7000494e49a78361006d3bbb00 : SUCCESS in 5m 09s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/888fd7a26ef044a8ad3d4f82878d564d : SUCCESS in 21m 02s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/ccc57a91ac9f47a6984abc294233beb6 : SUCCESS in 44m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a46bcc2671546fd80108d5684537481 : SUCCESS in 32m 54s","accounts_in_message":[],"_revision_number":2},{"id":"da08e8076974b294171a60e1f2fe081a4a794a00","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-06-27 23:56:06.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"4cf82f73a9536e9383343732cd861919143367a7","author":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"},"date":"2025-07-07 18:24:20.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"99573cfc0557a1a10e6fce15912d440d907335b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 18:24:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"93309aacf02d66cca9ced951677ed6df92274b88","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 19:15:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0d3f98edb754cdb80c42e91f83068b0\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86c56a1fcdfb4dab95d6c83a2c1cbd2f : SUCCESS in 7m 29s\n- grenade https://zuul.opendev.org/t/openstack/build/b890bf1f79164b59ac90f80279a3e6cd : SUCCESS in 28m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f39f1f9109db4bed8a0bf373df249259 : SUCCESS in 40m 33s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/f475e70f8d9e434595085db31727e878 : SUCCESS in 29m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f05d59a3e180455a826b46e6f8ac03ff : SUCCESS in 48m 40s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/ad94a1f43c3447009e3016f68f6f2012 : SUCCESS in 4m 41s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/ddd872da8b3b4b27a5d230a2dd1b9507 : SUCCESS in 8m 00s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/84323f88f25e4db5a4b325a13371f08a : SUCCESS in 16m 17s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/278080f1362f4cf786280875e674cf50 : SUCCESS in 19m 26s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/8b59ef70ee5b462d9881a448b4c327ac : SUCCESS in 12m 11s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/9ceb7ee7b11c453684440b134812a91a : SUCCESS in 12m 38s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/fbefeb6f29144af49ebff805453e5cb6 : SUCCESS in 7m 58s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/3c5af8bd2d3748478f900dd0933a1359 : SUCCESS in 3m 26s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/26e7e5fcacfc4fd4bf4d0e5c336ba37d : SUCCESS in 6m 29s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/65afb71087e54301a6f623648116ff68 : SUCCESS in 33m 25s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/1f7290a27cfe48829f5bea37805e762c : SUCCESS in 32m 00s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/45b08f8a6eee4721a6abb4b2eeee595d : SUCCESS in 31m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0ed51f344c7c433db683d20c53c58f90 : SUCCESS in 7m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecae719f5f414054b1b99a76e814148f : SUCCESS in 2m 18s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/06473c51d39743c78e75cb1a3bb0784e : SUCCESS in 9m 00s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/93f00ec1835d400bb5fd6dfc5aa6f616 : SUCCESS in 41m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4be9a8169bf841f89b5e834580a36e91 : SUCCESS in 41m 01s","accounts_in_message":[],"_revision_number":2},{"id":"c2bd205b5fdca35e72dd491ebbe9c86f5783e128","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 19:15:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"ca6334fae103c0dcfb7f1813a429ccc2d5399dca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-07 19:16:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46791216a4604b4eb2ab5dfce5b8e0fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89f6aa60b6f24ff5a7377c89755265a3 : SUCCESS in 44s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/b1fb7902508e41f5b66c0db6d0c6a0e0 : SUCCESS in 36s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"3dba681005c28b7f1e49404e7e53b9de32f543c4","revisions":{"ccaec571d374497a65db18b288e39b554ca112c3":{"kind":"REWORK","_number":1,"created":"2025-06-27 16:16:52.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/953615/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/953615/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/953615/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/953615/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/953615/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/953615/1"}}},"commit":{"parents":[{"commit":"b2a0d96f6d135c46d7d5cb839584d3c2a01bdcb5","subject":"Merge \"Update SAIO document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2a0d96f6d135c46d7d5cb839584d3c2a01bdcb5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-27 16:01:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-27 16:16:41.000000000","tz":-420},"subject":"Fix traceback in invalidate_hash","message":"Fix traceback in invalidate_hash\n\nChange-Id: I80142c6c0654b65b5755e7e828bcc4969a10f4f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccaec571d374497a65db18b288e39b554ca112c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ccaec571d374497a65db18b288e39b554ca112c3"}]},"branch":"refs/heads/master"},"3dba681005c28b7f1e49404e7e53b9de32f543c4":{"kind":"REWORK","_number":2,"created":"2025-06-27 18:12:04.000000000","uploader":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"ref":"refs/changes/15/953615/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/15/953615/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/953615/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/953615/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/15/953615/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/15/953615/2"}}},"commit":{"parents":[{"commit":"b2a0d96f6d135c46d7d5cb839584d3c2a01bdcb5","subject":"Merge \"Update SAIO document\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b2a0d96f6d135c46d7d5cb839584d3c2a01bdcb5"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-27 16:01:54.000000000","tz":-420},"committer":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-27 18:10:38.000000000","tz":-420},"subject":"Fix traceback in invalidate_hash","message":"Fix traceback in invalidate_hash\n\nChange-Id: I80142c6c0654b65b5755e7e828bcc4969a10f4f1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dba681005c28b7f1e49404e7e53b9de32f543c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/3dba681005c28b7f1e49404e7e53b9de32f543c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":597,"name":"Pete Zaitcev","email":"zaitcev@kotori.zaitcev.us","username":"zaitcev"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
