)]}'
{"id":"openstack%2Fswift~953978","triplet_id":"openstack%2Fswift~master~I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0","project":"openstack/swift","branch":"master","attention_set":{},"removed_from_attention_set":{"7847":{"account":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"last_update":"2025-08-11 19:06:40.000000000","reason":"Change was submitted"},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2025-08-04 15:50:57.000000000","reason":"\u003cGERRIT_ACCOUNT_15343\u003e replied on the change","reason_account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}}},"hashtags":[],"change_id":"I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0","subject":"ssync-receiver: terminate session if subreq read times out","status":"MERGED","created":"2025-07-02 17:38:55.000000000","updated":"2025-08-11 19:07:46.000000000","submitted":"2025-08-11 19:06:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":20,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"953978","meta_rev_id":"34bbb27e14a4fc035dfe7cec9de0fc62a2300da9","_number":953978,"virtual_id_number":953978,"owner":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-11 19:06:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":2,"date":"2025-08-11 17:20:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-08-04 15:50:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"all":[{"value":1,"date":"2025-08-11 17:20:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-02 18:34:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-07-03 19:01:31.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-07-07 17:14:37.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-08-04 12:11:32.000000000","updated_by":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2025-08-04 15:50:57.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2025-08-11 17:20:11.000000000","updated_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"reviewer":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"state":"REVIEWER"}],"messages":[{"id":"3ba05af95cb1f29891c91c575e5f4f4c0eddeca0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-02 17:38:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28e7518e8979210f5409a1e1d075971420728252","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 18:34:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8902eef15df844678f774058a954b1ce\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/824d5515cd454d4f99f79226d151b9a2 : SUCCESS in 11m 06s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b611144c491947808364389cba07813b : SUCCESS in 20m 41s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5d39ef11e86d4c8fa2687e90670ac821 : SUCCESS in 16m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/bcb901071ad240838f8a39aa4251de93 : SUCCESS in 21m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/18fb2c7a0bd0470692d1e417b6057131 : SUCCESS in 6m 01s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/51e55d86b5e04503b24fcb7a1de470d1 : SUCCESS in 19m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/3c308b168bd74a3698d695e8aeacc989 : SUCCESS in 19m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/696ff808fd87406087bc176a7b3eff70 : SUCCESS in 26m 27s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/125b4023ce8d4c48b4cd70145e5f8a28 : SUCCESS in 24m 18s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/477a80ffcc254134be9f608e048d6655 : SUCCESS in 15m 53s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1e008c5ea5c34b51929ffb3b045aa321 : SUCCESS in 15m 29s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d5d69443b5274203a0987688f195887a : SUCCESS in 16m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/57aec7ff765547b998200c2d663de2de : SUCCESS in 52m 49s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3ab481c35d7f4452985ccf8731c43ad4 : SUCCESS in 42m 08s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/328c0aed7d6f417d915fff2cfe2faf4e : SUCCESS in 16m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c79e4a515d5849f7bb1661f490066a0b : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/c541c7cf47c543fd9b6d5d94ec3a5cf9 : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":1},{"id":"61cad68dcb5bc78ed16fb2b4cf919c4beb7cd7db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 20:26:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f1f1bda056b461494fcf7bd63c257e6\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6c8616b430f544a6b1fea6d632a0078a : SUCCESS in 1h 22m 35s","accounts_in_message":[],"_revision_number":1},{"id":"c9802ae0bcc803bb9c8d5df754871c09c377c6cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-03 16:05:19.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"a1b8753dc2627cdd1d291d8d2515d4f4ae2645d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 17:18:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26e5f81bb1294706aa4adf6d08bf57b2\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a215fa1e13d14aaf9c0e17a4e07411b4 : SUCCESS in 9m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/631ab246943a4abfb801e7e5fe5f9776 : SUCCESS in 20m 23s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/984b8fb3ed414c7087d84e31c1b5cb87 : SUCCESS in 20m 14s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e7407519c23b4a00b946773a10743a1a : SUCCESS in 22m 54s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bffcc7e907214b2a8dd7e00e3353a044 : FAILURE in 3m 36s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5f7dd2cb2751468196cac32c456ebd5b : SUCCESS in 15m 22s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/347834550b4947cf96f2a9580275f534 : SUCCESS in 16m 19s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2b3c93fcf2cf4fb6a4c3d56b3c4cb847 : SUCCESS in 21m 40s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/e28eb9c4906c4167bdedc29fe99fadcc : SUCCESS in 18m 59s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc0f7bbda6964f0caed0c188b03d9249 : SUCCESS in 13m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/3c60554a30dd42d59ea30bfe538b2266 : SUCCESS in 14m 17s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/771a040ef9b44fef8503860aed243f1f : SUCCESS in 15m 35s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/851d4568933a428bb1ecffd33d672658 : SUCCESS in 1h 02m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/5c0fe489bfb84619a3799b13793da62e : SUCCESS in 23m 39s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5e670a7cc7094d52b33439828a494f70 : SUCCESS in 17m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f87376694bb14bd6b9874b6f6621eae9 : SUCCESS in 4m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/a4223b14b5d94e9292c07427e0b62252 : SUCCESS in 26m 22s","accounts_in_message":[],"_revision_number":2},{"id":"945477fd49bc5b5d4e5595df67332130a6f3519d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-03 18:09:50.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ebb30bc696860384d66237e6fc476b6edd297d71","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-03 19:01:31.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"87cd76c3a759b9fd8d2279b3bf3edbfc3ac89611","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 19:11:49.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8647f144bc064806a0c9f712e4833dfd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b1d4d10dafd47608a107a4fa03d8257 : SUCCESS in 4m 48s\n- grenade https://zuul.opendev.org/t/openstack/build/aedcb856d38d49f59ff1519f46e4aefc : SUCCESS in 28m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/3bbeb9490de542daaf53c82bf5dee11d : SUCCESS in 32m 04s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/096fd9bd8b9c402f8dffd6e18094a13f : SUCCESS in 29m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f865c151d43643b68a6ed4e46e56522e : SUCCESS in 32m 23s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/df1ac7dfd7b34129a705c982b6a0890b : SUCCESS in 14m 29s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/32dc37ce4e1c4c9d9eebbb0e80762e4d : SUCCESS in 19m 16s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/44e87bd1e5fa4654bf79b618eb4cb808 : SUCCESS in 19m 55s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/b0c45c8982b14253a7b2792b595390ca : SUCCESS in 3m 37s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/4943273540b845bb986a943d36843a02 : FAILURE in 9m 07s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/42fe3c973c114260b0deab40079def25 : FAILURE in 7m 48s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/2fac3d7cfbab4f149a40faff88d3f995 : FAILURE in 28m 24s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/fb8c17ade71543f08d1817c25cbb3cd0 : FAILURE in 10m 20s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/72330a96c2544b24a8a439e310546a20 : SUCCESS in 12m 24s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/6264709a32c444009ddc9b47ccb74040 : SUCCESS in 14m 47s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/d2b18d68de32493fa0d834f247bee153 : SUCCESS in 7m 56s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c528234f3ef941e39d89be30b6d07555 : SUCCESS in 5m 32s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a0c8fed552714621a89dff7e588f50e9 : SUCCESS in 9m 22s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/eb21c08138ec4ebdb93369b5fbdc7f5a : SUCCESS in 7m 14s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9930d35fe0da49d88aedff11cbe5891c : SUCCESS in 56m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/b2035392dd884bea96f19b827c5dcf1a : SUCCESS in 36m 36s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/2e6d204041ef406b8ce0fc0c95441078 : SUCCESS in 32m 10s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/8bb43cf41e324fd98ff7a7b20c33e85f : FAILURE in 19m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/644a46e3dace4ab5a6922a83d1bd1821 : SUCCESS in 6m 46s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/668d11f00c23445caf84adc37aada91b : SUCCESS in 17m 51s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81ac9301283a4b5ca97029387e04f77b : SUCCESS in 31m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/965c81d8b40743b891de07e877c40f49 : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":3},{"id":"e452f2e1e2a3d056d66ef6e4ce00ffc4094a4a0d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 19:25:29.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8911e6a8660242ebbe2aff2627f64dba\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fbdc4b7bd210424eb655273486b884b0 : FAILURE in 48m 35s","accounts_in_message":[],"_revision_number":3},{"id":"7a976eb19739fe7598d4232975d2cdd864bfa566","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-04 15:37:58.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"511e33338eae2ff0ba5533acefc856174c340fd3","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-04 15:38:35.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fd18b411b7362a75ad0934a5df833d94f136b5f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 16:47:53.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/56812f36ff804aed8ee3daf0fe209da5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39cb303e8ffb4e8383d1e32d8343a2a8 : SUCCESS in 10m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/2f8523e953ff42288d9423c22801a9fe : SUCCESS in 1h 03m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5ac5bd48e6874477a0cdd6c3d0880c24 : SUCCESS in 55m 05s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/055d9a8723844b1a928e9818e59ea1c4 : SUCCESS in 50m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9f5b11fd3364171b9b74b8bc6c67eb0 : SUCCESS in 1h 02m 52s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/270b6b208c954999a11711dae012f842 : SUCCESS in 20m 33s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ef6ca2de8f6a443d83a758de3b7cd995 : SUCCESS in 18m 37s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/c89a309a478249ca90879e300d402ded : SUCCESS in 21m 18s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/e311da105caf46b89ebaa9b19442cbae : SUCCESS in 5m 21s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/437df4bf1cd14632960422faef5bfce9 : SUCCESS in 19m 45s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/4a8bf3cee3a14575af3c88180d1312c2 : SUCCESS in 10m 20s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/91e3621324714a7fa67af5d44783b442 : SUCCESS in 26m 39s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3d9abcf6efef41609bc57c3cab4dfe58 : SUCCESS in 23m 19s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/00dba5f131954133a1c870f892e01ff7 : SUCCESS in 14m 25s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/da321ee47a8b4826820f9f88f2b6cdd7 : SUCCESS in 13m 30s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c27b4fcdb84a4f75915d5d9fa1c1ee07 : SUCCESS in 15m 31s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/a77a17180f9745809a915b875d07fa9c : SUCCESS in 6m 03s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5fd19bd9243b4b00bcd7e49a141d12f7 : SUCCESS in 15m 23s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/f4c727d9d0e04741a5748f752f58cdd1 : SUCCESS in 9m 31s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/712f908c830e4493bc51a51bec930549 : SUCCESS in 55m 06s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/7b66c1ada32f462488cd960ba7afea93 : SUCCESS in 37m 32s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/4b728cf598b046e9839bae24ae5ac6a1 : SUCCESS in 39m 44s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/edeefc1bec994f048d98f01bc45cfc15 : SUCCESS in 8m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01d6f21cbe3e41f0b523e290bdb3216e : SUCCESS in 4m 03s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/ede91d618eae432fa755848988ea0857 : SUCCESS in 19m 56s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3957de27132a400980fdaf72139a858e : SUCCESS in 48m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9a06fe949fdd440d8feceb929b0881e0 : SUCCESS in 54m 20s","accounts_in_message":[],"_revision_number":4},{"id":"03b584ae723011fa0301015b3d2386245660bdd4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-04 18:33:34.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f1611efc0ac4728a839fc3c66f17e8f\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/0759600631504b479d3c7a7d5459584b : SUCCESS in 45m 34s","accounts_in_message":[],"_revision_number":4},{"id":"a0804e180d6397fcc33e5da1bec920210855221e","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-07-07 14:59:47.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"fca0866029bd729a919f88aecee04d9fb7514315","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-07 17:14:37.000000000","message":"Patch Set 4: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"3ac6d1140051ef6571f617de67ebc22c9261e6f3","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-07-28 02:32:16.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"76ac9ba299690db78a52edc4879e3b5aedb19f9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-04 12:11:32.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d0669c4d83afc5f51f04a103d88c9affc1d54705","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 13:22:50.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bd32f0d8c77e4ae7802227ce6f1c6875\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e4bd39f5c7e44068f7668c6146f521c : SUCCESS in 10m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/72ec486925794e70b2074d499f321899 : SUCCESS in 56m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/bb7b79e66f4147bfb87bf29dc46dcef7 : SUCCESS in 1h 00m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/eb1715802ade4f0ea99f3e0fbbab57cc : SUCCESS in 50m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/826f7adeac814bd1bfa808a0f8da2035 : SUCCESS in 58m 42s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/12934f66a868498cb3211b9876d42b9b : SUCCESS in 17m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/89ac295ab7834730992c26ae2ad00dde : SUCCESS in 17m 17s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b348521ca3674fc09a102a8c1cbdd3c8 : SUCCESS in 20m 32s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/87c002595f3943e4997927c6b972c45f : SUCCESS in 4m 44s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/28763f18f45a49248a5e5e3504f6adec : FAILURE in 16m 41s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/89ad49b82638422d9c3805928094526e : FAILURE in 20m 06s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/19ea48e75c7c47fe8b6656971e651ccf : FAILURE in 20m 19s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/79cdff2ad5b5464b97bfb56065857345 : FAILURE in 30m 30s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/bb55aa38a61344b0bfb0cf3efd41e06f : SUCCESS in 12m 02s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a3e7cf310529471faca930ceafdd9466 : SUCCESS in 13m 50s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1e47d48ad7424681a2388410c1bed41c : SUCCESS in 15m 08s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/05548461a6494d7596ffbe3d24f41ce6 : SUCCESS in 6m 36s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/e3d143e801c24d74a2352504df77711d : SUCCESS in 14m 00s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/e62ee0ba6fcc4020b565dc2a5924c93b : SUCCESS in 7m 55s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/99e0e8234fc14d04b8891d6e31e39ce8 : SUCCESS in 50m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/82324c866e044c588f22b665bbe5f217 : SUCCESS in 40m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/361d571b992541e585fa129264e79bf5 : SUCCESS in 39m 42s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/06bd4e29f9664fa2ab4e87186103ccb9 : FAILURE in 17m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c953cbe1c5b444caa7119df7beb2a2bc : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/1212b67a1b4648f48ec8018441418b84 : SUCCESS in 17m 18s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/81c8668ec7504f9693fb52b0d33569f7 : SUCCESS in 57m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1d5523d1cb2c45fb899daeb2b17787fc : SUCCESS in 46m 26s","accounts_in_message":[],"_revision_number":5},{"id":"a11b8b3af3d551576274275213c384bb4a5ba0c7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 13:28:21.000000000","message":"Patch Set 5:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700229595f404d3aa64aaf6eb6a70706\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a1496a880cb44fb18b687af15401ff27 : FAILURE in 1h 14m 57s","accounts_in_message":[],"_revision_number":5},{"id":"b454981d62b4cd99fd91bec40397a8639b491e4f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"date":"2025-08-04 14:01:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"9f3771906ce6b69ca4a13f5585796fd7add38872","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 15:14:31.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ab090f727524f4094639ac4e6a4f475\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a0cb4650d3e4ea783bd536a9b0a340d : SUCCESS in 4m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/4a80429c50e14cd6adf16832fa1b3695 : SUCCESS in 1h 07m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ad300b1fb01146c7a1da9f0a9aaf8e9c : SUCCESS in 1h 01m 36s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/a360b4a602474cd8b68f5f5a554761e6 : SUCCESS in 44m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b5311ee61274af8a2cfeef8b26678ab : SUCCESS in 1h 01m 48s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9fe0a1f1ce784f93939e4ee3be3fa026 : SUCCESS in 10m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/5980404c60d648fdab1df39d397f9794 : SUCCESS in 19m 49s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/eb85c1977f774c60b8b3666e5d2bd90b : SUCCESS in 17m 05s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/ff25780521294a19a761b295e1ad3d94 : SUCCESS in 3m 49s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/9572ab3ab9274f48a39a893d9e1f2099 : SUCCESS in 24m 24s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/9d8ac28c39e74a03bfb6dc2908f0fe53 : SUCCESS in 19m 56s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/7d6fd560803343f3a5d33a14ba97f38f : SUCCESS in 23m 51s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/69d378c22a784f44898c94fe447ca17a : SUCCESS in 24m 06s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/f1966cfe10bc465399d8bd23eafc55eb : SUCCESS in 15m 04s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/910d1e895332441492d7f7de72c86da3 : SUCCESS in 13m 54s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/df498a5d8bcb4e6d86031cffd47acdcf : SUCCESS in 15m 30s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/fbcf496ea39b4c378333ebbe568e280f : SUCCESS in 4m 25s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/301c6195abce49baa0dc338e809b9636 : SUCCESS in 14m 24s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/7f250fb98a674abdb5e1efa734a6c4c5 : SUCCESS in 5m 01s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/9d20802db57844faa3c985b5ff941f00 : SUCCESS in 54m 30s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4e5d0420d3fd4046bc82e2d4995d446a : SUCCESS in 40m 12s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/ff39882a04cf4fa69f566038d262241f : SUCCESS in 21m 52s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/489d0be3a14a43f39340faa9a8163b58 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b4befecc8114b7c8fab9638cfd943ba : SUCCESS in 4m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/cfb77c6cc0f94c00bd1e475db0f26931 : SUCCESS in 17m 09s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ec6613c194e450e807dcf6524101f7b : SUCCESS in 50m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03af11ef1fd748b0ade44c189963260b : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":6},{"id":"c74cde980db8f308c906189133d9be28401306cd","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-04 15:18:33.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c99bac982014b99b2bd5c3f18837357\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8d37ac84518f4bf782a72b488208832c : SUCCESS in 1h 15m 17s","accounts_in_message":[],"_revision_number":6},{"id":"b0f4810a6bda74c45ff2677bd0c2ba1b3a9619d6","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-08-04 15:50:57.000000000","message":"Patch Set 6: Code-Review+2\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"4b77f30265c72808d43c211580e7ca028fd72ce6","author":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"},"date":"2025-08-11 17:20:11.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e7fd9c40cf896bc19e6a51450bda2ee12ff88618","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 17:20:35.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"c2f1b03bafd98d6d3fa779fa91b1ab7fd9dea3f3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:06:40.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d14c6b7ffff14f539a3b5af7d2da0252\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec84c4c5277c4c70abb6fa82eed6113b : SUCCESS in 8m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/e0e32b68dbed4accac5b5b051d8716e9 : SUCCESS in 59m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f20ddc796138403ead1b9f7711c5b369 : SUCCESS in 49m 56s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/4349a75148984d53a6007f0c7145d2fa : SUCCESS in 53m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/52e300c514c24d0e953cc59b6cde300e : SUCCESS in 1h 03m 12s\n- swift-upload-image https://zuul.opendev.org/t/openstack/build/2e836622026444dea981b2c31e7ab51b : SUCCESS in 7m 19s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/7c8e89df2292493289bfb44e50dc6f19 : SUCCESS in 19m 13s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/b277484b5b7c45eeb8aff4225690e295 : SUCCESS in 19m 26s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/db4abc08990b4e3384be2cc3e21a9061 : SUCCESS in 21m 58s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/353e01fca55d4743bb2bf67d7a8ab313 : SUCCESS in 14m 12s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/a32cb855c21749979f662e192794c902 : SUCCESS in 15m 11s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/1beaf51f0fdf498a8513b439c7ca3ee1 : SUCCESS in 15m 17s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/f51c70f319204e26a20deb6e04fdd2e7 : SUCCESS in 8m 35s\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/44b92d4292f54cf395f378d5747d82ef : SUCCESS in 10m 32s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/4ad7e6a7b4a641e2978a1b02ff57e034 : SUCCESS in 56m 16s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/59ece43d183b4c7a94e0fc8d05abf154 : SUCCESS in 52m 43s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a7f922ce501b4acdb2ce10d701668c85 : SUCCESS in 56m 01s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/de49fc948e8e48df9032318057b7386b : SUCCESS in 19m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc6f2b00d16c4dceb84df0b12311d108 : SUCCESS in 5m 36s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/18d3a23081a444948f81da007959f483 : SUCCESS in 24m 35s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0746addd841c499f85d6116f1d9190c6 : SUCCESS in 48m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d4927c8e812146e8ae46cce7d9afffcf : SUCCESS in 56m 09s","accounts_in_message":[],"_revision_number":6},{"id":"4f5beb8f79eb5291c0dd87346643de1e00bc8fcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:06:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"34bbb27e14a4fc035dfe7cec9de0fc62a2300da9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-11 19:07:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ef6b7d17204f8b9c47b6d93e44aed3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e7fd8886b324fb09cdc8a9ef449c597 : SUCCESS in 51s\n- swift-promote-image https://zuul.opendev.org/t/openstack/build/7889328c420c4c318d0b3752c228bad4 : SUCCESS in 41s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"1dc3307eaf0457e59337ca9a153c12aaade6696b","revisions":{"7b38a636bf23cac3bff6fc4c65b270450c33103f":{"kind":"REWORK","_number":1,"created":"2025-07-02 17:38:55.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/953978/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/953978/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/953978/1"}}},"commit":{"parents":[{"commit":"98f444ef375e2d528371e8dc3205e0d54eb6e425","subject":"ssync: Raise ChunkReadError if we don\u0027t read a whole line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98f444ef375e2d528371e8dc3205e0d54eb6e425"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-02 17:38:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-02 17:38:02.000000000","tz":60},"subject":"ssync: more timeout unit tests","message":"ssync: more timeout unit tests\n\nChange-Id: I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b38a636bf23cac3bff6fc4c65b270450c33103f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/7b38a636bf23cac3bff6fc4c65b270450c33103f"}]},"branch":"refs/heads/master"},"5cabd9ad86214eb26949365b4d3d9fa22607da1f":{"kind":"REWORK","_number":2,"created":"2025-07-03 16:05:19.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/953978/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/953978/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/953978/2"}}},"commit":{"parents":[{"commit":"98f444ef375e2d528371e8dc3205e0d54eb6e425","subject":"ssync: Raise ChunkReadError if we don\u0027t read a whole line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98f444ef375e2d528371e8dc3205e0d54eb6e425"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-02 17:38:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-03 15:07:57.000000000","tz":60},"subject":"ssync: more timeout unit tests","message":"ssync: more timeout unit tests\n\nChange-Id: I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cabd9ad86214eb26949365b4d3d9fa22607da1f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/5cabd9ad86214eb26949365b4d3d9fa22607da1f"}]},"branch":"refs/heads/master"},"4e94fc75d25075d088afc26db71e2b56afdd3513":{"kind":"REWORK","_number":3,"created":"2025-07-03 18:09:50.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/953978/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/953978/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/953978/3"}}},"commit":{"parents":[{"commit":"98f444ef375e2d528371e8dc3205e0d54eb6e425","subject":"ssync: Raise ChunkReadError if we don\u0027t read a whole line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98f444ef375e2d528371e8dc3205e0d54eb6e425"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-02 17:38:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-03 17:52:51.000000000","tz":60},"subject":"WIP ssync-receiver: terminate if subreq read times out","message":"WIP ssync-receiver: terminate if subreq read times out\n\nTODO: launchpad bug?\nTODO: consider different refactor\n\nIf a PUT subrequest body iterator times out, the object server will\nreturn a 500 response to the ssync receiver.  Previously, the ssync\nreceiver would attempt to drain the remainder of the subrequest body\niterator and then resume reading the next subrequest from the SSYNC\nbody. However, the attempt to drain the subrequest iterator would fail\nbecause the timeout caused the iterator to terminate.\n\nThe ssync receiver would therefore treat any remaining subrequest body\nas the preamble to the next subrequest. This remaining subrequest body\ncontent was likely to cause the protocol parsing to fail, but in the\nworst case could be erroneously interpreted as a valid subrequest.\n\n(The exact failure mechanism depends on what state the\neventlet.wsgi.Input is left in when the timeout fired.)\n\nThis patch ensures that the ssync receiver will terminate processing\nan SSYNC request if an exception occurs while reading a subrequest\nbody.\n\nChange-Id: I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e94fc75d25075d088afc26db71e2b56afdd3513"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/4e94fc75d25075d088afc26db71e2b56afdd3513"}]},"branch":"refs/heads/master"},"8ece3061d4889555919bc2da06e2f0574c64a442":{"kind":"REWORK","_number":4,"created":"2025-07-04 15:37:58.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/953978/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/953978/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/953978/4"}}},"commit":{"parents":[{"commit":"98f444ef375e2d528371e8dc3205e0d54eb6e425","subject":"ssync: Raise ChunkReadError if we don\u0027t read a whole line","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98f444ef375e2d528371e8dc3205e0d54eb6e425"}]}],"author":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-02 17:38:02.000000000","tz":60},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-07-04 15:36:08.000000000","tz":60},"subject":"ssync-receiver: terminate session if subreq read times out","message":"ssync-receiver: terminate session if subreq read times out\n\nIf a PUT subrequest body iterator times out while the object server is\nreading it, the object server will handle the timeout and return a 500\nresponse to the ssync receiver.\n\nPreviously, the ssync receiver would attempt to drain the remainder of\nthe subrequest body iterator and then resume reading the next\nsubrequest from the SSYNC body. However, the attempt to drain the\nsubrequest iterator would fail (silently) because the timeout had\nalready caused the iterator to exit.\n\nThe ssync receiver would therefore treat any remaining subrequest body\nas the preamble to the next subrequest. This remaining subrequest body\ncontent was likely to cause the protocol parsing to fail, but in the\nworst case could be erroneously interpreted as a valid subrequest.\n\n(The exact failure mechanism depends on what state the\neventlet.wsgi.Input is left in when the timeout fired.)\n\nThis patch ensures that the ssync receiver will terminate processing\nan SSYNC request if an exception occurs while reading a subrequest\nbody.\n\nCloses-Bug: #2115991\nChange-Id: I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ece3061d4889555919bc2da06e2f0574c64a442"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8ece3061d4889555919bc2da06e2f0574c64a442"}]},"branch":"refs/heads/master"},"b813b2916ab3f3a5775cd97f827befc57e22cffa":{"kind":"REWORK","_number":5,"created":"2025-08-04 12:11:32.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/953978/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/953978/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/953978/5"}}},"commit":{"parents":[{"commit":"220f791466fdeae7b81a6ba39ca8636a5187cb1f","subject":"Merge \"Clarify why there\u0027s a ShardRange.__hash__\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/220f791466fdeae7b81a6ba39ca8636a5187cb1f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-27 17:21:31.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-04 12:11:10.000000000","tz":60},"subject":"ssync-receiver: terminate session if subreq read times out","message":"ssync-receiver: terminate session if subreq read times out\n\nIf a PUT subrequest body iterator times out while the object server is\nreading it, the object server will handle the timeout and return a 500\nresponse to the ssync receiver.\n\nPreviously, the ssync receiver would attempt to drain the remainder of\nthe subrequest body iterator and then resume reading the next\nsubrequest from the SSYNC body. However, the attempt to drain the\nsubrequest iterator would fail (silently) because the timeout had\nalready caused the iterator to exit.\n\nThe ssync receiver would therefore treat any remaining subrequest body\nas the preamble to the next subrequest. This remaining subrequest body\ncontent was likely to cause the protocol parsing to fail, but in the\nworst case could be erroneously interpreted as a valid subrequest.\n\n(The exact failure mechanism depends on what state the\neventlet.wsgi.Input is left in when the timeout fired.)\n\nThis patch ensures that the ssync receiver will terminate processing\nan SSYNC request if an exception occurs while reading a subrequest\nbody.\n\nCloses-Bug: #2115991\nChange-Id: I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0\nCo-authored-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b813b2916ab3f3a5775cd97f827befc57e22cffa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/b813b2916ab3f3a5775cd97f827befc57e22cffa"}]},"branch":"refs/heads/master"},"1dc3307eaf0457e59337ca9a153c12aaade6696b":{"kind":"REWORK","_number":6,"created":"2025-08-04 14:01:42.000000000","uploader":{"_account_id":7847,"name":"Alistair Coles","email":"alistairncoles@gmail.com","username":"acoles"},"ref":"refs/changes/78/953978/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/78/953978/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/78/953978/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/78/953978/6"}}},"commit":{"parents":[{"commit":"220f791466fdeae7b81a6ba39ca8636a5187cb1f","subject":"Merge \"Clarify why there\u0027s a ShardRange.__hash__\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/220f791466fdeae7b81a6ba39ca8636a5187cb1f"}]}],"author":{"name":"Tim Burke","email":"tim.burke@gmail.com","date":"2025-06-27 17:21:31.000000000","tz":-420},"committer":{"name":"Alistair Coles","email":"alistairncoles@gmail.com","date":"2025-08-04 14:01:30.000000000","tz":60},"subject":"ssync-receiver: terminate session if subreq read times out","message":"ssync-receiver: terminate session if subreq read times out\n\nIf a PUT subrequest body iterator times out while the object server is\nreading it, the object server will handle the timeout and return a 500\nresponse to the ssync receiver.\n\nPreviously, the ssync receiver would attempt to drain the remainder of\nthe subrequest body iterator and then resume reading the next\nsubrequest from the SSYNC body. However, the attempt to drain the\nsubrequest iterator would fail (silently) because the timeout had\nalready caused the iterator to exit.\n\nThe ssync receiver would therefore treat any remaining subrequest body\nas the preamble to the next subrequest. This remaining subrequest body\ncontent was likely to cause the protocol parsing to fail, but in the\nworst case could be erroneously interpreted as a valid subrequest.\n\n(The exact failure mechanism depends on what state the\neventlet.wsgi.Input is left in when the timeout fired.)\n\nThis patch ensures that the ssync receiver will terminate processing\nan SSYNC request if an exception occurs while reading a subrequest\nbody.\n\nCloses-Bug: #2115991\nChange-Id: I585e8a916d947c3da8d7c0e8a85cf0a8ab85f7f0\nCo-authored-by: Tim Burke \u003ctim.burke@gmail.com\u003e\nSigned-off-by: Alistair Coles \u003calistairncoles@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dc3307eaf0457e59337ca9a153c12aaade6696b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/1dc3307eaf0457e59337ca9a153c12aaade6696b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":1179,"name":"Clay Gerrard","email":"clay.gerrard@gmail.com","username":"clay-gerrard"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
