)]}'
{"id":"openstack%2Fswift~954188","triplet_id":"openstack%2Fswift~master~I28d8a57da468dc3a051e96ee711b7523b0adb517","project":"openstack/swift","branch":"master","topic":"fix-s3-list-bucket-creation-date","hashtags":[],"change_id":"I28d8a57da468dc3a051e96ee711b7523b0adb517","subject":"Fixing constant creation-date for s3api list-buckets Closes-Bug: #1856938 Change-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61 Signed-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e","status":"NEW","created":"2025-07-05 21:05:04.000000000","updated":"2025-07-05 22:13:20.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96cbf3ab733b8b55b24b97eb03f2cb84a806a89f","_number":954188,"virtual_id_number":954188,"owner":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-05 22:13:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-05 22:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-05 22:13:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1049209de96eac44fd6b2072cc9c4d642b14d606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"date":"2025-07-05 21:05:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6715c3b7987f63f172992fe864b05b5984afa673","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 22:06:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02ccdae3deff4210872f4ad0d1a14766\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/73f24d5e708641c6ac0f0f16eefbde63 : SUCCESS in 59m 39s","accounts_in_message":[],"_revision_number":1},{"id":"96cbf3ab733b8b55b24b97eb03f2cb84a806a89f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-05 22:13:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81b996e7a8a34b859b1934e6337a99bb\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86e850345da24d7f8e40ae90947410b9 : SUCCESS in 11m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/cea81ef29bba45f9a371967c4d009581 : SUCCESS in 1h 01m 41s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c7b148424fdc4c32b8d39feaedeb6811 : SUCCESS in 33m 41s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/aedb99e16ba7446b94e134a2c7029fdd : SUCCESS in 31m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f0316e030d7447e91dfa77f6c194807 : SUCCESS in 32m 21s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/0f448edf08be4264bf29424c185f02db : SUCCESS in 14m 54s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d0f500e6b02240eea0b795a8db7967f8 : SUCCESS in 15m 30s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/d6c88c015c9244c3ac225a89f2835a5e : SUCCESS in 19m 52s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/300121cb4d3f4140a3691b71b168abbe : SUCCESS in 7m 15s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/b7f2c4090ab14f38a89a3f7f1a2182f2 : SUCCESS in 8m 08s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/96648757ffe147a99aebba9468a6f99b : SUCCESS in 19m 18s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/6ef5dc6d37154a58aa4ec2c9692129f0 : SUCCESS in 10m 20s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/635f5862259d4bd79ef99eb72bdd5eea : SUCCESS in 10m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/c25d2cb5dd5d46c5aa914cc3dadf6712 : SUCCESS in 7m 51s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/ae6af6002a8c47ea9fe71a1008e47074 : SUCCESS in 7m 48s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/cb0a6d5daa6942faa6444e8e7cac7b77 : SUCCESS in 8m 01s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/cff2ec61b74d434e91c1d8af0b420f3d : SUCCESS in 5m 53s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/58c716ea44e84777bc57d016365875d8 : SUCCESS in 18m 39s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5bd748c1f2054616816404b9e06be7c5 : SUCCESS in 7m 45s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/32233b9a730c47c4917cf6758175deb7 : SUCCESS in 43m 00s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/377ecc176a804a48a29bdf413dee2d47 : SUCCESS in 19m 52s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6d38a9ee0662450d81f7a73d6cef205a : SUCCESS in 49m 34s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/aa28457aaf304fefa217c501c925efa2 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b66764f084a4234af132f21e1e7e7d8 : SUCCESS in 2m 14s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/b364480f01fc4825bf80c1b25fab5feb : SUCCESS in 15m 01s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fd43c8af7db142fd8fa61800933f3d3e : SUCCESS in 1h 02m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ed8aa38a6a04347aca73cde0ef3f5dc : SUCCESS in 28m 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"926e6801c80b3f290bc670f93993fe370bb6d646","revisions":{"926e6801c80b3f290bc670f93993fe370bb6d646":{"kind":"REWORK","_number":1,"created":"2025-07-05 21:05:04.000000000","uploader":{"_account_id":37337,"name":"Nathan Nguyen","display_name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","username":"nathang15"},"ref":"refs/changes/88/954188/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/88/954188/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/954188/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/954188/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/88/954188/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/88/954188/1"}}},"commit":{"parents":[{"commit":"076771462cf5333644170b069ad95e8bd999d024","subject":"Merge \"Fix the inconsistent use of printing errors\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/076771462cf5333644170b069ad95e8bd999d024"}]}],"author":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-05-08 01:20:44.000000000","tz":-240},"committer":{"name":"Nathan Nguyen","email":"nguyennathan1502@gmail.com","date":"2025-07-05 21:04:39.000000000","tz":-240},"subject":"Fixing constant creation-date for s3api list-buckets Closes-Bug: #1856938 Change-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61 Signed-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e","message":"Fixing constant creation-date for s3api list-buckets\nCloses-Bug: #1856938\nChange-Id: I3239e4c43ed18c02c94effc951cc1ad6066afc61\nSigned-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e\n\nChange-Id: I28d8a57da468dc3a051e96ee711b7523b0adb517\nSigned-off-by: Nathan Nguyen \u003cnguyennathan1502@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/926e6801c80b3f290bc670f93993fe370bb6d646"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/926e6801c80b3f290bc670f93993fe370bb6d646"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
