)]}'
{"id":"openstack%2Fswift~954193","triplet_id":"openstack%2Fswift~master~I7a5de760143488567c4930a0c893f7ce9cadf05b","project":"openstack/swift","branch":"master","attention_set":{"6968":{"account":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"last_update":"2026-07-29 17:24:04.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"15343":{"account":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"last_update":"2026-09-06 19:39:41.000000000","reason":"Added by \u003cGERRIT_ACCOUNT_37598\u003e using the hovercard menu","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-07-29 18:37:44.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I7a5de760143488567c4930a0c893f7ce9cadf05b","subject":"Drop lower-constraints.txt and its testing","status":"NEW","created":"2025-07-06 15:12:15.000000000","updated":"2026-09-06 19:39:41.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aba3a5d506d76ad3aad0f936f887ff80c876578a","_number":954193,"virtual_id_number":954193,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-07-29 19:43:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-06 17:12:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2025-07-06 17:19:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-15 20:28:05.000000000","updated_by":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"CC"},{"updated":"2026-07-29 17:22:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"state":"REVIEWER"},{"updated":"2026-07-29 17:24:04.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":6968,"name":"Christian Schwede","email":"cschwede@nvidia.com","username":"cschwede"},"state":"REVIEWER"}],"messages":[{"id":"c449c6648213d32101927c910baa973ec3bef0de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 15:12:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7b5b944062914cb6fe55b155540158e47e2d3835","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 15:14:58.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"faaae97536462d82ff49a829e8ccd4e253b9422d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 17:12:52.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d973bf2bdfa5455289ad853da4dad0f3\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/37fc10f549af451b9074eaf7f4e710b3 : FAILURE in 1h 56m 18s","accounts_in_message":[],"_revision_number":2},{"id":"7309bd2f51eee248c084443c0a66d0ed91f4ab77","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 17:19:55.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/727e4e65deb4431e8951ee838d50edb5\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d3c90d5501744088ea27cccff81bc5a : SUCCESS in 8m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/996d8fc2babe477597f8405fc814f948 : FAILURE in 44m 32s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/6d0b3e6d0c3b4d0c8f5e866ae0a08a7b : FAILURE in 49m 20s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/54cfb7c030994463856ac3416e2421ff : FAILURE in 17m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8921a326259440a4bf1ccbbef4d87c74 : POST_FAILURE in 10m 50s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/970233d5b08b4f339dfe38490098b806 : FAILURE in 6m 43s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/71f21b99e57e4fb199def76164f3e162 : FAILURE in 6m 56s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/01c33a0364f94830aa8c39e009a927ba : FAILURE in 7m 13s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/bc2e05f620d1431f810b47f22c2a28ab : FAILURE in 3m 45s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/5803c8557cd74091b0a5ad97ee23fcc0 : FAILURE in 42m 27s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/d3e9bc50be904c498086dde7b7e702a1 : FAILURE in 46m 49s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/0703fb542d61424abd76270a1069c394 : FAILURE in 52m 34s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/626e1c57689f43e7951ea2d25abb9f94 : FAILURE in 53m 44s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/0eb78cdf764e4df2908199394e314de3 : FAILURE in 4m 07s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/7a3e7a6b7e6f43eabe3ea0dd738b58c9 : FAILURE in 4m 20s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/c47e1bc98873422f820a4ffabd3881e3 : FAILURE in 3m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/2f7c19b7caf745b884f68fbbcf8a4c29 : FAILURE in 6m 14s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/b43a6c3539c94584b13aca6a22191fd3 : FAILURE in 5m 02s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/5acfc8c5ae6e4ba584409666cdbaad88 : FAILURE in 10m 12s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/c839e7e3b0fa4c328306701423f19f15 : TIMED_OUT in 2h 00m 38s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/6a3b10c80edf43ba808395ebcb40de54 : FAILURE in 19m 24s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/6a22e8a7bec54cd086722fc1bd755175 : FAILURE in 18m 29s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/5a9ced677559438d924b32721208377d : FAILURE in 24m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a4b7186e10b44b29039e7d01949784d : SUCCESS in 5m 00s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/edd4a71d2b234686b9641b871f2e0dc8 : TIMED_OUT in 1h 30m 48s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3fc2cb4a06434c519107d6fd60a51631 : FAILURE in 23m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d3963b57c3804fa8b86a16c22596aee3 : FAILURE in 20m 27s","accounts_in_message":[],"_revision_number":2},{"id":"2678c775d1d9c31fddd965a0f6c4119bda75ce3a","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-07-06 21:26:14.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"82fcc119f142c54afe8e9cc7e624a68faca8cf49","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 23:28:19.000000000","message":"Patch Set 2:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd27700b891f4421856356c48b2b14b0\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/97621d09eff34fbdaf5e117a60b82a83 : FAILURE in 2h 00m 18s","accounts_in_message":[],"_revision_number":2},{"id":"c0cfe7b5fb82ce5e65786a3d3c4dfd1bb9e653bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-06 23:32:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71bdb717dbce4bf5bff12c704f9c681d\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78d21045407415c872f1f0e15edbe24 : SUCCESS in 11m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/7725cc5103c646f18e6710c1a35b964b : FAILURE in 21m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ec7b05b4b169427cabad190a196aa54c : FAILURE in 23m 10s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/7c887573cbc449eab0ac09a14ab0c83e : FAILURE in 10m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f058c1a0fc3344ba94d75d086cfdc903 : POST_FAILURE in 30m 35s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/9487fd8313c0431baf7d56872d65c968 : FAILURE in 3m 46s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/8ce60f466ef54b7d98712c48bc68eb98 : FAILURE in 6m 03s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/ddf629e539f148b4b1b8d50ca1ddc346 : FAILURE in 5m 53s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/3c93b0836d924b4e924ec2a7ff101d85 : SUCCESS in 5m 29s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/a570903723c34b9b9363e2c4abb8a6ec : FAILURE in 43m 36s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/e5723913cc464090838f2d98ee24b21e : FAILURE in 26m 00s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/a8e1d5401dea4517878c483fed71ad4b : FAILURE in 45m 17s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/4c34f1710d1d45db8e7a366df02d8b50 : FAILURE in 30m 21s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7865325fa13547beae5dc857e4add98a : FAILURE in 4m 18s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/1c993b480f224dffa76a1cc25f970d3c : FAILURE in 3m 58s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ff3e8df38c8d4cb291ac0d490bafb0f0 : FAILURE in 4m 19s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/59680be1332a4df6a2d2bf6b27a5ef56 : FAILURE in 5m 51s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/00ff8b6a35d64f46bafc45c78bbe8628 : FAILURE in 2m 51s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/67f3bca2193346638a2477eb8ad92bb9 : FAILURE in 6m 37s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/6dfc22f5cc8c4231b1f127a3e12942f2 : TIMED_OUT in 2h 00m 56s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/ea89ccf5d96c4af89c960f778136fad7 : FAILURE in 13m 28s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/3b05391a03314b2b989824a396b27adb : FAILURE in 11m 09s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/f397e77f98fa4e1982505d1e0dccd472 : FAILURE in 24m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efcaaed01ee549358ba0d1812c330352 : SUCCESS in 4m 52s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/924b5e4f146e4f1eb3904b59daf2bba6 : TIMED_OUT in 1h 31m 50s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/19bead952a2e46c7bea2306e5c173561 : FAILURE in 18m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc16dd6fde864d5489c58d2d1696e45e : FAILURE in 13m 42s","accounts_in_message":[],"_revision_number":2},{"id":"254700d5bce01630c91c6a5451f5dce61d71ce6e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-15 13:25:48.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"0448cdf32f02ba15224f2209a7448de80764d7cc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 15:36:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83787fc45a6d47c99a0dd2c582996f5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cdafd8031a34b68b04b6417e6736e91 : SUCCESS in 10m 43s\n- grenade https://zuul.opendev.org/t/openstack/build/815b418eca1a4cac8fcf77bc19c9b156 : FAILURE in 41m 08s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/581167522e274039b1e704cb7b2b484c : FAILURE in 49m 31s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/435b87a76222489f8758efa61cf6f27e : FAILURE in 25m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f23aa77d1559407e88e13b85145af52e : POST_FAILURE in 25m 30s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/a04af1fa5ad24346ac0aa39072bb8628 : FAILURE in 7m 12s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e788dd0225964731b496a800490a5614 : FAILURE in 6m 34s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/493336cc93b440418765a02971ca9eca : FAILURE in 8m 08s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/131eb9b1cf6545769402b379d7da611e : SUCCESS in 7m 20s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/3fe9d44f68eb45a09b35737b5a19db8f : FAILURE in 1h 01m 51s\n- swift-tox-py39 https://zuul.opendev.org/t/openstack/build/298da18ac9a04a8c919a4bd753da5cfe : FAILURE in 49m 24s\n- swift-tox-py312 https://zuul.opendev.org/t/openstack/build/bd13a5f85a9c4dca9ef94b01cb2f1ba3 : FAILURE in 53m 10s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/3fd8336d371a4e769cd84541fe6a0846 : FAILURE in 1h 09m 13s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/d1316a48127c4e2db774611b10aa552a : FAILURE in 4m 17s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/f4fe9417d2084d57ba7148eb4d5336e7 : FAILURE in 5m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/a5fb1be2d947477898413bdc1811300e : FAILURE in 5m 35s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/c14dfa8e963543ff80c53c7a5c230795 : FAILURE in 6m 20s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/5809f2959d09446c91c7ad7ee0f97700 : FAILURE in 5m 58s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/65cb087ea6cd433aa848d36f451d4605 : FAILURE in 9m 36s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/551c4c527d064cb0b0de4aae30530467 : TIMED_OUT in 2h 01m 07s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/4bf0dcc8dfd9439f8daa4005ed34a3b9 : FAILURE in 18m 29s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/a3aba6d242a245a7bc50dfcafb6cd827 : FAILURE in 14m 03s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdd87364a4a04c9a98d9e48cd88d4acc : TIMED_OUT in 41m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/43aab569788d4f09a12667306492e3f0 : SUCCESS in 5m 30s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/fce15fd50495420dbb9df865a6d5bd83 : TIMED_OUT in 1h 30m 54s\n- tempest-integrated-object-storage-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/fc0d22b70bd34250848c00098fbac40e : FAILURE in 14m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6cb52b811ddc46b48719027f58eabc24 : FAILURE in 18m 32s","accounts_in_message":[],"_revision_number":3},{"id":"015a44e8f20d40ee34e6874d3ba8b8cc4556dd5f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-15 19:03:34.000000000","message":"Patch Set 3:\n\nBuild failed (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b46fd8fb251f424f9ecdb511c018f721\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/04d92acea3bb496499d0ffc03a9fdf24 : FAILURE in 2h 07m 14s","accounts_in_message":[],"_revision_number":3},{"id":"6cd342328f59f9e599f913a2def11e40ca78f1b2","author":{"_account_id":15343,"name":"Tim Burke","email":"tburke@nvidia.com","username":"tburke"},"date":"2025-09-15 20:28:05.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"869964d4c0acac7ec9e48c86d2e29c615bd67dbc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-15 20:55:10.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"82e9266884c778575f2fe8665bd43ffb4298bd35","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-09-15 20:55:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":3},{"id":"0eb99d8bad5c2f98e6099f59151be515cfd79dcf","tag":"autogenerated:gerrit:restore","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-18 21:54:56.000000000","message":"Restored","accounts_in_message":[],"_revision_number":3},{"id":"cf0833c207eb85313aa2ab78b0b512e07da998ec","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-18 21:55:07.000000000","message":"Patch Set 4: Published edit on patch set 3.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"b09861b21df25ee77ba84f7ffd0f0c4f1d591a94","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-18 21:55:11.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8953257e4b97665813bb29982b8c82181d95c490","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-06-18 21:56:27.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"4049ceffb62233191f1c04981dc96c5b70d1693c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 22:51:50.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d0567910f5649bba9c12cba17b54557\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/888b9d44b37644e599ff1ccf20e41f17 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1908d72fd43d418ebd16de691cd26653 : SUCCESS in 7m 46s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/de91f68a2f154207ad79c5a16a11ab70 : SUCCESS in 19m 25s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b79770c19e904b309199a7223ce9eaa3 : SUCCESS in 21m 36s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/818067539bcb4db7b188c29c29386389 : SUCCESS in 19m 50s (non-voting)\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/7d7fcc4c773b4e7d918d8ce55d523847 : SUCCESS in 7m 58s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8cd51b261ab6424cbe67ef429ce507df : SUCCESS in 15m 22s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/749fa7f41ba44159b5f83b0e94d8ad94 : SUCCESS in 8m 32s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/55d3a8d922fa4151aa4b583e544a64f9 : SUCCESS in 51m 21s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/f3193306a20642ac8c92c1db9c090314 : SUCCESS in 36m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/475a15fc86f743cd8f8190ff5365c714 : SUCCESS in 4m 16s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/7ad4522bb7e948ce90e9a3b1651a23bd : SUCCESS in 19m 53s","accounts_in_message":[],"_revision_number":6},{"id":"6e86398041e2df56045d7fb4d2a755f71cb5edc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 22:57:01.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/adc8ef776f364329b144798a6ea96e62\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/47c0effa57634aac921982795dbece15 : SUCCESS in 56m 34s","accounts_in_message":[],"_revision_number":6},{"id":"79ed6af2b0e64c2f468929b94738c67839da18bc","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 17:20:57.000000000","message":"Patch Set 7: Commit message was updated.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":7},{"id":"9c7b2f8815884451fe45668841cd9abc68ac7f0b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 17:21:05.000000000","message":"Patch Set 8: Published edit on patch set 7.","accounts_in_message":[],"_revision_number":8},{"id":"b080883400182e632af3bcc0addbd5d5330cb1de","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 17:21:08.000000000","message":"Patch Set 9: Patch Set 8 was rebased","accounts_in_message":[],"_revision_number":9},{"id":"7c1b313700a5e571fa573ce44bd1b0935955ef4e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 17:22:15.000000000","message":"Patch Set 10: Published edit on patch set 9.","accounts_in_message":[],"_revision_number":10},{"id":"7663c3a52e4bf05d2cd164dfcccb96c6d3564fe1","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 17:22:22.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":10},{"id":"3845b70711f5be6125ab36a0bb89e8c80ac98375","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 18:34:06.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb68d077030c440186c99b019a9b4336\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/bb0cc68683e94a55876faf228a02927d : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81fe69f79a3c465a9c2fc689d381e89d : SUCCESS in 10m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ed6988b1a33643c2837f16d7fa4479e0 : SUCCESS in 3m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e82aa118ba6d4ab8869115919348003f : SUCCESS in 7m 39s\n- grenade https://zuul.opendev.org/t/openstack/build/89bb10931791433f85fdbb145d5743ae : SUCCESS in 57m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d87eddc259a44b2ab7225160770d8b4c : SUCCESS in 1h 07m 08s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/5f99118c49d844bb8e2e4f2db670a5fb : SUCCESS in 43m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/26cdda314adb45b7afb4a25aae2eec72 : SUCCESS in 55m 38s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/b5c52f4b809a4a7a8d87d376fb2ea950 : SUCCESS in 19m 20s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/1ec87435e4ed43aeb86cd32842e4302c : SUCCESS in 20m 40s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/636b08cc7a4545c0b8b1e4ae8b8169b4 : SUCCESS in 15m 43s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/51c685a0a92b44b8b38136e2baeebc05 : SUCCESS in 4m 48s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/0d9ac8656e9347a6af8dafb7caad72db : SUCCESS in 22m 31s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/dcf0f554ba8a4faaa0894e38bf6fc84f : SUCCESS in 17m 36s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/9391352f869e4c18a627ccba23a186f4 : SUCCESS in 30m 36s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/576ccc3c7e80495ba5ed7ac70ff0d825 : SUCCESS in 23m 09s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/5c4cbd3ed90e4850a5971c9b0ca555cb : SUCCESS in 16m 15s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/8892c09a9d2e43d7a7eaec72eb9ac2ab : SUCCESS in 16m 06s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/ce39bca57e9a48b996ae0adb075e90f0 : SUCCESS in 16m 00s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/c0775af63a2949d598fc39a27d532b65 : SUCCESS in 17m 23s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/4ef5b3c5a9d94eeba450b1ac55b4187e : SUCCESS in 6m 43s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/a8c9f6efcd0c40548834f764583d4ca7 : FAILURE in 5m 59s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/b81c9b68563c464ebffcc8d451dcfd0a : SUCCESS in 5m 11s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/e55004ad168d446a9b7cd2485521b66b : SUCCESS in 1h 06m 10s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/68c2e95bdf194d429c91ccfeb810b73e : SUCCESS in 55m 47s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/de3e4f0743894ad68995dbda011eddf1 : SUCCESS in 22m 49s\n- swift-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2647a8b221a2452ea9f3472cae73958f : FAILURE in 5m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91f02c984ca46fabc5763e3ce9d2b34 : SUCCESS in 3m 59s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/2e1f2de7c8064b78b41c4cf36ad3fb8d : SUCCESS in 26m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/816ab47e841c4a398b99e823db7fafd3 : SUCCESS in 36m 48s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":10},{"id":"8347896f3037ccbf2de33e2c623b36299253da62","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-07-29 18:37:41.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":11},{"id":"e7cc9def28546b155997ed0029325dcae2099b70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 19:43:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf496df5eca843cdb49ace156ec2516a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/dbdad3c2dcb14b4488111ce9fa76a174 : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e84b4e4f53c4a789c5fc9a2f143f08b : SUCCESS in 10m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9950a511bf96416ea011d67a0f7c80d6 : SUCCESS in 2m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f42ff7fed69410b8f6339e9241afeb7 : SUCCESS in 7m 51s\n- grenade https://zuul.opendev.org/t/openstack/build/69973c054c6a49beaadccb2933126f3f : SUCCESS in 1h 03m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4a999b68470542b6ad38fd9b92e4e8fb : SUCCESS in 58m 37s\n- tempest-integrated-object-storage https://zuul.opendev.org/t/openstack/build/641c705df7b24bb8814b4539f55efd2c : SUCCESS in 56m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9704744cfda7431189eea1db5248fee5 : SUCCESS in 44m 47s\n- swift-tox-func-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e10666072d7f44aa9885e73b7d1b46b9 : SUCCESS in 19m 28s (non-voting)\n- swift-tox-func-encryption-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/e3abee3c9d6c4fcaa2ca4be8756af221 : SUCCESS in 17m 10s (non-voting)\n- swift-tox-func-ec-py39-centos-9-stream-fips https://zuul.opendev.org/t/openstack/build/f81d7290730949369ccd5d8319e4c179 : SUCCESS in 11m 16s (non-voting)\n- swift-build-image https://zuul.opendev.org/t/openstack/build/2b819249cad24018bb15bacbeda83e2a : SUCCESS in 7m 24s (non-voting)\n- swift-tox-py37 https://zuul.opendev.org/t/openstack/build/10198c263c434122967df3fef0321eef : SUCCESS in 26m 15s\n- swift-tox-py310 https://zuul.opendev.org/t/openstack/build/c3392c7c073a4477837bfb941c9bbdb3 : SUCCESS in 9m 48s\n- swift-tox-py313 https://zuul.opendev.org/t/openstack/build/c539c30a4559484f9a3eb54d50bcb466 : SUCCESS in 23m 33s\n- swift-tox-py314 https://zuul.opendev.org/t/openstack/build/08abc6f1089d4c99b57450ed03e54cc8 : SUCCESS in 13m 55s\n- swift-tox-func-py312 https://zuul.opendev.org/t/openstack/build/fc061718367343fa958d65414b812304 : SUCCESS in 17m 29s\n- swift-tox-func-encryption-py312 https://zuul.opendev.org/t/openstack/build/b119ad135cda4685aefca39939d176af : SUCCESS in 15m 24s\n- swift-tox-func-ec-py312 https://zuul.opendev.org/t/openstack/build/64c7ed69eddf4b30bee8b738661cdcfb : SUCCESS in 15m 44s\n- swift-tox-func-py39-centos-9-stream https://zuul.opendev.org/t/openstack/build/f259766a8ec04958b6557b3e70a1c6f4 : SUCCESS in 18m 44s\n- swift-func-cors https://zuul.opendev.org/t/openstack/build/6e8853a55999422c80ea01e163adb484 : SUCCESS in 4m 29s\n- swift-tox-func-ceph-s3tests-tempauth https://zuul.opendev.org/t/openstack/build/4cce77cb30d841aba35d71c70f82df2b : FAILURE in 7m 38s (non-voting)\n- swift-tox-func-s3api-compat-tests-tempauth https://zuul.opendev.org/t/openstack/build/525548e86a414cffb44d59c6c01d4fff : SUCCESS in 9m 04s\n- swift-probetests-centos-9-stream https://zuul.opendev.org/t/openstack/build/f687b1b8ce7b403db594d7eace6285f7 : SUCCESS in 37m 48s\n- swift-dsvm-functional https://zuul.opendev.org/t/openstack/build/aa9836b400584f58b59b97e3d8910028 : SUCCESS in 40m 27s\n- swift-dsvm-functional-ipv6 https://zuul.opendev.org/t/openstack/build/d1b08a7c826a483b8d1c7f2265fec030 : SUCCESS in 23m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc7fae2f369b40f786afa7b482d9c56b : SUCCESS in 5m 38s\n- swift-multinode-rolling-upgrade https://zuul.opendev.org/t/openstack/build/9e7e104fcaa1495d9e2a88f1ecf88f95 : SUCCESS in 9m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a09999dd04c4d299abde06e14a8b41c : SUCCESS in 55m 21s\n\nWarning:\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-ec-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-encryption-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/zuul/zuul-jobs/playbooks/unittests/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/cleanup.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.\n  File .ssh/known_hosts.old found in build home directory during playbook opendev.org/opendev/base-jobs/playbooks/base/post-logs.yaml for job swift-tox-func-py39-centos-9-stream-fips.  Add to preserve-home-paths if safe, or otherwise remove.","accounts_in_message":[],"_revision_number":11},{"id":"660b94b9ce869e40f56b612c98abf07f2caf63e8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-30 00:52:59.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/037975152c60470281b4bb1686719c21\n\n- swift-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1ec755a581c544a29706d0204b9876fb : SUCCESS in 53m 07s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"98538b16f2c7379f8a25fe0b579798a12ae72d44","revisions":{"20025481fa499bed0553946dddebb7d0c5252d01":{"kind":"REWORK","_number":1,"created":"2025-07-06 15:12:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/1"}}},"commit":{"parents":[{"commit":"364cc6556fcdbf12ee66c6d5a7ef5609fef51eaa","subject":"Merge \"s3api: fix multi-upload BadDigest error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/364cc6556fcdbf12ee66c6d5a7ef5609fef51eaa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20025481fa499bed0553946dddebb7d0c5252d01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/20025481fa499bed0553946dddebb7d0c5252d01"}]},"branch":"refs/heads/master"},"8691d9676e08177201ff74c39d13d1a63ce3dd34":{"kind":"REWORK","_number":2,"created":"2025-07-06 15:14:58.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/2"}}},"commit":{"parents":[{"commit":"364cc6556fcdbf12ee66c6d5a7ef5609fef51eaa","subject":"Merge \"s3api: fix multi-upload BadDigest error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/364cc6556fcdbf12ee66c6d5a7ef5609fef51eaa"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:14:57.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8691d9676e08177201ff74c39d13d1a63ce3dd34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/8691d9676e08177201ff74c39d13d1a63ce3dd34"}]},"branch":"refs/heads/master"},"91efa51db2fc2e7461addfab6d2adf9510bc67cd":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-15 13:25:48.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/3"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-09-15 13:25:48.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91efa51db2fc2e7461addfab6d2adf9510bc67cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/91efa51db2fc2e7461addfab6d2adf9510bc67cd"}]},"branch":"refs/heads/master","description":"Rebase"},"ad9c4899cb3fc3fa25a24d114e44b84d42de2ccd":{"kind":"REWORK","_number":4,"created":"2026-06-18 21:55:07.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/4"}}},"commit":{"parents":[{"commit":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","subject":"AUTHORS/CHANGELOG for 2.36.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/82cb5a5d78b91c6af258cd8f06d30f69be90fa18"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-18 21:55:06.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad9c4899cb3fc3fa25a24d114e44b84d42de2ccd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/ad9c4899cb3fc3fa25a24d114e44b84d42de2ccd"}]},"branch":"refs/heads/master"},"a202620477c4759c52178a864fb9fd9e9af7d6a4":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-06-18 21:55:11.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/5"}}},"commit":{"parents":[{"commit":"0dc24a9fb7f3dfefca6ff4443ba98891534e1868","subject":"Merge \"s3api: 501 unsupported bucket subresources at other levels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc24a9fb7f3dfefca6ff4443ba98891534e1868"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-18 21:55:11.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a202620477c4759c52178a864fb9fd9e9af7d6a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/a202620477c4759c52178a864fb9fd9e9af7d6a4"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"82cb5a5d78b91c6af258cd8f06d30f69be90fa18","ours":"ad9c4899cb3fc3fa25a24d114e44b84d42de2ccd","theirs":"0dc24a9fb7f3dfefca6ff4443ba98891534e1868","merge_strategy":"recursive","contains_conflicts":false}},"03c1664f16265c936b5512c4f076f8ca8ba6b672":{"kind":"REWORK","_number":6,"created":"2026-06-18 21:56:27.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/6"}}},"commit":{"parents":[{"commit":"0dc24a9fb7f3dfefca6ff4443ba98891534e1868","subject":"Merge \"s3api: 501 unsupported bucket subresources at other levels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc24a9fb7f3dfefca6ff4443ba98891534e1868"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-06-18 21:56:26.000000000","tz":0},"subject":"wip","message":"wip\n\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03c1664f16265c936b5512c4f076f8ca8ba6b672"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/03c1664f16265c936b5512c4f076f8ca8ba6b672"}]},"branch":"refs/heads/master"},"bdc56278d1ca0ab73f0871fe45782397fa912595":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2026-07-29 17:20:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/7"}}},"commit":{"parents":[{"commit":"0dc24a9fb7f3dfefca6ff4443ba98891534e1868","subject":"Merge \"s3api: 501 unsupported bucket subresources at other levels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc24a9fb7f3dfefca6ff4443ba98891534e1868"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 17:20:57.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nCo-authored-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdc56278d1ca0ab73f0871fe45782397fa912595"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/bdc56278d1ca0ab73f0871fe45782397fa912595"}]},"branch":"refs/heads/master","description":"Edit commit message"},"690b7d28ab5b07d24d31d49a4257e5c4eaa87f72":{"kind":"REWORK","_number":8,"created":"2026-07-29 17:21:05.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/8"}}},"commit":{"parents":[{"commit":"0dc24a9fb7f3dfefca6ff4443ba98891534e1868","subject":"Merge \"s3api: 501 unsupported bucket subresources at other levels\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/0dc24a9fb7f3dfefca6ff4443ba98891534e1868"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 17:21:03.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nCo-authored-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/690b7d28ab5b07d24d31d49a4257e5c4eaa87f72"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/690b7d28ab5b07d24d31d49a4257e5c4eaa87f72"}]},"branch":"refs/heads/master"},"accaf32b8cfb3ddb8aab688ba8f0af3f67547a09":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2026-07-29 17:21:08.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/9"}}},"commit":{"parents":[{"commit":"05afa6c87f289bb63a76a0e1f882a2eebe66bd79","subject":"Merge \"s3api: drop native Swift control headers from client requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05afa6c87f289bb63a76a0e1f882a2eebe66bd79"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 17:21:08.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nCo-authored-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/accaf32b8cfb3ddb8aab688ba8f0af3f67547a09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/accaf32b8cfb3ddb8aab688ba8f0af3f67547a09"}]},"branch":"refs/heads/master","description":"Rebase","conflicts":{"base":"0dc24a9fb7f3dfefca6ff4443ba98891534e1868","ours":"690b7d28ab5b07d24d31d49a4257e5c4eaa87f72","theirs":"05afa6c87f289bb63a76a0e1f882a2eebe66bd79","merge_strategy":"recursive","contains_conflicts":false}},"abac9c7f0b29def601d3f7b9f7e821389af06c64":{"kind":"REWORK","_number":10,"created":"2026-07-29 17:22:15.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/10"}}},"commit":{"parents":[{"commit":"05afa6c87f289bb63a76a0e1f882a2eebe66bd79","subject":"Merge \"s3api: drop native Swift control headers from client requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05afa6c87f289bb63a76a0e1f882a2eebe66bd79"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 17:22:15.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nCo-authored-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abac9c7f0b29def601d3f7b9f7e821389af06c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/abac9c7f0b29def601d3f7b9f7e821389af06c64"}]},"branch":"refs/heads/master"},"98538b16f2c7379f8a25fe0b579798a12ae72d44":{"kind":"REWORK","_number":11,"created":"2026-07-29 18:37:41.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/93/954193/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/swift","ref":"refs/changes/93/954193/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/swift refs/changes/93/954193/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/swift refs/changes/93/954193/11"}}},"commit":{"parents":[{"commit":"05afa6c87f289bb63a76a0e1f882a2eebe66bd79","subject":"Merge \"s3api: drop native Swift control headers from client requests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/05afa6c87f289bb63a76a0e1f882a2eebe66bd79"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-07-06 15:12:15.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2026-07-29 18:37:40.000000000","tz":0},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: I7a5de760143488567c4930a0c893f7ce9cadf05b\nCo-authored-by: Ghanshyam Maan \u003cgmaan.os14@gmail.com\u003e\nSigned-off-by: Ivan Anfimov \u003clazekteam@gmail.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98538b16f2c7379f8a25fe0b579798a12ae72d44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/swift/commit/98538b16f2c7379f8a25fe0b579798a12ae72d44"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
